)]}'
{"id":"openstack%2Fmanila~928608","triplet_id":"openstack%2Fmanila~master~I275a3680d3ea6b1b1ea69d78d4b7fecb4f61af03","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-09-10 10:09:52.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2024-09-10 18:30:03.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2024-09-10 18:30:03.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-10 17:35:44.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2024-09-10 18:30:03.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-09-10 10:02:22.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}}},"hashtags":[],"change_id":"I275a3680d3ea6b1b1ea69d78d4b7fecb4f61af03","subject":"Fix the remaining review comments from SnapLock patch","status":"MERGED","created":"2024-09-09 13:15:04.000000000","updated":"2024-09-10 18:31:23.000000000","submitted":"2024-09-10 18:30:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928608","meta_rev_id":"4cd1cc1989938dc57412243a6b9769f4f530ccc3","_number":928608,"virtual_id_number":928608,"owner":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-10 18:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-09-10 10:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-09-10 17:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-09-10 10:02:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2024-09-10 17:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 13:17:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2024-09-09 13:17:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"REVIEWER"},{"updated":"2024-09-09 13:17:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-09-09 13:17:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-09-09 13:17:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-09-09 13:43:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-09 13:48:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-09 15:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 15:16:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-09-09 15:21:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-09 15:38:24.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"}],"messages":[{"id":"282dbad00cf284d3094202b1c58e44a3db71c918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-09 13:15:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dc9ae3614075bfd65d61217755514f7efaa1f53","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-09 13:43:45.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/74ba751cccc54a468b29494cf7a39060\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ddbb027d7743462b94a831dac40a356e : FAILURE in 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8c0274c32a849369b12b8df5afb941f2c31f88e1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-09 13:48:00.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/1/check/DellEMC_PowerFlex_Manila_dhss_false/a23cb1a/DellEMC_PowerFlex_Manila_dhss_false/1645 : SUCCESS in 32m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"68ff74d4e3fbf4a14e6a7c4125609f87dcd5ff98","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-09 14:38:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"672b6123ba8558dd959b7c038181fef14135f0d9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-09 14:39:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac59b7194d6a193dd8410b930479c2bb27a597ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 15:02:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8464a6eeaf20490bb0affd7c81d40025\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6e915e20234152bfb047f581f267de : SUCCESS in 10m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a319c6fd09df4b31a49426255ef04370 : FAILURE in 29m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43d711353964011baa866f539cfd506 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4b94b25a997430b811c1ceaa7f3218e : SUCCESS in 8m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b7c2e8ca6b84cff8a718e8a5056eaa8 : SUCCESS in 7m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1dbb2a03979439f8668fa818a518470 : SUCCESS in 9m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3375b148196f4ab3a00dee4cf381eba4 : SUCCESS in 7m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f3d1b08e6b434818a4d57fe0e41494d7 : SUCCESS in 1h 09m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b71f8e384e574f9580bf50a74308e1df : SUCCESS in 1h 45m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d980dc311f5e47d8b8b5e4a6b14fc279 : FAILURE in 57m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8132a715bd1d40389545e9928246261e : SUCCESS in 38m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5ab53994b0724c93aee04994730ba333 : SUCCESS in 1h 00m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5ebe206592f24d0cb4a5f10d1ec2b0fd : FAILURE in 1h 18m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/57a02ddc542c4b72a44781c1f58a0213 : SUCCESS in 4m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78de40cebd6d4b089662a8412519367f : SUCCESS in 4m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c195d651cc514340927bbc9d143a03b8 : SUCCESS in 24m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0baafc8f6e134a76886c34b132e00377 : SUCCESS in 37m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0a8f577051f84b4cac7b673095d62cfa : FAILURE in 45m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6137a76456b6408280e3ec049c6b49bd : SUCCESS in 1h 19m 41s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/763754afff0f408d8f7fbadd3c9a5708 : SUCCESS in 1h 01m 22s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/63f0a34834914d16ae6510f5a0017372 : SUCCESS in 21m 20s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ba0b931b70e249fcaf1a457f3fa0a4f8 : SUCCESS in 39m 52s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b98906b74ad9494aaaf3810724331dc4 : FAILURE in 4m 33s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/cd758c7b4d614fa8a4409cba14521c51 : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acc9780a079d752c4471c4d8720b0340700d3f87","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-09 15:16:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/08/928608/1/check/DellEMC_Unity_Manila_dhss_true/52204ab/DellEMC_Unity_Manila_dhss_true/1642 : FAILURE in 14m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/1/check/DellEMC_Unity_Manila_dhss_false/63a7c4e/DellEMC_Unity_Manila_dhss_false/1750 : SUCCESS in 2h 01m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"99c4eda8e8665d28335df45b9ec094e997250b3e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-09 15:21:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/08/928608/1/upstream-check/manila-tempest-plugin-ontap-dhss-rest/999727d/ : RETRY_LIMIT in 15m 13s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/08/928608/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/3c4ac1e/ : RETRY_LIMIT in 18m 00s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d63272ccc6b974e25c8a5d50ca7c9b82ad08d7f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-09 15:35:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/1/check/DellEMC_PowerFlex_Manila_dhss_false/b0423da/DellEMC_PowerFlex_Manila_dhss_false/1646 : SUCCESS in 32m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"ab3efebd52dee6ad868e4a5eb6a54b052e55c9f1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-09 15:38:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/1/check/DellEMC_PowerStore_Manila_dhss_false/9260b51/DellEMC_PowerStore_Manila_dhss_false/480 : SUCCESS in 35m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"8506749b297a7d1c93ed523e5106494aaee01dcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-10 05:25:02.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f9b289729c4df148404518db536a3c6fbfd7ca6","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-10 05:26:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fb9a97d759aad6736f0b58b74ee1d877a8e5152","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-10 05:55:48.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b397a482044c4e3a964d5629298911ea\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5ad332c690a4be895a8b7a19b21c0fa : FAILURE in 30m 15s","accounts_in_message":[],"_revision_number":2},{"id":"5c3d5a6ac5db36bc3411dbdef60779127adc753b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-10 06:00:06.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/2/check/DellEMC_PowerFlex_Manila_dhss_false/f40dc51/DellEMC_PowerFlex_Manila_dhss_false/1653 : SUCCESS in 34m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc50e6757a0f1a2197cbd6afa5fc0d3dcd8c3775","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-10 06:51:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/08/928608/2/check/DellEMC_Unity_Manila_dhss_true/db37461/DellEMC_Unity_Manila_dhss_true/1649 : FAILURE in 13m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/2/check/DellEMC_Unity_Manila_dhss_false/7f920c7/DellEMC_Unity_Manila_dhss_false/1757 : SUCCESS in 1h 25m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ec873937903c6c5ee86747c05c388430d2379da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 08:38:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/060cf252841d401b8b609cd5dea260b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ec6605b4404fce9e3ca240da5fb68d : SUCCESS in 8m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fca91e0b8c4e4fb58707ee2cc06f1289 : SUCCESS in 21m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06bd305f55bb433292a1c363c9051796 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0292036bd30c4c17afc2a4818b70e76a : SUCCESS in 11m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c70f2c5f450b4064ab06f26cf3f0b4c0 : SUCCESS in 7m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/968e908220f343a1a044b9285553164d : SUCCESS in 9m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8f9332d456b4c999f32460a4180219c : SUCCESS in 6m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51012abb89bd4144862fefab86c7d0b4 : SUCCESS in 38m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3c8da6635a8649b1ae8cd85ca94cd5be : TIMED_OUT in 3h 02m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f9d96fefbd9740658f1340b005c4fd11 : FAILURE in 1h 21m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b741668d39c04c7d8a6ff2ff97473000 : SUCCESS in 50m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/774026ad286040488f100ab7fdbea7a8 : FAILURE in 9m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5078f268e7324bf3b97251a46ddf648c : SUCCESS in 1h 07m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bf24c33491f844f1819454e874226c1f : SUCCESS in 3m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1492550dcde4444485f4ca222b9e91d8 : SUCCESS in 3m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a4ec6f243e324ce6b634d9803493ddcd : SUCCESS in 23m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b3d87aef8c0f411189ed35fea92671a8 : SUCCESS in 37m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2e2a47fe2b964515be6a0d9c81322916 : FAILURE in 42m 39s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/7e94ebc1979c4a3184253c41c0c2d096 : SUCCESS in 42m 00s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76be2d220a3946d0981b16cef3905534 : SUCCESS in 46m 02s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1465aade119e486bacab9381b8dbfb27 : SUCCESS in 21m 08s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/97c0720a4b734af5bcc0d79f03258399 : SUCCESS in 23m 45s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/349634cea1cb4865997e458559c39e01 : FAILURE in 3m 57s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/badca1dd9af343acace14dd8bf45ecf4 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19b2ef8db9656821773d9a7a1492434a329a4e6e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-10 08:56:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/08/928608/2/upstream-check/manila-tempest-plugin-ontap-dhss-rest/480e521/ : RETRY_LIMIT in 9m 21s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/08/928608/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/984c9eb/ : RETRY_LIMIT in 11m 36s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/08/928608/2/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/cc6f9d9/ : RETRY_LIMIT in 13m 04s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"6510b3848b8649e7a2fed5ca3d0d0ec8d1e662ed","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-10 09:11:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/2/check/DellEMC_PowerFlex_Manila_dhss_false/71992f6/DellEMC_PowerFlex_Manila_dhss_false/1654 : SUCCESS in 33m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f53c2295c23f8a459bfdc3b2e30c8d91410045b3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-10 09:12:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/2/check/DellEMC_PowerStore_Manila_dhss_false/1ece654/DellEMC_PowerStore_Manila_dhss_false/483 : SUCCESS in 33m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"08f33887a76798709ed66bb5e0c1449c901c25cf","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-09-10 10:02:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cf6ad6d8c3d138c62c5ea6839f32f4db688e4fb8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-10 10:07:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/08/928608/2/check/DellEMC_Unity_Manila_dhss_true/144e769/DellEMC_Unity_Manila_dhss_true/1650 : FAILURE in 13m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/928608/2/check/DellEMC_Unity_Manila_dhss_false/7524a79/DellEMC_Unity_Manila_dhss_false/1758 : SUCCESS in 1h 28m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a059e34f33e8591643d35a2539b889be3dd42cc1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-10 10:09:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eabcc8017c328b954ddc8feb77d1fea97cd5123a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-10 17:35:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67613913ff3e14c29715bea9a8a173e90c7b5264","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-10 17:35:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e4b76b6cc4c31702ae72e66779f62220d6697df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 17:36:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8500996aee4ba9e87cec0830108dba92756e2258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 18:30:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad44965cd4284b06a35d00c39c743234\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a19c7219b9a401b91fedd4cc21936c6 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743348f96ce64ad6a3b6c36c131365ca : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56b4ff36c04b49089443fcace532ea81 : SUCCESS in 9m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffa0176a0db24a1cbc91a4756bc0bd5f : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/360e5c43139842d583eb4d793c0c4eca : SUCCESS in 7m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/951ebfc9177c4df180f4c4ebf70a2211 : SUCCESS in 52m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/87bd09fcb85a434196888cabd8412b0c : SUCCESS in 27m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ff16004da0841afb3d7c4edd1bc5631 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":2},{"id":"60ec4d30dc9f1bf5e592b3640e6ab3567c05e2d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 18:30:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4cd1cc1989938dc57412243a6b9769f4f530ccc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 18:31:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18d2a8f8aa34003a5d41c70d0c36a9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beca7bfee4974da8a9ce0881ec10b932 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a93de18551da4e2d8d763812ed516d98 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2439d95993eb6c001a3a8c0e47c5095c0c6b5da4","revisions":{"84d7ff9b28df1464281b93a7db9a039bd18f71b8":{"kind":"REWORK","_number":1,"created":"2024-09-09 13:15:04.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/08/928608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/08/928608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/928608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/928608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/928608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/08/928608/1"}}},"commit":{"parents":[{"commit":"84ab5df84a9de4ad6df1fa31b95b0b9503ec1572","subject":"Merge \"Metadata for Share Export Location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ab5df84a9de4ad6df1fa31b95b0b9503ec1572"}]}],"author":{"name":"Gireesh Awasthi","email":"gawasthi2010@gmail.com","date":"2024-09-09 13:11:10.000000000","tz":-240},"committer":{"name":"Gireesh Awasthi","email":"gawasthi2010@gmail.com","date":"2024-09-09 13:11:10.000000000","tz":-240},"subject":"Fix the remaining review comments from SnapLock patch","message":"Fix the remaining review comments from SnapLock patch\n\nThis is follow-up patch to fix the few left over comment from\npatch https://review.opendev.org/c/openstack/manila/+/926049\n\nCloses-Bug: #2058027\nChange-Id: I275a3680d3ea6b1b1ea69d78d4b7fecb4f61af03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84d7ff9b28df1464281b93a7db9a039bd18f71b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84d7ff9b28df1464281b93a7db9a039bd18f71b8"}]},"branch":"refs/heads/master"},"2439d95993eb6c001a3a8c0e47c5095c0c6b5da4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-10 05:25:02.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/08/928608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/08/928608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/928608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/928608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/928608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/08/928608/2"}}},"commit":{"parents":[{"commit":"84ab5df84a9de4ad6df1fa31b95b0b9503ec1572","subject":"Merge \"Metadata for Share Export Location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ab5df84a9de4ad6df1fa31b95b0b9503ec1572"}]}],"author":{"name":"Gireesh Awasthi","email":"gawasthi2010@gmail.com","date":"2024-09-09 13:11:10.000000000","tz":-240},"committer":{"name":"Gireesh Awasthi","email":"gawasthi2010@gmail.com","date":"2024-09-10 05:23:00.000000000","tz":-240},"subject":"Fix the remaining review comments from SnapLock patch","message":"Fix the remaining review comments from SnapLock patch\n\nThis is follow-up patch to fix the few left over comment from\npatch https://review.opendev.org/c/openstack/manila/+/926049\n\nCloses-Bug: #2079967\nChange-Id: I275a3680d3ea6b1b1ea69d78d4b7fecb4f61af03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2439d95993eb6c001a3a8c0e47c5095c0c6b5da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2439d95993eb6c001a3a8c0e47c5095c0c6b5da4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
