)]}'
{"id":"openstack%2Fmanila~929151","triplet_id":"openstack%2Fmanila~stable%2F2023.2~I33204657401f3eadd490fc50fedaef899dcff297","project":"openstack/manila","branch":"stable/2023.2","topic":"bug/2017501","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-13 21:03:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I33204657401f3eadd490fc50fedaef899dcff297","subject":"Fix export locations update when subnet is added","status":"MERGED","created":"2024-09-12 16:21:01.000000000","updated":"2024-09-13 21:04:19.000000000","submitted":"2024-09-13 21:03:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929151-bug/2017501","cherry_pick_of_change":929093,"cherry_pick_of_patch_set":1,"meta_rev_id":"88860abdd7a26e11dd0038b6b4546f13d79e9c53","_number":929151,"virtual_id_number":929151,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-13 21:03:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-09-13 20:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-09-13 20:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-12 16:21:01.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"state":"CC"},{"updated":"2024-09-12 16:28:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-12 17:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-12 18:03:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-13 20:17:56.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"657427c0e09882c1808f3a68a3a7f5e9b1b7b184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-12 16:21:01.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"cfb3adde8d60626243ffa3a6cedfa2c929ea3722","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-12 16:21:18.000000000","message":"Topic bug/2017501-stable/2023.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"96488d2f10cfb08753c5b8043296924d89b45774","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-12 16:21:22.000000000","message":"Topic set to bug/2017501","accounts_in_message":[],"_revision_number":1},{"id":"467819eea9025090844ed80e321ff71507783e02","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-12 16:28:12.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2aeea7b2b2a349e8a7cf42c1f282945b\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de580d13168b407e82bf8e6c2ebba74f : FAILURE in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1066b664834d82359f25c5f8c7b5e12dffd4350b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 17:44:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4865440618141ce8e549ae94c8a581d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417fdfe64d9e4768a34e050c85a47792 : SUCCESS in 8m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9f4be043738461391b382bdb2a13755 : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9d7b5910ec84794a5c718280f303d92 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf01c8d24fa49019b4fef1ddb10ed2b : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83f803036dd34456b69bb54d442fc374 : SUCCESS in 7m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4aed814881f43379a4a7d3bf13fec86 : SUCCESS in 10m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c70fcb272d734924ae0b8a2520856d65 : SUCCESS in 9m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad3e3eeb0c3a4ef3b287de57d067f580 : SUCCESS in 8m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/399df588273346af8ef0ebefd5d460c3 : SUCCESS in 1h 07m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5bd97a97eaaf40fcb6066fd43c98b95a : SUCCESS in 1h 11m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ef3a15599b08420d860f05651c573513 : SUCCESS in 1h 18m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8eb42cbf97d8421897a12c2331bbd3b6 : SUCCESS in 1h 06m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/84c52222a79541899a8cff0a0911f19d : SUCCESS in 1h 05m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/298e0a0638814ec689007492dbeeb0d0 : SUCCESS in 41m 20s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/10aff72232d34da394ae9cffb65c69da : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b89d86a6ec23479f8b338af255a13816 : SUCCESS in 4m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/506fe093c88c48b983691465560a0fd0 : SUCCESS in 41m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f660f1483ee044babcfe90fd05474572 : SUCCESS in 34m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a8ff078aaf084217b7ffe6b52ce5329f : SUCCESS in 49m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/93191d6d405f410d900965fc4e0e378f : FAILURE in 25m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/00126935d6354780b2a594ae231359ca : FAILURE in 20m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/326bc5ee76854da6970c4689c00faf84 : SUCCESS in 1h 07m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7dc48f70278343f18a57e3d0948b0e76 : SUCCESS in 47m 18s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdabfc6c57db40739e3dcb008159d75d : SUCCESS in 57m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d570beb0165a4fa9be749321fd272a0c : SUCCESS in 39m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a851c692db3e428bac27eba6a6abce5b : SUCCESS in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3618d9e4e324118767a8d6d164888c4bbf4d13d5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-12 18:03:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/929151/1/upstream-check/manila-tempest-plugin-ontap-dhss/0ae5002/ : RETRY_LIMIT in 9m 22s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/929151/1/upstream-check/manila-tempest-plugin-ontap-dhss-rest/614f285/ : RETRY_LIMIT in 9m 22s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/929151/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/0d1b817/ : RETRY_LIMIT in 12m 53s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"cae94cc4afd388857785da61758647ef2fc7b5b9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-13 20:17:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a543382b2a56b1af4da5ae13eb4b9e42dabd5daa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 20:18:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16e91866c2a6b0b2030c3eed433022860811898e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 21:03:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6268095481044f9ad1aaf4cc80b17bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/192e4b6f004440f69f7d3216df06550d : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0601582f02b940259356031c0818711a : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7799ac86309402192522be6ff7e3afb : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a085d14f1774eac99051bbc779d119d : SUCCESS in 10m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2089936a7b64203829a1dc64f8d8091 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8149753f4afa4b8cb6ac19f346d24ae6 : SUCCESS in 9m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2ca20f2596b1487c90bac340df6d3611 : SUCCESS in 44m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d6ffec21ae554f6b84fc1b7dd9b3a6f4 : SUCCESS in 26m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1eff21475b7543359a5092818b6980b4 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4891edbdeafe9e261199eab687044c551ae44b2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 21:03:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88860abdd7a26e11dd0038b6b4546f13d79e9c53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 21:04:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c527f9871d74659ac00ed39bd7b9383\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b989f9984eb446b382ba22c798a9db50 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/866f1dc5f5f04ead8e35894008732dea : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09ec18ad87b68b205c47fc56002f18a0c4fdd393","revisions":{"09ec18ad87b68b205c47fc56002f18a0c4fdd393":{"kind":"REWORK","_number":1,"created":"2024-09-12 16:21:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/929151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/51/929151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/929151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/929151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/929151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/51/929151/1"}}},"commit":{"parents":[{"commit":"f509be3364eeba9be1c8f69a2dc03beb0a0006af","subject":"Create ports as disabled with external networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f509be3364eeba9be1c8f69a2dc03beb0a0006af"}]}],"author":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2023-05-03 08:49:12.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-12 16:21:01.000000000","tz":0},"subject":"Fix export locations update when subnet is added","message":"Fix export locations update when subnet is added\n\nExport locations of active share replicas was updated\nbut not those of the share itself.\n\nCloses-Bug: #2017501\nChange-Id: I33204657401f3eadd490fc50fedaef899dcff297\n(cherry picked from commit 07ce64462c63c5253226a733f5686e78eb7223a1)\n(cherry picked from commit 5b77e4756104554fe024d086241e81e46d71d1ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/09ec18ad87b68b205c47fc56002f18a0c4fdd393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/09ec18ad87b68b205c47fc56002f18a0c4fdd393"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
