)]}'
{"id":"openstack%2Fmanila~929152","triplet_id":"openstack%2Fmanila~stable%2F2023.1~I33204657401f3eadd490fc50fedaef899dcff297","project":"openstack/manila","branch":"stable/2023.1","topic":"bug/2017501","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-09-20 20:21:34.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-20 21:13:59.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-09-20 21:13:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I33204657401f3eadd490fc50fedaef899dcff297","subject":"Fix export locations update when subnet is added","status":"MERGED","created":"2024-09-12 16:21:43.000000000","updated":"2024-09-20 21:15:07.000000000","submitted":"2024-09-20 21:13:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929152-bug/2017501","cherry_pick_of_change":929151,"cherry_pick_of_patch_set":1,"meta_rev_id":"fdccb5b6ef37faadc9674ed46aca2d1caf7c7371","_number":929152,"virtual_id_number":929152,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 21:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-09-20 20:21:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-09-20 20:21:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-12 16:21:43.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"state":"CC"},{"updated":"2024-09-12 16:41:22.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-12 18:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-12 18:42:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-13 21:05:13.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-09-13 21:05:13.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"583230050a5e645b4633a261c91816f133181c64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-12 16:21:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"34bc3437389c8ed51e65255a52fdf190e6147984","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-12 16:22:01.000000000","message":"Topic bug/2017501-stable/2023.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"d70017b48b05dc5e7f26c8b261be5b184a41ee29","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-12 16:22:05.000000000","message":"Topic set to bug/2017501","accounts_in_message":[],"_revision_number":1},{"id":"b5a589fea4d8e4fe468c7ad42a0aa4d5bf43cc0c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-12 16:41:22.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/924f660eb683458fb40dd41ad43bf1f8\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b8bc0cb98b9a4e08bb9225933485d5ac : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6ef90a860b451f38edf3a3d12f851ad466ffd431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 18:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc0d561f78794ac48fdfda8e7112e528\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907a52b5a565448b8b2b4a2d77c3e67c : SUCCESS in 9m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2f777d9683e4218b1c3f039392f5bf1 : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78a1707d23d4881b80ca2f8c92c5c7f : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2a50284d810402f867cf1c66ef7872b : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/373918a1365346c9b35763bae6ef957a : SUCCESS in 8m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/de7abaf5ed7147a6ac3bf3789f6dedca : SUCCESS in 38m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dcf065451d0943c08d191eb44a41d4d5 : SUCCESS in 1h 59m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c7ab11511ee54dcd8fef63a7519db698 : SUCCESS in 1h 22m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a55100ac975f453e938839abaac8a063 : SUCCESS in 54m 10s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0eceb539dea04e5e8714ee7dc9721b79 : FAILURE in 10m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2da0850d6d6f47d9b39eda793ec6f661 : FAILURE in 1h 10m 35s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ceae4b374184868b5c2166deeafdf90 : SUCCESS in 4m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e642eed35cd4776826d03c93689ec37 : SUCCESS in 4m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/096629d80d46422d893581d235939fb6 : SUCCESS in 44m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7bbd16cee9524ee3ba5e98c32fe94d71 : SUCCESS in 45m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/081c4fcba7164d17aa68fd3bcb09c234 : SUCCESS in 47m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/c874d9cc9bfc44d7875d7df5e572d5f8 : FAILURE in 20m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/61931f850e8042b2a3d064d1b06cc64f : FAILURE in 19m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2582d29c108a4a8faa7381105c2e64cf : SUCCESS in 46m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1745e8cb31494e019e6945bd5fcd18e5 : SUCCESS in 58m 35s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e3bacf0288444f73b98c940542eb1a81 : SUCCESS in 49m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7fc2126ee1224fab8a331dc4b015d1d6 : SUCCESS in 42m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/60a47a6a76da4aad9ea3d597c27ab3b2 : SUCCESS in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31f859047a66dc7092ffc70782f4f0625f40b7df","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-12 18:42:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/929152/1/upstream-check/manila-tempest-plugin-ontap-dhss/805fc82/ : RETRY_LIMIT in 9m 59s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/929152/1/upstream-check/manila-tempest-plugin-ontap-dhss-rest/57dd642/ : RETRY_LIMIT in 8m 45s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/929152/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/c9980f2/ : RETRY_LIMIT in 17m 14s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/929152/1/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/cc5a0fb/ : RETRY_LIMIT in 13m 41s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4116534957e092c99c6fa7911efcd09dd8668fe","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-20 20:21:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"001af8accb97416963caf0942bc6974ae9346492","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 20:22:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10471a83e6ccf36e7f287852f4becdd7a235b7e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 21:13:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a87873a5a49244bb8437ae2da0481251\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/615a556e2617430b847a1896ffbe7278 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b27c358d64b415aaf3e360e24b0c99f : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16d3505c812848d3a5cc52ddd0a432ec : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96a6a1d692ed4219bfdb4e3347bf82b9 : SUCCESS in 10m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/efc8b169de684baeaf83e0de7121d01a : SUCCESS in 49m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b6ef56d87162445db5be9758dd4bc3a6 : SUCCESS in 37m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8617c91f78914bda9a194d7679899e45 : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"99218b0634ed3624668a6592b04842ab5663bf23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 21:13:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fdccb5b6ef37faadc9674ed46aca2d1caf7c7371","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 21:15:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daba088da14c4a6fb0f51928910b33a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7485f5b66d254fb6a00534e943dfef24 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/803651d70584434388b51598a5464109 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2944880f07aa31dea5e812ca832a5bb17851ee38","revisions":{"2944880f07aa31dea5e812ca832a5bb17851ee38":{"kind":"REWORK","_number":1,"created":"2024-09-12 16:21:43.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/52/929152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/929152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/929152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/929152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/929152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/929152/1"}}},"commit":{"parents":[{"commit":"58dbf997ea93e5c38b6e7db01a85ae8d9836cd9e","subject":"Create ports as disabled with external networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58dbf997ea93e5c38b6e7db01a85ae8d9836cd9e"}]}],"author":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2023-05-03 08:49:12.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-12 16:21:43.000000000","tz":0},"subject":"Fix export locations update when subnet is added","message":"Fix export locations update when subnet is added\n\nExport locations of active share replicas was updated\nbut not those of the share itself.\n\nCloses-Bug: #2017501\nChange-Id: I33204657401f3eadd490fc50fedaef899dcff297\n(cherry picked from commit 07ce64462c63c5253226a733f5686e78eb7223a1)\n(cherry picked from commit 5b77e4756104554fe024d086241e81e46d71d1ed)\n(cherry picked from commit 09ec18ad87b68b205c47fc56002f18a0c4fdd393)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2944880f07aa31dea5e812ca832a5bb17851ee38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2944880f07aa31dea5e812ca832a5bb17851ee38"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
