)]}'
{"id":"openstack%2Fmanila~930642","triplet_id":"openstack%2Fmanila~stable%2F2023.1~I5e86d09ea5ba8e61ecf04c52b25c639dfc823737","project":"openstack/manila","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-10-10 18:03:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5e86d09ea5ba8e61ecf04c52b25c639dfc823737","subject":"Manage/unmanage for CephFS drivers","status":"ABANDONED","created":"2024-09-26 18:44:34.000000000","updated":"2024-10-10 18:03:14.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d848c1c6b4226603f8c6de2eea1565ff018de02b","_number":930642,"virtual_id_number":930642,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-26 20:38:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":-1,"date":"2024-09-26 18:44:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 18:58:30.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-26 20:38:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 20:58:41.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"957f30a99539bc3ee27444e330bd53b34172c54e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-26 18:44:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef1b7cc3f60dc0e6b26066fc81cb7ca3c3f17d15","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-26 18:44:48.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b6dd9b31df4f11f80f57f1458bcfb5aac46496c4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-26 18:58:30.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0c21e75f5e964a389c61797ac241f32f\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c720c27ef38b47f8be4132cba53721ca : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f8f924b906d8d5f6f1341b7dcf557bde0e45dc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 20:38:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31a2b889aff54c3ba430e6de33b768c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006792438d1847089333dd151ba35e9e : SUCCESS in 9m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6e0aae0374c48e585a35d4a5de57ff8 : FAILURE in 20m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd32137c6e6e4e0ba80e7788782183c7 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c419f0e3bba446ba744b047c4cec2ae : SUCCESS in 11m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d3e7a19897d04e91b79660d3d370e83f : SUCCESS in 57m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f21ffa4c18f34966b4d8c5aae95475fb : SUCCESS in 1h 47m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/80ae6777e9a74fbea4b6e5e6871b466a : SUCCESS in 1h 13m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6407204f10974047965ec985d8b8f193 : FAILURE in 54m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cefb38ffab29482286ceb5e8b60928b5 : FAILURE in 6m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bc23256b92a549ccaace7a6e14333c2c : FAILURE in 1h 05m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/42294787344848dcb9e60db0b09fcd6c : SUCCESS in 4m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b7a618f44fd4e83947f5714a84ab4a9 : SUCCESS in 5m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3783aff58e384befb1588c0cf0f76508 : SUCCESS in 25m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c1b21d23685344d5b2b8a72f01b7d3cf : SUCCESS in 30m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c6d8952cdcdd477db2cd7b2c24d57e42 : SUCCESS in 40m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/13757ab206df4d2285881995d62835b7 : FAILURE in 12m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3391b80a4b1845158e141a111c59bb78 : FAILURE in 23m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/e455e2c72a6c4883bb24c0d22e9459dd : SUCCESS in 1h 08m 49s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a91fac567a6b4dc78c710f52c6924f01 : SUCCESS in 47m 53s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/72891263e87f4be8acdb4b147c442e78 : SUCCESS in 48m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4b850ec743844228bf53d1fa56aecabd : SUCCESS in 28m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/960285c120f2432197f99adee9b4cef1 : SUCCESS in 41m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50ebacb7b26254a0f3843ac25b0f271390a7db15","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-26 20:58:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/930642/1/upstream-check/manila-tempest-plugin-ontap-dhss/6a94ca5/ : RETRY_LIMIT in 8m 49s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/930642/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/ff62d54/ : RETRY_LIMIT in 8m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6231a8fbd1cd2ccc29c86172523a6c2dd1cb9ad","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-30 12:51:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74dde3bafb271de645473a15a96ba718d20569fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 14:39:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de12856ef00d45d2a82c840ca7d76167\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10141ad8480f4382ad970ab2886a84fb : SUCCESS in 9m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cdfdfab5bff492097de822e819802aa : FAILURE in 21m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79a9c6a984c4a939856e4f82e642b3f : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/958e0553bbad43d7b8c1781b0b55fecf : SUCCESS in 10m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d1ebb0df1db84b48ae23bd5ec2bf1e48 : SUCCESS in 1h 20m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/950d4326f2804215ad9eeea22459bb19 : FAILURE in 1h 09m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eea3566dd75c47a394bd6ff9b30946a1 : SUCCESS in 1h 19m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/03e89803195a431ab25df83d42b64e97 : SUCCESS in 1h 28m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/46acfa9a3591451199acb1691e041268 : SUCCESS in 1h 22m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1c3da5a5794943b8805266ea003dca84 : SUCCESS in 1h 38m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/77f930a3d7a0437aa828d8a144eced4c : SUCCESS in 3m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/532cfebf19ca46d1a90b8a6e22f01843 : SUCCESS in 3m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b4b62d64db5b4da592504b503179c916 : SUCCESS in 21m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bf83e66c77454f83ada3b35308aee265 : SUCCESS in 28m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4d8dcc39210e486eb3de46432960edc5 : FAILURE in 30m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5149163aa9d0411caa373b8026a3f435 : FAILURE in 11m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a8f2f090a2f34566b151318fad625489 : FAILURE in 20m 51s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8b548dc2142e482faf24d44f7114c071 : SUCCESS in 1h 24m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/200c571ed41148239d1d7dd9735b8628 : SUCCESS in 1h 01m 48s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8b71461808514de0a62246fe19b25b69 : SUCCESS in 57m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/86f902a2611b4f6abd89ff1f63b11bd4 : SUCCESS in 39m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5a0c358694aa43c1b0c88b8d78afe99e : SUCCESS in 40m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd751b80ed47703d511dcccd2346c9aa090a91ad","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-30 14:56:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/930642/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/ea2ea2a/ : RETRY_LIMIT in 8m 43s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d848c1c6b4226603f8c6de2eea1565ff018de02b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-10-10 18:03:14.000000000","message":"Abandoned\n\nThis change was submitted mostly for testing purposes. Abandoning it for now.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48e025df71c80f35c5f46f4bbf48e1c771a6675b","revisions":{"48e025df71c80f35c5f46f4bbf48e1c771a6675b":{"kind":"REWORK","_number":1,"created":"2024-09-26 18:44:34.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/42/930642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/42/930642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/930642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/930642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/930642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/42/930642/1"}}},"commit":{"parents":[{"commit":"2944880f07aa31dea5e812ca832a5bb17851ee38","subject":"Fix export locations update when subnet is added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2944880f07aa31dea5e812ca832a5bb17851ee38"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-23 18:30:47.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-09-26 18:44:22.000000000","tz":-180},"subject":"Manage/unmanage for CephFS drivers","message":"Manage/unmanage for CephFS drivers\n\nImplement manage/unmanage for shares and snapshots to the CephFS\ndrivers.\n\n- Instead of providing the export locations, users will need to\nprovide the subvolume/subvolume snapshot name in the\nexport/provider location parameter.\n\n- The size of the snapshots will follow the shares\u0027 size, as we\ncan not currently be sure of the real size of the snapshot. This\nmust be fixed after the issue is addressed within the ceph\nbackend [1].\n\n- We will attempt to resize subvolumes whose size are \"infinite\"\nto the size value provided within the driver_options parameter.\n\n- Start storing the provider location of the snapshots after\ntheir creation, so it can be reused later.\n\n[1] https://tracker.ceph.com/issues/67215#change-270669\n\nPartially-Implements: bp manage-unmanage-cephfs-shares-and-snapshots\nChange-Id: I5e86d09ea5ba8e61ecf04c52b25c639dfc823737\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/48e025df71c80f35c5f46f4bbf48e1c771a6675b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/48e025df71c80f35c5f46f4bbf48e1c771a6675b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
