)]}'
{"id":"openstack%2Fmanila~930643","triplet_id":"openstack%2Fmanila~stable%2F2024.1~Id2214412fe75456cdb52edb5ff261d32b063655f","project":"openstack/manila","branch":"stable/2024.1","topic":"bug/2053100","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-10-03 12:01:48.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-10-03 13:40:15.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-10-03 13:40:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2214412fe75456cdb52edb5ff261d32b063655f","subject":"Fix \"preferred\" export locations in the CephFS driver","status":"MERGED","created":"2024-09-26 19:05:59.000000000","updated":"2024-10-03 13:41:20.000000000","submitted":"2024-10-03 13:40:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930643-bug/2053100","cherry_pick_of_change":930602,"cherry_pick_of_patch_set":1,"meta_rev_id":"27469946ef65d9522292eb00c41a66124817851e","_number":930643,"virtual_id_number":930643,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 13:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-10-03 12:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-10-03 12:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 19:13:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-26 21:27:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 21:47:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-30 21:39:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-09-30 21:39:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"}],"messages":[{"id":"f6a29ac48f5412f77fbbde15c32feded6b60f166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-26 19:05:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"ceb1b06c567488bef631d53d1b10add79dd88b75","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-26 19:06:21.000000000","message":"Topic bug/2053100-stable/2024.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"6820de46a30d3bb208b96c6f1179443351c7440d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-26 19:06:25.000000000","message":"Topic set to bug/2053100","accounts_in_message":[],"_revision_number":1},{"id":"0fa2a5fe8a12c7899541384876b589e6f6bdda8a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-26 19:13:06.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ff7b897012ef4c8e8f25e665db59ff00\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/63f0067ab5b44711b60c300c59e2bc6f : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"88816011624e239debdfe9a7a75555dddcfaf24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:27:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c14f88558e184277a3c3700f28a1a357\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f2017d355f43e0ac2c2c9eec835867 : SUCCESS in 12m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/439f69aba7dd4cb3ad55374e0a3a899f : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d597d0f79bc4ed2bc55869bcb2c557b : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e0068246734879a43ed6d3d6fcb3d7 : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51a0969ac94b4d9ca4d9c1d0e5121de4 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/270192e5ee0c47e4b9d7309e13e17f40 : SUCCESS in 9m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/054876977bd34f588235492e07304f9e : SUCCESS in 1h 07m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6f8097c10ace4cd29800fcda6c6af664 : SUCCESS in 2h 13m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b23ed23db69d48f5bef748710c850177 : SUCCESS in 1h 03m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0f9e4778758847b49130ee0aaa249e98 : SUCCESS in 1h 01m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/528ddccbd4014629bb26488b65847667 : FAILURE in 10m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fb775ce0167341a789a14ff4e55b8b7f : FAILURE in 55m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/341f8dac0e564464aa0633f04344b4bd : SUCCESS in 3m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3fb3056ef5e044939c802a3e70291656 : SUCCESS in 4m 04s (non-voting)\n- manila-tox-py311-with-sqlalchemy-20 https://zuul.opendev.org/t/openstack/build/6a0141d89cb641f8aff3186b83f23226 : SUCCESS in 11m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3b4f447571bc432b9af26ef898928ba3 : SUCCESS in 26m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c45b8f1667c04c1a8cd0a4484ff3c422 : SUCCESS in 33m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/df61e0f7183f49efbd4426b4d5777434 : FAILURE in 26m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ac7c3779df8541f3a955f67ad9f3a1c5 : SUCCESS in 55m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/660c2b06407540ffa91b1cc206ce8ab1 : SUCCESS in 48m 00s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/951843f66c9f498c87fbe991f08c4752 : SUCCESS in 1h 01m 31s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a385dd2fb45d4c198b6f2df5b18606ac : SUCCESS in 45m 50s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2c53b0eabd37425a81a1c6396cd336a5 : SUCCESS in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3ba18fd72144a6696d739a40f169155f928527e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-26 21:47:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/930643/1/upstream-check/manila-tempest-plugin-ontap-dhss/a3d2db4/ : RETRY_LIMIT in 8m 41s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/930643/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/57be6e7/ : RETRY_LIMIT in 8m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"657150dcdadf2bfc3c30f3f36ce91fb7d7f2ecc4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-26 22:54:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"121c9f21cdc58b8e3520e5c8785388260446d6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 00:26:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9a26f47551b4fe3bd57e2783d431cef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03b957173c645078e9e1022755e15ed : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e574b85253bb4362a3e15ced52c67bfd : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c28a5458ce41fe8552aa242f10982d : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e326f763a854a6b96a4f8b05e7998b1 : SUCCESS in 9m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33ad42335b674f46995f6fc407d6f8d1 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbb1e5a1ca1248bc94b0f9912a27c7e0 : SUCCESS in 9m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a90cc7877dc40ddba8e3450a3f9fb1b : SUCCESS in 38m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1c2a05740a30440fa31ee2fe21ca7a17 : FAILURE in 1h 30m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c5cacea615594003afea97e966dda0a0 : SUCCESS in 50m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/29c3e9e78e38464887b36538b2829158 : SUCCESS in 41m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e2a74d68a7724466873813e0b3a31be5 : FAILURE in 6m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/243f7dfc0db645568228d3866e961727 : SUCCESS in 52m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8acf36ab5e06441ba8b505fa15349505 : SUCCESS in 3m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/878b542a18524fc1bf17a57b383fabf9 : SUCCESS in 4m 37s (non-voting)\n- manila-tox-py311-with-sqlalchemy-20 https://zuul.opendev.org/t/openstack/build/8fa8eb2aa7fd4db4be4eea9723ab6ac4 : SUCCESS in 8m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5381ef178e8743ac8b80b247d9bcf35b : SUCCESS in 21m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4cdde149fbde4888bbd0196a8e320d71 : SUCCESS in 36m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/212e2002f3804fa38258f01b2012887f : SUCCESS in 54m 42s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/3d4901d0e31a4f4abe175cea607e369c : SUCCESS in 48m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9dc4b2e4779c4faf9d5ba76bbbc18c3a : SUCCESS in 55m 28s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db7a3b92019143bb8f95c96e9b901bab : SUCCESS in 56m 27s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/29474312658e4a179e01f2ea73276992 : SUCCESS in 23m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ba89d00a0f284c48a5ab24926cc87467 : SUCCESS in 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fcfcdaf1009869033fe8be517267cb00507cb33","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-27 00:42:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/930643/1/upstream-check/manila-tempest-plugin-ontap-dhss/cfe021d/ : RETRY_LIMIT in 8m 40s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"88b9d4f50c9c04433ba62d629ca4b28e31ccdf88","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-10-03 12:01:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49eca0e151ccb2f29e2da2471d15f659556928bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 12:02:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e414181eaf58c080320a64b80a37cc445fa3de53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:40:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1be349de8224161b775971337134a35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3e8b3cde9849e4aeba75d2209056d5 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe127ae55dd74314bbb08c88ecc7881e : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/377eeef3833a41839473412fa253808b : SUCCESS in 13m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90c10530efed457381b411f2bf1f5b61 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa499cbec855440ebe7b31e33cf1a875 : SUCCESS in 8m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/db735285b9d945e493ca3922dbd45b94 : SUCCESS in 1h 32m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/42e5890e25ca42949bfa174984cacad5 : SUCCESS in 34m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2feaa628c88741b9ad6ccfab24398c38 : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":1},{"id":"303dc0b2534ce68ca501cadbd521c92095ae3000","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:40:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"27469946ef65d9522292eb00c41a66124817851e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:41:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a4b9be0321469f9cf22ba43a5311ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf41399d0c2745b49296b75e61fd0139 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a44bd816289f4dbaa3502f6ed30b623f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06da39f3733df0ae9b6d630d60aae2f0389af3b6","revisions":{"06da39f3733df0ae9b6d630d60aae2f0389af3b6":{"kind":"REWORK","_number":1,"created":"2024-09-26 19:05:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/43/930643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/43/930643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/930643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/930643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/930643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/43/930643/1"}}},"commit":{"parents":[{"commit":"983189916aaf9f68113d3c4958f81fde7a71580a","subject":"Fix share/snapshot show for deferred deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/983189916aaf9f68113d3c4958f81fde7a71580a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-12 05:23:19.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-26 19:05:59.000000000","tz":0},"subject":"Fix \"preferred\" export locations in the CephFS driver","message":"Fix \"preferred\" export locations in the CephFS driver\n\nThe CephFS driver used to set the \"preferred\"\nexport path by sending back an export location\nmodel update; however, the \"preferred\" keyword\nis expected in the export location\u0027s metadata,\nnot in the export location dict.\n\nChange-Id: Id2214412fe75456cdb52edb5ff261d32b063655f\nCloses-Bug: #2053100\n(cherry picked from commit 17088aa57af3ae2dd5a16e2e8558ec471d2c1cc5)\n(cherry picked from commit 620933cdb98f3560a480923ba096a6215d7d0edd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06da39f3733df0ae9b6d630d60aae2f0389af3b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06da39f3733df0ae9b6d630d60aae2f0389af3b6"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
