)]}'
{"id":"openstack%2Fmanila~930644","triplet_id":"openstack%2Fmanila~stable%2F2023.2~Id2214412fe75456cdb52edb5ff261d32b063655f","project":"openstack/manila","branch":"stable/2023.2","topic":"bug/2053100","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-10-03 14:56:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2214412fe75456cdb52edb5ff261d32b063655f","subject":"Fix \"preferred\" export locations in the CephFS driver","status":"MERGED","created":"2024-09-26 19:06:44.000000000","updated":"2024-10-03 14:57:40.000000000","submitted":"2024-10-03 14:56:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930644-bug/2053100","cherry_pick_of_change":930643,"cherry_pick_of_patch_set":1,"meta_rev_id":"7effa6e0911b19bf2766af0781605ed9f8dc33da","_number":930644,"virtual_id_number":930644,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 14:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-10-03 13:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-10-03 13:29:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 19:18:15.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-26 20:42:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 20:59:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-03 13:29:22.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"b107747af327ff6ce47167bc0cbbb57e3587006f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-26 19:06:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"497cd72592a00c273f2186b3a4405225dbe1fc11","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-26 19:07:05.000000000","message":"Topic bug/2053100-stable/2023.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"e54e5b2ce4e007800d0c311a5817fca50d29611d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-26 19:07:09.000000000","message":"Topic set to bug/2053100","accounts_in_message":[],"_revision_number":1},{"id":"bb829497679c76bf65bb44deb7bc2617e4fee20d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-26 19:18:15.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9686b0d2254c47b7b428e4f8ac87bf4a\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c622efd73004a90a5e611b34cdadc90 : FAILURE in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"beace5408944d1ac847eb537cbfa54c01e24b28f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 20:42:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c6c9ddcf99b40f386017eb0e4969e48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07eff60728b64b679b30a96915b94001 : SUCCESS in 8m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d94796c7824e4296b07286a632beb480 : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6461201b9c2c438d8a3242d18ac98475 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e9e045ba64a45029654d0a294a8120b : SUCCESS in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/075d2db7aa0346ab90db13e0d62131c5 : SUCCESS in 10m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2889242f506f4fd6a9c4e83ebd0fd23a : SUCCESS in 10m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41c8818429c64b009b8b833e0be3abb7 : SUCCESS in 7m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17909019a27945c7a80b1b4ba753ca09 : SUCCESS in 9m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/82383af61d694175b03e24cdcd43eb15 : SUCCESS in 39m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dee1a24fbc4f4c1b9a9c03f3adc306ae : SUCCESS in 1h 34m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4b4cf802cbc49a0a96684f334912612 : SUCCESS in 55m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/27dba2a109f84c458c56773f618719f8 : SUCCESS in 36m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/28d1413dd0c5436eb3103b53cc4f75b9 : SUCCESS in 1h 03m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/537c1dcf10f54a70ab56b7c26c2aa626 : SUCCESS in 1h 15m 58s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a9c1789c69b44ac88b2ea60061b22276 : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b8d8d79498cb4ae395a3d4ff0adfe7a5 : SUCCESS in 5m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a63f7c723cc74e688404079152d0e3d2 : SUCCESS in 34m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a528050eb6154e8aad83691185c82f42 : SUCCESS in 37m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/019d1504285a4378ae451795b0664893 : SUCCESS in 44m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4e6aa51c9ef747a39d7ae4bad2ff3831 : FAILURE in 19m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f48105c4fbd942c4ad79c8bba295731e : FAILURE in 21m 59s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/18edb3cec1224357bb24b246c8db6871 : SUCCESS in 1h 04m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5e7e858e6a7f44e7997a8e1b38f2b125 : SUCCESS in 58m 43s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f6d3c3d0f20a4433919091e9d1d77b36 : SUCCESS in 1h 01m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6327fe3bf3a04e6bbd362a382d9cb837 : SUCCESS in 44m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6cad0e753f584d44892394a5ac4733f8 : SUCCESS in 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2db1c399946f4622911c8f8325732b3f10345384","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-26 20:59:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/930644/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/9294ad4/ : RETRY_LIMIT in 8m 58s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"cb46f485a641bbb2b5ca26f01374775fc972d1c4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-10-03 13:29:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88e913a5df42801f4fe92ef36b266a0abe778fba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:29:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b3d526e1da837dc7ef126a1c0ca329a0d285159","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:56:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04fc3fcb9bff4baeba65a7888446dfc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1ba31c93484ad696c20dd22c65ee23 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e18dc8526dee4b63be83a6fe7c56200a : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34bbafb1e4ae48edaeb52477e6831047 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc690fd7e1444f7bda3ff935f97a03b : SUCCESS in 9m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/078439f34752415ab090da22b47c7d0d : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33fa63c45e464033ba6df2db9f2e492a : SUCCESS in 9m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/178a102098aa46a7913c6ec2acf5a8e7 : SUCCESS in 1h 17m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/98e3871c766445dcacea1c1839c0a8ba : SUCCESS in 28m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0f9e0ad977d14ddfa5db20d61c62035c : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5955fad888b2b06e189524b9f66205e82cc605d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:56:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7effa6e0911b19bf2766af0781605ed9f8dc33da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:57:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6f79e1de492454e8e82ac5311a179b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4e107d68724037b92275c0dfecc23b : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eea8d98f6edc481d8c1c6300cfb694c0 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90c390013a61228341f59c46b173f2b78293ce58","revisions":{"90c390013a61228341f59c46b173f2b78293ce58":{"kind":"REWORK","_number":1,"created":"2024-09-26 19:06:44.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/44/930644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/930644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/930644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/930644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/930644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/930644/1"}}},"commit":{"parents":[{"commit":"7933e6cd80c1ad4a74a3cc7852eb862d673b800c","subject":"Merge \"Ensure password values are not logged\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7933e6cd80c1ad4a74a3cc7852eb862d673b800c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-12 05:23:19.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-09-26 19:06:44.000000000","tz":0},"subject":"Fix \"preferred\" export locations in the CephFS driver","message":"Fix \"preferred\" export locations in the CephFS driver\n\nThe CephFS driver used to set the \"preferred\"\nexport path by sending back an export location\nmodel update; however, the \"preferred\" keyword\nis expected in the export location\u0027s metadata,\nnot in the export location dict.\n\nChange-Id: Id2214412fe75456cdb52edb5ff261d32b063655f\nCloses-Bug: #2053100\n(cherry picked from commit 17088aa57af3ae2dd5a16e2e8558ec471d2c1cc5)\n(cherry picked from commit 620933cdb98f3560a480923ba096a6215d7d0edd)\n(cherry picked from commit 06da39f3733df0ae9b6d630d60aae2f0389af3b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/90c390013a61228341f59c46b173f2b78293ce58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/90c390013a61228341f59c46b173f2b78293ce58"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
