)]}'
{"id":"openstack%2Fmanila~930683","triplet_id":"openstack%2Fmanila~master~I8781359e338ffb44e5f55211bec147000a2097cb","project":"openstack/manila","branch":"master","topic":"bug/2082944","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-10-10 22:28:21.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"35979":{"account":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"last_update":"2024-10-25 21:15:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8781359e338ffb44e5f55211bec147000a2097cb","subject":"Add ability to trace SQL transactions with profiler","status":"MERGED","created":"2024-09-27 08:34:54.000000000","updated":"2024-10-25 21:16:58.000000000","submitted":"2024-10-25 21:15:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930683-bug/2082944","meta_rev_id":"0e2ead0365a2b07e59271f3439f0b0003fb23f5d","_number":930683,"virtual_id_number":930683,"owner":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-25 21:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-10-25 20:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-10-10 22:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-10-25 20:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 09:18:26.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-27 09:26:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-27 10:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-27 10:53:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-27 11:14:47.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-09-27 11:26:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-09-30 21:54:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-10-25 20:02:58.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"46535556b556129e86c6e9563d0662991d8d9d3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"date":"2024-09-27 08:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f29d43dee06a409e4570aa1186f4d7072f9b94f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"date":"2024-09-27 08:45:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3e873091b155a572483c8fc5ab3038387ed863a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 09:18:26.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/2/check/DellEMC_PowerFlex_Manila_dhss_false/5a02014/DellEMC_PowerFlex_Manila_dhss_false/1757 : SUCCESS in 32m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c57d94a9d1e22f020b346579d01689582c0cc7ff","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-27 09:26:12.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3e627bac7ad54e8fbae280a7fd7d110a\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d2902b335cf4455ea4a25f31c5a6b1d8 : FAILURE in 29m 26s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da1fdce277b4405f9bcf64678f67639d : RETRY_LIMIT in 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a58fd13197140174375779a50d0aefd42c97ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 10:36:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b945eac8ed04991a14cc1e49a4bc545\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52fb5f2bb5b44edd87903963980c0d36 : SUCCESS in 10m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef0e5a9e3f749f384e4cb9402ce540c : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9e4f8a576f420cb039ee3c9a37953d : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/038b6e02745d4703b1cfff7fe043bf91 : SUCCESS in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/760ae875aafe4de4a50e74901c6bce43 : SUCCESS in 7m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07a5ebf5d1e047a8b130f156c328169a : SUCCESS in 8m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd3f2b55865b46b3863e6b13cc6e3097 : SUCCESS in 9m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a58cc0e867f0460f810cbd6f658c0886 : SUCCESS in 39m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5dc18e9703b54ddeb31eab65b8e234ae : SUCCESS in 1h 48m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9e167d6ea3a741ce896de4314775f399 : FAILURE in 1h 18m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/843f1f46cecd49f8ac2c6f4e1e977d98 : SUCCESS in 59m 48s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a29d259d5eb140138f7bcae884747ba5 : SUCCESS in 1h 03m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e10cb854a37e4b42b1f6f103bc71126e : FAILURE in 18m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8ac28a21fe754c07a47a91b40127cc49 : SUCCESS in 3m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0ab42ee66a34590bf979c51c08003af : SUCCESS in 4m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9731408833ad4a0eabd20de1820ab127 : SUCCESS in 36m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07bb451010e94b6a8535b586d9cbbd7a : SUCCESS in 36m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f178186633b44cbf98c83db82c430f25 : FAILURE in 26m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5e56fc1f7a9f46948a995b7e435bfdb5 : SUCCESS in 1h 14m 49s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f7a9d44a205459aba115ed4c2d92120 : SUCCESS in 1h 01m 53s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1cb6526abd16469b868246a4d81cf162 : SUCCESS in 37m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d6bb67e7d1e343d4a12d6215a140a0cf : SUCCESS in 39m 29s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/26fc93d288c9409498eadc06516e4458 : FAILURE in 4m 28s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f060200aade84a4c9ec0248feb23a2f8 : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77dd62b0853d34e23378360115916b397d10c182","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-27 10:53:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/930683/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/1667e8c/ : RETRY_LIMIT in 8m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2204a5286f823ea258485f2dd1486af7a6c62bb7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 11:09:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/2/check/DellEMC_PowerFlex_Manila_dhss_false/956efd4/DellEMC_PowerFlex_Manila_dhss_false/1759 : SUCCESS in 33m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"389cac0a143304d863f8f29ac9a1d62fd6399e16","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-27 11:14:47.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/2/check/DellEMC_PowerStore_Manila_dhss_false/6901ee5/DellEMC_PowerStore_Manila_dhss_false/516 : SUCCESS in 38m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"28c61ebc5d84ddde63d21cabc580cd9b856f36fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-27 11:26:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/83/930683/2/check/DellEMC_Unity_Manila_dhss_true/39dc9a4/DellEMC_Unity_Manila_dhss_true/1746 : FAILURE in 13m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/2/check/DellEMC_Unity_Manila_dhss_false/52e5f86/DellEMC_Unity_Manila_dhss_false/1849 : SUCCESS in 1h 19m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d46c6af20d5355d630ad8e4d7d2279b226e42709","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-30 21:54:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"76f4354d607c0f6b28f3cc68d3b8f11fc5414516","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"date":"2024-10-10 12:46:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"68264a2055d5b073668486504d2303638620d8ca","author":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"date":"2024-10-10 12:49:54.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"55928c488006b186fc2f6334c45b5ee76b3a2137","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-10 13:15:12.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/777a715f445f4d539556e15ccae3ca3b\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e573297d18e4c6c9bf895a26ba48460 : FAILURE in 27m 56s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/125359830cf84314b02b24eb22ac875f : RETRY_LIMIT in 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f72aea3e36dbb8d588246887a110bcb36b3ffc21","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-10 13:40:13.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/3/check/DellEMC_PowerFlex_Manila_dhss_false/1615b8b/DellEMC_PowerFlex_Manila_dhss_false/1851 : SUCCESS in 32m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"d27bcaa7318489ac90e35ff5e55532d84197caf9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-10 13:45:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/83/930683/3/check/DellEMC_Unity_Manila_dhss_true/4f2c529/DellEMC_Unity_Manila_dhss_true/1834 : FAILURE in 13m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/3/check/DellEMC_Unity_Manila_dhss_false/03196a9/DellEMC_Unity_Manila_dhss_false/1937 : FAILURE in 34m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b8abf7c6d3ed5341baecc4776ad854770660263c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:18:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab835d2c5b04c23afaa00676a5a20b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd482cf07cf4494d93987bb40c5f2db9 : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47174478d3c64d88b8bf8b5d94a10520 : SUCCESS in 29m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e351e26dca3746b0ac8f3eece9edaeed : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/751b52666d1e4624a9954b77bcbb3f0c : SUCCESS in 12m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/203228d9fd964a83ac3165a873443dbc : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4690b8bb8240467bb86e300cf04a6a8b : SUCCESS in 12m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/84a8eb56988f4daca625ca836a04aec8 : SUCCESS in 1h 13m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f436914dfe684f2d89be1d9eeb1bf584 : FAILURE in 2h 25m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/70f391dd46db407bacd0faefcac81c11 : FAILURE in 1h 18m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/82e39968b6764cf19767216d003db2c2 : SUCCESS in 1h 07m 10s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6ff02530ef0b429099f3a1c662e1219e : SUCCESS in 1h 01m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/575985c693544e4ba82823e4c0ec8fe9 : FAILURE in 21m 44s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e4e615e440dc44059d24f5a53f8b51bd : SUCCESS in 4m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/050bd59a0b8d48459d163c6f0faf78a6 : SUCCESS in 4m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c229d6774fff4ef1b45d0229b5061c1d : SUCCESS in 31m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c90ec0c0735946d1bbf145c26bebcb5e : SUCCESS in 31m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/42f64991f21244b3b4a8e7c33c9ca300 : FAILURE in 18m 30s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0b1752489e0047f1ad27cd45fe3e906b : SUCCESS in 1h 20m 28s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18d9dbceec5f40efa1c448b570facd77 : SUCCESS in 1h 05m 15s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d92f9a9875ff4b58afefc16c5cef8b42 : SUCCESS in 27m 40s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/171009a3210a479e91fd1318c52adc49 : SUCCESS in 30m 59s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b8cb9190b94741efbce4abed886b2737 : FAILURE in 4m 18s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d078b078629149f7b539b5cb250b8249 : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eb7e9cf0d61922730873bd413aa7c824812e8c70","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-10 15:31:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/930683/3/upstream-check/manila-tempest-plugin-ontap-dhss/24d12ae/ : RETRY_LIMIT in 4m 58s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/930683/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/46d378f/ : RETRY_LIMIT in 5m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"28aee12fc7af2ed88d9962992137fd06c3903b17","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-10 15:51:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/3/check/DellEMC_PowerFlex_Manila_dhss_false/9819317/DellEMC_PowerFlex_Manila_dhss_false/1854 : SUCCESS in 32m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"ab7f818f3aaa320925662e55525216f26f117dd3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-10 15:53:53.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/3/check/DellEMC_PowerStore_Manila_dhss_false/f0217bf/DellEMC_PowerStore_Manila_dhss_false/543 : SUCCESS in 35m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"67aece20f4ceb16ad88ab55fdd4e0837bea3ac1c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-10 15:54:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/83/930683/3/check/DellEMC_Unity_Manila_dhss_true/30268e0/DellEMC_Unity_Manila_dhss_true/1837 : FAILURE in 14m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/83/930683/3/check/DellEMC_Unity_Manila_dhss_false/40eaaa2/DellEMC_Unity_Manila_dhss_false/1940 : FAILURE in 35m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b4eb083be0dc6300008a34cebf52fa07430145aa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-10 22:28:21.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5c5e8362ef085189156aec78913c36fa4cdb17a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-10-25 20:02:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c3e857d451faddbc9fe19c99a1252cc365ec1df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 20:03:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e6a9028a11a3246be279ac56a28e2658b13e4f7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 21:15:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d82a677e2e44b8d8baa069f148fe354\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b200a2a6b6f048dea7ce78a3cf4d8ef0 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee8f4d1b25241c2a909ab30b265ca65 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7111a6a25fa947e6b1f7fe647da04c6e : SUCCESS in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57ad04cf944c49cabac106874348c98c : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59cfa89630724ecfba50822fd72709e4 : SUCCESS in 9m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/069ac7f588144a7bba751dc356b5f87b : SUCCESS in 1h 07m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/de803b9073354cecaa0db885a3f3207c : SUCCESS in 38m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00cc9ebfda5b4f289d2c28a792fce788 : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ac1e43c160136908dede71e165df26c73c5985d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 21:15:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0e2ead0365a2b07e59271f3439f0b0003fb23f5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 21:16:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0458114d9f684a30808660d8c0424626\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2774d543a28044359c5bee4d1f5490ff : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c59031aa9e14a2bb136898fdfb52f81 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b796905a595f06df6cf3cece420b164dd8bb429","revisions":{"326358c608f05fa793bbd1aaa4b54bb5916e8638":{"kind":"REWORK","_number":1,"created":"2024-09-27 08:34:54.000000000","uploader":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"ref":"refs/changes/83/930683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/83/930683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/83/930683/1"}}},"commit":{"parents":[{"commit":"dca04e2270a745e73226adbdf80f740f002bd133","subject":"Merge \"Fix \"preferred\" export locations in the CephFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dca04e2270a745e73226adbdf80f740f002bd133"}]}],"author":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2024-09-27 08:31:25.000000000","tz":120},"committer":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2024-09-27 08:31:25.000000000","tz":120},"subject":"Add ability to trace SQL transactions with profiler","message":"Add ability to trace SQL transactions with profiler\n\nCloses-Bug: #2082944\nChange-Id: I8781359e338ffb44e5f55211bec147000a2097cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/326358c608f05fa793bbd1aaa4b54bb5916e8638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/326358c608f05fa793bbd1aaa4b54bb5916e8638"}]},"branch":"refs/heads/master"},"197c2df25c6b5a4eff6c12dcc633cc1d5d4fe441":{"kind":"REWORK","_number":2,"created":"2024-09-27 08:45:41.000000000","uploader":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"ref":"refs/changes/83/930683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/83/930683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/83/930683/2"}}},"commit":{"parents":[{"commit":"dca04e2270a745e73226adbdf80f740f002bd133","subject":"Merge \"Fix \"preferred\" export locations in the CephFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dca04e2270a745e73226adbdf80f740f002bd133"}]}],"author":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2024-09-27 08:31:25.000000000","tz":120},"committer":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2024-09-27 08:45:22.000000000","tz":120},"subject":"Add ability to trace SQL transactions with profiler","message":"Add ability to trace SQL transactions with profiler\n\nCloses-Bug: #2082944\nChange-Id: I8781359e338ffb44e5f55211bec147000a2097cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/197c2df25c6b5a4eff6c12dcc633cc1d5d4fe441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/197c2df25c6b5a4eff6c12dcc633cc1d5d4fe441"}]},"branch":"refs/heads/master"},"6b796905a595f06df6cf3cece420b164dd8bb429":{"kind":"REWORK","_number":3,"created":"2024-10-10 12:46:43.000000000","uploader":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"ref":"refs/changes/83/930683/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/83/930683/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/83/930683/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/83/930683/3"}}},"commit":{"parents":[{"commit":"dca04e2270a745e73226adbdf80f740f002bd133","subject":"Merge \"Fix \"preferred\" export locations in the CephFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dca04e2270a745e73226adbdf80f740f002bd133"}]}],"author":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2024-09-27 08:31:25.000000000","tz":120},"committer":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2024-10-10 12:42:05.000000000","tz":120},"subject":"Add ability to trace SQL transactions with profiler","message":"Add ability to trace SQL transactions with profiler\n\nCloses-Bug: #2082944\nChange-Id: I8781359e338ffb44e5f55211bec147000a2097cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b796905a595f06df6cf3cece420b164dd8bb429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b796905a595f06df6cf3cece420b164dd8bb429"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
