)]}'
{"id":"openstack%2Fmanila~930706","triplet_id":"openstack%2Fmanila~master~Ica7ac704ae358d0271562d3c02b4b56dfb7ded7f","project":"openstack/manila","branch":"master","topic":"remove-eventlet-workaround","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-11-19 17:32:54.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2024-11-19 19:05:33.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-11-19 19:05:33.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-19 19:05:33.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-11-19 19:05:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica7ac704ae358d0271562d3c02b4b56dfb7ded7f","subject":"Remove workaround for eventlet bug #592","status":"MERGED","created":"2024-09-27 14:15:11.000000000","updated":"2024-11-20 00:38:39.000000000","submitted":"2024-11-19 19:05:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930706-remove-eventlet-workaround","meta_rev_id":"c12a6c660eeea3f9fd10f09a1b09f429a6950a46","_number":930706,"virtual_id_number":930706,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 19:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-11-19 17:32:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-10-02 19:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-09-28 05:54:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-11-19 17:32:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 14:40:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-09-27 14:42:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-27 15:21:45.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-27 16:23:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-09-27 17:56:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-27 18:11:20.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-27 18:32:51.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-10-02 19:53:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-10-02 19:57:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2024-10-02 19:57:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-11-19 15:47:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-11-19 17:33:12.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"043cebfc369d3ef94e107f81b16a8441fd18117b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-27 14:15:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cadb2646213f5d7c22560d14464dd123539d3dba","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-27 14:40:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6f31dffd83f99978f2bce358e4402161879b3c0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-27 14:42:09.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6957046f7b434b598110e7dc8dae6d83\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f59ee45ae0840fa83604125a8d57ed0 : FAILURE in 26m 27s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b891e454f6b442ba1242f9e8f1a2319 : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f2631fc946436d31aa0bd09b9100ac37ba0cc07","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-27 14:45:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be4432ca647d9a0d073c41a67b77da6fabcfa548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-27 14:45:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"45e11535c3a0fe9a6953eb43198f5a77304140f6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-27 14:50:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc170b0c5b1a1c663ab5619fcaf0e754e80f6a14","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-27 15:14:11.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9682d43d67e242338eb4ae20664c7ae2\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec181b58438144a8935fa2c8b36e5b28 : FAILURE in 27m 54s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef5071146b41401a8ddb43b8385522f4 : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a84cce9a207c831133b06794b513f36b40bd672e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 15:21:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/2/check/DellEMC_PowerFlex_Manila_dhss_false/dbc74fa/DellEMC_PowerFlex_Manila_dhss_false/1762 : SUCCESS in 35m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"b6e58d06b9e1d7033fa601b8b531b41bea95f6af","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-27 16:23:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/06/930706/2/check/DellEMC_Unity_Manila_dhss_true/1eabbf3/DellEMC_Unity_Manila_dhss_true/1749 : FAILURE in 13m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/2/check/DellEMC_Unity_Manila_dhss_false/fd1911e/DellEMC_Unity_Manila_dhss_false/1852 : SUCCESS in 1h 37m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"07afa2ef34806ba9b450e30661f0a0f19b86a474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 17:56:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c978f468a1db434899b3967db086fad8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc22b99cd51463ca98b900366db3e56 : SUCCESS in 12m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/803c276eadfc434ea3bd17447eccade3 : SUCCESS in 28m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b58e3cf937e94c9c9ec8717d0cdafe02 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53d52b698f464173a967b3fd5b523543 : SUCCESS in 10m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e189b2022dd44d79a029460c23fc8d0 : SUCCESS in 8m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/671cd44bc86c4a55857407fb65174d74 : SUCCESS in 10m 22s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/971c1c654d42430aa4711860933c056e : SUCCESS in 3m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/57ecd8329ed04853a5c0db4ec281f0d9 : SUCCESS in 1h 19m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d8d2715cb89b4bd18e280eac4608fcee : TIMED_OUT in 3h 02m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e88b73c921794c5e93882473173c9949 : FAILURE in 49m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3f9ca5f6c44243a394dec915a1c22b28 : SUCCESS in 1h 03m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/22f80103077c4447bfbf986cda4f3387 : SUCCESS in 58m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/08ab552ea05641af83a760e91874c386 : FAILURE in 16m 49s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ab7c3e0cdbc41329a309a528c541280 : SUCCESS in 4m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aebb6a19019b46098d2b9dd614304545 : SUCCESS in 5m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/75c894f43ee04e6bbdcd163a70ed47cf : SUCCESS in 33m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/81e5125437f54cdfbe2d1f5a8392ded1 : SUCCESS in 44m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/32cae98299584a83a2f46ac72f755768 : FAILURE in 16m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5ab801c1f7a94eb5bb3acc2b0e722dee : SUCCESS in 1h 20m 54s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c64fa3bc38c7477f91f2f1eb9993598f : SUCCESS in 1h 02m 52s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d9f5619112604e53987553fb215a289a : SUCCESS in 29m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/07c81a8a8a8e4de3afd463e33baf36a7 : SUCCESS in 33m 57s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/47af061667ca42f4afb897f6c3054647 : FAILURE in 4m 39s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d5d2c491bc3d4b22a36537c7f8179ce6 : SUCCESS in 7m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4886c4062daf3933bb54d52158b60a199f59ee8b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-27 18:11:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/930706/2/upstream-check/manila-tempest-plugin-ontap-dhss/645c425/ : RETRY_LIMIT in 14m 12s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/930706/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/cf5225d/ : RETRY_LIMIT in 9m 07s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"fcd108982cd12494cea043e41e125e1164151aa7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-27 18:32:51.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/2/check/DellEMC_PowerStore_Manila_dhss_false/515cd56/DellEMC_PowerStore_Manila_dhss_false/518 : SUCCESS in 36m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"ffbef4654fba29e4388b12a3a0dda69164a72e76","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-27 18:57:07.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/2/check/DellEMC_PowerFlex_Manila_dhss_false/9571575/DellEMC_PowerFlex_Manila_dhss_false/1765 : SUCCESS in 34m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"af00499a307af77716a7ea7d82eed96c2bfe8d50","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-27 21:20:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/06/930706/2/check/DellEMC_Unity_Manila_dhss_true/36dc497/DellEMC_Unity_Manila_dhss_true/1752 : FAILURE in 13m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/2/check/DellEMC_Unity_Manila_dhss_false/9f6efd3/DellEMC_Unity_Manila_dhss_false/1855 : SUCCESS in 1h 12m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8aec3e260e8e411dce6ced19f302858ddd8b7bf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 05:05:27.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"12a9f2b8ed75b216652eed9b0392a1359aab5f47","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 05:06:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6cbbb40c738a997fc1b1c132f21d90ffef40f0e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-28 05:35:42.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2ad5a9bf253e42f29331ebd99ba915cf\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d9b76daead4b40b48edf4c0ec49c3f69 : FAILURE in 29m 43s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6ddf0dfe05f6405b8619a427af64d7fa : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"96978c9a00382f5d2bba8a02686b373b68ddbd21","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-28 05:39:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/3/check/DellEMC_PowerFlex_Manila_dhss_false/0e895e7/DellEMC_PowerFlex_Manila_dhss_false/1770 : SUCCESS in 33m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"a7b48c496f6ac9ff52f116f424e078ebea08c3e5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 05:54:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c1a4983da56c376c0fb9f7a6700d97cf9fc2b288","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-28 06:20:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/06/930706/3/check/DellEMC_Unity_Manila_dhss_true/2b24be6/DellEMC_Unity_Manila_dhss_true/1754 : FAILURE in 13m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/3/check/DellEMC_Unity_Manila_dhss_false/0fb6b18/DellEMC_Unity_Manila_dhss_false/1857 : SUCCESS in 1h 14m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"213b005b02b83befb85c7c5417fa558a1b13e229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 06:33:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be2b9ffd41b4295857699dfa1e34273\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e686b8e53924aa78a4ea37268d631b9 : SUCCESS in 10m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81759cc2af0b4c1eb8c2ad80877c844e : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba74309f755408abd11c8a0b2d71321 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec7015c65d44452ab65a1daa30c959b8 : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19df7227344c464fb21ed9cb1b6d8155 : SUCCESS in 10m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd25a300a49e4abdbcbd6e481e70e6d3 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9ec58d7d2f814dfa8c95c6573becf123 : SUCCESS in 50m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c38bba13699c48689ab7c7984dbaf94a : SUCCESS in 1h 26m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/36fbc0c3d320454f8d16d6b3e3265cdb : FAILURE in 49m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0979370ffc854628a7b45b1f3b3f27e6 : SUCCESS in 46m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5d945484c3ff49e2b5c481e41bb0e7eb : SUCCESS in 58m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/44abf845b6fe49cb8748e604cdca8ff3 : FAILURE in 13m 20s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3f96c8ca84a442669225fbdabf2e7eed : SUCCESS in 4m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b41a2fed1e70413dac91115a830b0735 : SUCCESS in 5m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fdcdc0046b9842c9b0b5614146edfa3e : SUCCESS in 28m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b292630feca64277a340488dae1e267e : SUCCESS in 43m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d734b1ef34a4407b3d024cfa59c6361 : FAILURE in 20m 14s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1a1b363e6b994e81ad5e66be937104c1 : SUCCESS in 49m 46s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53a0bca6685546d38b75733da72310d8 : SUCCESS in 52m 29s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/39e0baa8d75b42da9bdc7ddc4d28cedd : SUCCESS in 20m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e1eeb70fcc1a459db9772fa390b54e6a : SUCCESS in 38m 36s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/34e3ae46c217491a8aea01124a0dee1d : FAILURE in 4m 31s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1808213104594e3d89bbeb0fc59867ea : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b08fa6acbc03a2bd8716223f64031458a6bdbf56","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-28 06:48:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/930706/3/upstream-check/manila-tempest-plugin-ontap-dhss/59ca2da/ : RETRY_LIMIT in 8m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/930706/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/4e53955/ : RETRY_LIMIT in 9m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"dfc29633524c083ef9c2cbaefceee015d3eb39b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-28 07:04:40.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/3/check/DellEMC_PowerFlex_Manila_dhss_false/5771b23/DellEMC_PowerFlex_Manila_dhss_false/1771 : SUCCESS in 31m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"d0f618375060c63e3324339205b059d3107bb665","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-28 07:08:59.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/3/check/DellEMC_PowerStore_Manila_dhss_false/3a7f516/DellEMC_PowerStore_Manila_dhss_false/519 : SUCCESS in 35m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"6d1b33787fd61f13b5cee9a24b8549e515e6d351","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-28 08:18:16.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/06/930706/3/check/DellEMC_Unity_Manila_dhss_true/12e3d8b/DellEMC_Unity_Manila_dhss_true/1755 : FAILURE in 13m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/930706/3/check/DellEMC_Unity_Manila_dhss_false/8ed0258/DellEMC_Unity_Manila_dhss_false/1858 : SUCCESS in 1h 44m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"f6c7758018bb0b8a949592ad297487ddf509c7b7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-10-02 19:53:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0af9959f77a26b32ffe7afb62b06c58dded98d28","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-11-19 17:32:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2dae38999c6931acdf95000df8401d17ed6183f","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2024-11-19 17:33:12.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"34fed4c48643e83f6da49fc396391f122d1bfa5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:33:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f9e506d1e5b28d125e0fce456fc025af6770bf9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 19:05:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dfa0cc00db746b0a501a1561fd31d2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded60f6589a54db1b5994c1c7546f903 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15821e6d294e4f1f88732e620c36ede2 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0990367e95184edca5832ce8d4004e0a : SUCCESS in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/901c12ce681a432caa12320accabb8eb : SUCCESS in 9m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99283b5f8dd647d7ae223888a9329f8d : SUCCESS in 2m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a0595b7e56254c3587f48d9af138be78 : SUCCESS in 1h 14m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/066b28b1edb3499187f0d4d5f07a64c9 : SUCCESS in 36m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d6d700e4773f4fb4b1e6a81450e74549 : SUCCESS in 45m 07s","accounts_in_message":[],"_revision_number":3},{"id":"37215a0f11a7b48b2449ec6e243bb7ccd06e1dce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 19:05:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"aacacfb0860682512f656402bf1b3a0fdb46c972","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 19:06:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d33beea91dd4a48a1f8b36a1846f0f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49918fa8c0de4a8ab577dee73b7a9fbc : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3},{"id":"c12a6c660eeea3f9fd10f09a1b09f429a6950a46","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2024-11-20 00:38:39.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/e7cd595306b44017bef8dfbab53ac1b3\n\n- manila-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/bb54ab81c66d45cb899c957405457269 : FAILURE in 7h 05m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"353feba0b0842e99145b35bde3787ee29516c727","revisions":{"bae201c562e3f9ba20cc77e57b880d67361b184a":{"kind":"REWORK","_number":1,"created":"2024-09-27 14:15:11.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/06/930706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/930706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/930706/1"}}},"commit":{"parents":[{"commit":"fd87d87f609e62075eb1a6b7d4c303d75eb86bc8","subject":"Merge \"bug fix: error message fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fd87d87f609e62075eb1a6b7d4c303d75eb86bc8"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-27 14:11:38.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-27 14:11:38.000000000","tz":120},"subject":"Remove workaround for eventlet bug #592","message":"Remove workaround for eventlet bug #592\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nChange-Id: Ica7ac704ae358d0271562d3c02b4b56dfb7ded7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bae201c562e3f9ba20cc77e57b880d67361b184a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bae201c562e3f9ba20cc77e57b880d67361b184a"}]},"branch":"refs/heads/master"},"dc8fdc52ed8527339a906b89cb7840df5e4750de":{"kind":"REWORK","_number":2,"created":"2024-09-27 14:45:46.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/06/930706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/930706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/930706/2"}}},"commit":{"parents":[{"commit":"fd87d87f609e62075eb1a6b7d4c303d75eb86bc8","subject":"Merge \"bug fix: error message fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fd87d87f609e62075eb1a6b7d4c303d75eb86bc8"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-27 14:11:38.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-27 14:44:52.000000000","tz":120},"subject":"Remove workaround for eventlet bug #592","message":"Remove workaround for eventlet bug #592\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nChange-Id: Ica7ac704ae358d0271562d3c02b4b56dfb7ded7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dc8fdc52ed8527339a906b89cb7840df5e4750de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dc8fdc52ed8527339a906b89cb7840df5e4750de"}]},"branch":"refs/heads/master"},"353feba0b0842e99145b35bde3787ee29516c727":{"kind":"REWORK","_number":3,"created":"2024-09-28 05:05:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/930706/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/930706/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/930706/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/930706/3"}}},"commit":{"parents":[{"commit":"fd87d87f609e62075eb1a6b7d4c303d75eb86bc8","subject":"Merge \"bug fix: error message fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fd87d87f609e62075eb1a6b7d4c303d75eb86bc8"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-27 14:11:38.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 05:05:26.000000000","tz":0},"subject":"Remove workaround for eventlet bug #592","message":"Remove workaround for eventlet bug #592\n\nThis code worked around a bug in eventlet[1] that has been fixed in\n115103d5608cbe8f15df10e27eba1644f5364e95. The fix has been available in\nevery eventlet release since v0.27.0.\n\n[1] https://github.com/eventlet/eventlet/issues/592\n\nChange-Id: Ica7ac704ae358d0271562d3c02b4b56dfb7ded7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/353feba0b0842e99145b35bde3787ee29516c727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/353feba0b0842e99145b35bde3787ee29516c727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
