)]}'
{"id":"openstack%2Fmanila~935849","triplet_id":"openstack%2Fmanila~stable%2F2024.2~Iac0bee438365ab0902859df74647f1502c619270","project":"openstack/manila","branch":"stable/2024.2","topic":"bug/2084529","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-11-21 23:29:12.000000000","reason":"Change was submitted"},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2024-11-21 23:29:12.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-11-21 22:09:07.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-11-21 23:29:12.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-11-21 08:27:30.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}}},"hashtags":[],"change_id":"Iac0bee438365ab0902859df74647f1502c619270","subject":"decorate all missing *_update methods with db_retry on deadlock","status":"MERGED","created":"2024-11-21 07:02:17.000000000","updated":"2024-11-23 10:04:54.000000000","submitted":"2024-11-21 23:29:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935849-bug/2084529","meta_rev_id":"366f00692ca772ec73426de6e83c9d55be93a862","_number":935849,"virtual_id_number":935849,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-21 23:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-11-21 22:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-11-21 08:27:30.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2024-11-21 22:09:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-21 07:03:36.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2024-11-21 07:03:36.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-11-21 07:03:36.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-11-21 07:03:36.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-11-21 08:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-21 20:18:40.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-11-21 22:09:27.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"c8513b60a87fc92dd6d5e46baab04a160324c7ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-11-21 07:02:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a15963b5a789e9f1a8c2f89212fcc6f89695037","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-11-21 08:27:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de1902f8a6b45008a51560b499f485bf1aaaa2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 08:47:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79c0bb88a91147d4ae232b41f2857950\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2cf619754274bfeb2db7a07984ba9d1 : SUCCESS in 10m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ca0f6ae8f7489095c0cd4c0ac09f85 : SUCCESS in 28m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37660378447b46378ef9ee3ff020c002 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08822cd4e0df4798a2ec7802e0b6bf3e : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c556d4bea77e4c13813077000804b66f : SUCCESS in 8m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/520fa1a9ddfa4b399c5574a31eaedb29 : SUCCESS in 7m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6a5e95d67864b04847b552d01b8a10a : SUCCESS in 7m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a32e558773d8447193038bbed7972237 : SUCCESS in 50m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c1a72de5d5564d5ea043f88937b09888 : SUCCESS in 1h 43m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3c9b4ae9d9584bee870f06655ae5aa26 : FAILURE in 1h 15m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ee24914ef1e441cba2b401586c2dad85 : SUCCESS in 1h 08m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/598f6d450b344385a50429d74aedb79f : SUCCESS in 1h 10m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e3f490c6782c4a78b0598e500e9225b2 : SUCCESS in 1h 18m 52s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f777fb477ae84b7d9848cf293c8cc8bb : SUCCESS in 4m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/992ef688484540fd95703f4972aad2ae : SUCCESS in 3m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/714caeec5d634c598531e7af205ddb27 : SUCCESS in 33m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/14b5c2291ade4096b06bcb870e333448 : FAILURE in 14m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bc140cccb3bb4df0b8829ac973cefa95 : FAILURE in 21m 50s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1a667ee20e6942289d83882825dedfe5 : SUCCESS in 1h 24m 30s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f86b99dce1f48898103f0fac5cfa2cd : SUCCESS in 1h 06m 01s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b3fe640e36d84e0eb0ba236143b699f0 : SUCCESS in 26m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/aec52643ed4f4dbe88b34e5c71674405 : SUCCESS in 22m 33s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/8ba8d682962f40bc9a24b0cbfbdd9ee5 : FAILURE in 5m 06s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3e5b0c0461d54a01b4225171a8efc287 : SUCCESS in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c75071972bbbc8637cdeb9c5e6b7fbb61fcecef4","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-11-21 09:38:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"194c1001946e48039c59395fc58c9ed58cd81011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 13:26:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7301442912fd4e479f9c0da06ac31b8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/855b81028ac84ae0880991251438d373 : SUCCESS in 7m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c27d41e43de6429aa9515e7487356898 : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192eb556a8c84c41b4851113ca1bf244 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/016e2197b03541a69be293b330dd8730 : SUCCESS in 10m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b42e04f82d94e20a4cdc5fb5d82f2b7 : SUCCESS in 8m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e790c38aefd343e79102f744a58a5f0d : SUCCESS in 11m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae462bb5dce46f1bc62d5a6f811c85a : SUCCESS in 8m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/496c7a252b7f4513a6e15ee73c1866ad : SUCCESS in 1h 13m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cf216de6573142b99a06c5e72f07c457 : FAILURE in 2h 55m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0cca14a0b92742ed9b00f9003000b6fc : FAILURE in 1h 31m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ce1a855118764fc99bb97eb6b5f7a7ad : SUCCESS in 1h 02m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0eaf7b204d67431485e0003cf1a303d3 : SUCCESS in 1h 04m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c0a4acbe74f04befa081c0466d2a6d4e : SUCCESS in 1h 00m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c9651f4a17654b46b7b276e9aab5b3ca : SUCCESS in 3m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/030be3d5f42f496ca300c15f8acfb673 : SUCCESS in 4m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0cf39771757048eb8fcbef85dc2bd4c9 : SUCCESS in 26m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/73882b9f33524a9984c7c30784f3db61 : SUCCESS in 38m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e0ad136bbf634e34af9364c56c1df2ed : FAILURE in 49m 16s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4f26153c5dca47dc89a06f7c2dbb776e : SUCCESS in 1h 16m 50s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21ecfbe331d34168baa45e1f4351f191 : SUCCESS in 1h 12m 35s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0cd05009e0074545a63ca32dd2a8c8b8 : SUCCESS in 41m 40s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0304e66dba6541808efb96900aaf7b66 : SUCCESS in 36m 59s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/bece6690b7914df9b088d065204f76ad : FAILURE in 4m 13s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4ce137fc570c46768594d61f79db5ef8 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7446202f7ca9498ce22f3a3f7f2b031d4cdf5027","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-21 20:18:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/52f512099ffb49f58c5cc2ff39c20c74 : RETRY_LIMIT in 11m 46s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f639c4e29f8b7347b1f90983cf2b710f1951c7d3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-11-21 22:09:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0007f332d7c2c1276dc9a97d652f9bb7cf710a20","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-11-21 22:09:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9bf4c45cab07d2d710f4a1964b5415283f1873c6","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2024-11-21 22:09:27.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d9bde19260bc677876d8a4952250716df04c59e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:09:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f09a66ba1dcb446940863b8d1ea80dbe8d6a99b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:29:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970162d029014c47a6125793fa461943\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d716c91165414081cbad4190df4df9 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41fa8e206fb843eba5607c2225584bb0 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a64e599af9b9493f9d953154cffd1abd : SUCCESS in 11m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7771ed10a22846a6b0c0baebc17ea2d6 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/495b22bad7d44666bab627c0e7abb2ec : SUCCESS in 8m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c378ffd7180844ac8cc352d51da3af9d : SUCCESS in 1h 14m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2fce53988cdf4a949f8a1e26fd92412e : SUCCESS in 25m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/949b37d2cb34471eb0f0767b730703a0 : SUCCESS in 25m 13s","accounts_in_message":[],"_revision_number":1},{"id":"31a073d87550d44bafbd4c07da4415716391c3bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:29:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"82751e5054e42512b14a73ec1fb92e9f8567f1e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:30:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc7804b05a0416e835eff58c155d952\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8bc53f589ba4596b8c88391bbd78c02 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68fecddaa26c4bdf87fb031e65933ea2 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"366f00692ca772ec73426de6e83c9d55be93a862","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2024-11-23 10:04:54.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/eec15160943a48cda0af7e419fc55d38\n\n- manila-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/16a07bbb3ed248f5ac924b6233f32da2 : FAILURE in 15h 20m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"052e324e26b6ebdc4b8e011df9f41b43f1cea09b","revisions":{"052e324e26b6ebdc4b8e011df9f41b43f1cea09b":{"kind":"REWORK","_number":1,"created":"2024-11-21 07:02:17.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/49/935849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/935849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/935849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/935849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/935849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/935849/1"}}},"commit":{"parents":[{"commit":"a1706e52a1faa1e1ed367fc996ec17117a247b7a","subject":"devstack: Fix missing manila command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a1706e52a1faa1e1ed367fc996ec17117a247b7a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-10-15 10:01:21.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-11-21 06:35:42.000000000","tz":330},"subject":"decorate all missing *_update methods with db_retry on deadlock","message":"decorate all missing *_update methods with db_retry on deadlock\n\nShare instance records and their relations were missing db_retry and\nso added the db_retry decorator.\n\nCo-Authored-By: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nCloses-bug: #2084529\nChange-Id: Iac0bee438365ab0902859df74647f1502c619270\n(cherry picked from commit b64e5a35e1067976de3c92372476582b4d99a3ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/052e324e26b6ebdc4b8e011df9f41b43f1cea09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/052e324e26b6ebdc4b8e011df9f41b43f1cea09b"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
