)]}'
{"id":"openstack%2Fmanila~944849","triplet_id":"openstack%2Fmanila~stable%2F2024.1~I3f9d60509ab01bcc4c808341dda8ea6c46f04738","project":"openstack/manila","branch":"stable/2024.1","topic":"bug/2101015","attention_set":{},"removed_from_attention_set":{"1313":{"account":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"last_update":"2025-03-20 04:12:07.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-03-20 01:31:03.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2025-03-20 04:12:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f9d60509ab01bcc4c808341dda8ea6c46f04738","subject":"fix volume resize issue when host not recognized","status":"MERGED","created":"2025-03-18 09:51:40.000000000","updated":"2025-03-21 02:05:07.000000000","submitted":"2025-03-20 04:12:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944849-bug/2101015","cherry_pick_of_change":944820,"cherry_pick_of_patch_set":1,"meta_rev_id":"454e99fb2590f5e2116c0e26a33c250977737158","_number":944849,"virtual_id_number":944849,"owner":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-20 04:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-03-20 01:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-03-20 01:31:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 00:33:48.000000000","updated_by":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2025-03-20 00:33:48.000000000","updated_by":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-03-20 00:44:10.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-03-20 01:31:26.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-03-20 02:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-20 05:32:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"1d7854285d177b5c4f8ef59597936ff3d8210625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-03-18 09:51:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"c5aa21064c109a7377486e890fbe958a0a36a004","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-03-18 09:54:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"2b5ff891d91d66136a5d5f1b16485bef573a49f9","tag":"autogenerated:gerrit:restore","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-03-20 00:30:42.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"e0bd6768782d3e44a67aa7bdd47c7c39d9ac2d15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-03-20 00:31:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"11b2f5f1f94a9c20823aee7ca82c79f8aa42ba3e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-03-20 00:32:52.000000000","message":"Topic fix-exportfs-resize-stable/2024.2-stable/2024.1 removed","accounts_in_message":[],"_revision_number":2},{"id":"b1a89b5f3771b09a3c2019a3822d92ecca5754a2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2025-03-20 00:32:59.000000000","message":"Topic set to bug/2101015","accounts_in_message":[],"_revision_number":2},{"id":"260f0da2be7501a648cc8bd824f7aac866a6f643","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-20 00:44:10.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0bea34d0ab7749d597ecee1bb6b6c91b\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71f2983af7ad40e4b3f952669cc5953c : FAILURE in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"694391de17a336fc1f0539c3ac9841d1d05712c8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-20 01:31:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ae2efde4527cf554f33d5f16f3e4f79a98b0105","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-20 01:31:26.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28bbbfad24d0b3b9e88c3af9bfc5bcd6f1e277d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 02:31:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c6d91bca03429e815b036afe05cfe1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb622b26af3e4b5fa6677d9efdb19f10 : SUCCESS in 7m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7092416f262c4dd38a74d3dad115794a : SUCCESS in 16m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07a2b133acc4c3693e1ddb33473b3a2 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc383100c88f46098c4e15726992be5f : SUCCESS in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bca3f618c089498bac0247d938f6739a : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d079193656e4217a00ff0454be852e5 : SUCCESS in 6m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c4d9bc08a5f1456da407c31f614478db : SUCCESS in 1h 02m 46s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/c33678c0f0af45b5b23545e36c1ce3a6 : SUCCESS in 1h 10m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e92d73b03b764799a898b827963ebf19 : SUCCESS in 1h 58m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ad4e30174dd8409795d334828332fefa : SUCCESS in 1h 03m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d932aa1664be4d74833becfc20bb2669 : SUCCESS in 53m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cdce58f74d2e41489d5c2441beb49783 : SUCCESS in 33m 42s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ed51195ca3b54d718b5d6ff967b969a1 : SUCCESS in 2m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/74951114930243499e5acbb013b9a2f8 : SUCCESS in 2m 14s (non-voting)\n- manila-tox-py311-with-sqlalchemy-20 https://zuul.opendev.org/t/openstack/build/a9b0c31289d040bdac76087d71e2784c : SUCCESS in 5m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bd2120f0d9524532a74537dbdcf34bc8 : SUCCESS in 30m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a4cfdd7447e94ba789c0c9f27f8bad78 : SUCCESS in 24m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e2e52586d64e4403a5522b33212b6ccb : FAILURE in 32m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ca678c9560124cde8234e93cde53d660 : SUCCESS in 31m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8d839e2f38b7413b982d02ba897297d1 : SUCCESS in 51m 57s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7551b7a3a734ae2981f9d84a6cfdfe9 : SUCCESS in 48m 03s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/92f29d4d89f545bcb2829b9a7c081093 : SUCCESS in 22m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1283c27a1fe042998648686fbe63e1e0 : SUCCESS in 35m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb988ee6100dd85bf4ed722cebe13550e86b1e39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 02:32:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"777c00b43c3d91a79b5cea796083b6a6439696de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 04:12:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa185edcc47642c695c9989513d19b0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6b40be41934f3aa88b30d06c078837 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6563c4124be4471589abd65d533f9cb8 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c61548129f3f4ea39413ca42d451a2ee : SUCCESS in 9m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0632ead4b99b4c1eb77090be77bd3edd : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2080211c4f9b47cc9b61229f34b2eaf1 : SUCCESS in 13m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/72a9e311fd644d0a9a49eb7f496f7cf6 : SUCCESS in 1h 35m 00s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/fb949db5a4b246ccbb16b8299ef441fb : SUCCESS in 1h 15m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/73feef559a324ff398ac2e6916140e44 : SUCCESS in 35m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d213640342624f178baf2f175b4a4771 : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":2},{"id":"65a85710581ae5185c43337c705e56e5269d048d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 04:12:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6d7d42bb677153ce5ec1079c90838239ecefd727","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 04:13:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db816bab6cb439990d04b932071b011\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6586e8871e4c2ba7527ca927795b3f : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64f3e7229e6d4cc7882cd7d5b5d8ce1b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2},{"id":"63923395d278da25fa06411e2dfdefb4608c7262","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-20 05:32:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"454e99fb2590f5e2116c0e26a33c250977737158","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-21 02:05:07.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/4378e6227df34f7ead1a3b73a5a2c4aa\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/626cb770f30b4f659f4b2941bf638491 : FAILURE in 1h 29m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82428a03db414cddd9b7c89433f73e10034359de","revisions":{"06834c1d8e1fd3fba0ea5eb9822b65adc7d9e06a":{"kind":"REWORK","_number":1,"created":"2025-03-18 09:51:40.000000000","uploader":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"ref":"refs/changes/49/944849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/944849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/944849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/944849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/944849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/944849/1"}}},"commit":{"parents":[{"commit":"221dae96df038d572dbe99f71e75e7bc7d2cf55e","subject":"fix manila to use device uuid instead of name for mounting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/221dae96df038d572dbe99f71e75e7bc7d2cf55e"}]}],"author":{"name":"Yaguang Tang","email":"heut2008@gmail.com","date":"2025-03-12 06:43:23.000000000","tz":480},"committer":{"name":"Yaguang Tang","email":"heut2008@gmail.com","date":"2025-03-18 09:51:40.000000000","tz":0},"subject":"fix volume resize issue when host not recognized","message":"fix volume resize issue when host not recognized\n\nCloses-Bug: #2101015\nChange-Id: I3f9d60509ab01bcc4c808341dda8ea6c46f04738\n(cherry picked from commit 0ac0e43e0e7afbd1d9be95e92c06a34d26892029)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06834c1d8e1fd3fba0ea5eb9822b65adc7d9e06a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/06834c1d8e1fd3fba0ea5eb9822b65adc7d9e06a"}]},"branch":"refs/heads/stable/2024.1"},"82428a03db414cddd9b7c89433f73e10034359de":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-20 00:31:00.000000000","uploader":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"ref":"refs/changes/49/944849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/944849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/944849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/944849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/944849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/944849/2"}}},"commit":{"parents":[{"commit":"221dae96df038d572dbe99f71e75e7bc7d2cf55e","subject":"fix manila to use device uuid instead of name for mounting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/221dae96df038d572dbe99f71e75e7bc7d2cf55e"}]}],"author":{"name":"Yaguang Tang","email":"heut2008@gmail.com","date":"2025-03-12 06:43:23.000000000","tz":480},"committer":{"name":"Yaguang Tang","email":"heut2008@gmail.com","date":"2025-03-20 00:31:00.000000000","tz":0},"subject":"fix volume resize issue when host not recognized","message":"fix volume resize issue when host not recognized\n\nCloses-Bug: #2101015\n\nChange-Id: I3f9d60509ab01bcc4c808341dda8ea6c46f04738\n(cherry picked from commit 0ac0e43e0e7afbd1d9be95e92c06a34d26892029)\n(cherry picked from commit 502f1c8d5b9e8dee1bae53158c98ccb3b0c4a7ae)\n(cherry picked from commit ce9e7e4c898a5654d8c6fe5523e74c842f2828a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82428a03db414cddd9b7c89433f73e10034359de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82428a03db414cddd9b7c89433f73e10034359de"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
