)]}'
{"id":"openstack%2Fmanila~945552","triplet_id":"openstack%2Fmanila~master~I6b809e093413a9a5d574023db73bbb01023f0e9f","project":"openstack/manila","branch":"master","topic":"bug/2104153","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2025-05-22 19:21:21.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-05-22 15:37:17.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2025-05-22 19:21:21.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-05-22 19:21:21.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-05-22 18:45:27.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-05-21 08:35:39.000000000","reason":"\u003cGERRIT_ACCOUNT_36180\u003e replied on the change","reason_account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-05-22 19:21:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b809e093413a9a5d574023db73bbb01023f0e9f","subject":"NetApp: Fix export location during replica promote","status":"MERGED","created":"2025-03-26 10:32:37.000000000","updated":"2025-05-22 23:27:49.000000000","submitted":"2025-05-22 19:21:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945552-bug/2104153","meta_rev_id":"7d92a4ab3a8ab4b0d45abd23972672e4331186ac","_number":945552,"virtual_id_number":945552,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 19:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-05-22 15:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2025-05-22 18:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-05-21 08:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2025-05-22 18:45:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 10:34:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2025-03-26 10:34:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-03-26 10:34:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2025-03-26 10:34:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-03-26 10:34:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-03-26 11:03:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-03-26 13:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 16:23:58.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-28 16:56:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"CC"},{"updated":"2025-04-11 11:51:36.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2025-05-22 18:45:44.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"69b33cde0dbac706c65412ab0204cff5ac55bfff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-03-26 10:32:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c771dabca7c0e8cddb177f2437898aa213a31af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-26 11:03:46.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b7b9f6696b7a4034996f9c207aad8fa3\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c11b97b3d9c947eabfe198c6b1234a28 : FAILURE in 30m 38s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec29d8d7a6bd4ad9a2f68d07e98e19c0 : RETRY_LIMIT in 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c93e4ccc56972f57b269e641b3db7844e81d2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 13:22:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de93d150374348f9a3e99d02e19c5739\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b819fbb124e0456aa4333fda84a64eb6 : SUCCESS in 15m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbf4f4c5dc6543e8a8e04626b80ff863 : SUCCESS in 31m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ecc6e46b5b4d50b0191db74cfa9e63 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5231d47a1f2546db930050d190172d56 : SUCCESS in 12m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9445374a92374166b6070bee4e55aee1 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee71cf915a414d25bf46e6ca55292762 : SUCCESS in 7m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fa229aa1573e4a0b8427b81710f11566 : SUCCESS in 1h 17m 37s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/3d3df1266dfd475e9bacc733a6ab653f : SUCCESS in 1h 05m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bc063418046841d0bfd35f9f2bc2011e : FAILURE in 2h 41m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f2d5e8d63a8343b69f572104214a6fea : FAILURE in 1h 10m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/396e9f85cc944f79b102ec285d4f9dbb : SUCCESS in 1h 04m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2cae189ba95e48a6bf334f850a47627c : FAILURE in 1h 36m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e4d68578a9394bf89a93687cac20b8c9 : SUCCESS in 34m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6612fea7827f4e2b9f484659f46d9260 : SUCCESS in 2m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b543de36032b41eb90b739b38bfee591 : SUCCESS in 4m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3cec8e1d133c45e7b996d6286c9b5454 : SUCCESS in 32m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/08c2727e446f432fae14e71b2c658da4 : SUCCESS in 34m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/450c38751fdb4c0394ddf955a53754c8 : FAILURE in 42m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8ba9d4729f6e4f9cb9ccc4c77f10badf : SUCCESS in 1h 00m 28s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/63ad5c5095a4451d8c4ffe3f2f6a46a2 : SUCCESS in 43m 43s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d4c14624b57409cb373b24d47de79b9 : SUCCESS in 46m 22s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4021f8acc4ba42e0a5a1c437136c09ce : FAILURE in 5m 25s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/29e5ca863f99485db2e227003c0b03e4 : FAILURE in 4m 51s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/10ce75301849405084abebc1b4ce334f : FAILURE in 5m 08s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/05728c184bf644b2aa74d13dcc8d520a : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f35fa610e5397c0b4d9a0fe25fe0c313be502233","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-26 16:23:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest manila-tempest-plugin-ontap-dhss-rest : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9efaba240545ccde2eeae537e0f674fc5cc434b","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2025-03-28 16:12:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"622fd21f2bd6e54edc10ee22fdc1e54f134fb678","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-28 16:56:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f616dc34026e7d4fc0f7a5e8d056582ff3e933e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-03-31 05:42:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"93b36f1ef8fdb8261d07d66d0f4c1e65990bcb70","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-03-31 05:47:18.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"af638b60bf85577ad17e5a1ff645fb8eb5a8ff1a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-31 06:18:05.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/38c0cc4465284f6694e4b612286973b8\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/edba6249664e44f0a10370c8d81f01e1 : FAILURE in 34m 50s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cae16b2916ad4dbfa6fd667302700601 : RETRY_LIMIT in 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8b7a5ab521106172a7dc07724f16b2f7c165cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 08:18:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a401489e0634bc9abe62e5afafd7db0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330ffced5b454bcb8d13afa38b1b42fc : SUCCESS in 11m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60e2661607b142978cee7f18d40ddbf8 : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818aec80489f44d7abea334761ab9e16 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff24e4d5622d4a078971620721e99734 : SUCCESS in 9m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9de45a1d99d241198b6074a1aa1240c1 : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ab7eb9c6d2b4f5eac4bd8e2d199df68 : SUCCESS in 7m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cead1bbc29a24b2abca3b50c442e5567 : SUCCESS in 1h 11m 21s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/46a342ea76c741d39211ea195b4ed6e7 : SUCCESS in 1h 04m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6e83a0f2ed744cf596e8269b293a0fe2 : SUCCESS in 2h 31m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f1c0892353b1404091aeee965261aaa9 : FAILURE in 57m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/50fa29f4b5ff4df0b05792d77e1b7ab4 : SUCCESS in 1h 06m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/16eab77ed49344d684e7acb560858138 : SUCCESS in 1h 29m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/99156b3e2558441a91251cfeac567197 : SUCCESS in 47m 54s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8bde7a1175ec4f8bbccb1028c2215fa9 : SUCCESS in 4m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/31c6cf4f4e8b4ab1a0b201aa66f83af8 : SUCCESS in 2m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/653bca8de3ef471989b092dffdbfd04b : SUCCESS in 38m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/82cddad82f5a46eda865fce89bd436e8 : SUCCESS in 25m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1cd6b217c23a485b905e8173b7d5a0bf : FAILURE in 45m 10s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5447ab4d28da4c479794484f07da0fed : SUCCESS in 58m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6dc66c863c2f47c08e0e526ac7bdf787 : SUCCESS in 54m 29s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb693f26d7054de5840e0605c2f321d6 : SUCCESS in 44m 20s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c859608da9be429d8c32f3ddc0df5173 : FAILURE in 9m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6737b32fd62c4410bd97688f2d13196e : FAILURE in 5m 19s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/7c8bb17e94df46e2bd9f63b3e3f4c181 : FAILURE in 4m 06s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c5fc10ce6325414c8aadbe41a7a67d3c : SUCCESS in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34c2854a520e977ae45fb960c41518880298e1a9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-31 11:19:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest manila-tempest-plugin-ontap-dhss-rest : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"13675477fa3e2a28569624819aabe868b2916466","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2025-04-10 09:10:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3072e796b14064f42c9859ce7eda11047b46e0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-04-10 11:05:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"01591ff9d2eb1aa32f7ddcfcf49f073a7f6af35c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-04-10 11:05:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0258f83f6ad2b3ea40c062488e8cba9ea746561","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-10 11:41:51.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d8bbb34ecc6a47c699f96062b68d11f5\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dbf7796f366340848facbe21ddf8db86 : FAILURE in 35m 03s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46630f2d42dc47adbb6f11e9cfc05287 : RETRY_LIMIT in 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aac232d3cc65478754ff318556f7eaaa9766fc7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 13:25:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f0e0e02c06486abccd4267bd7fce72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c3fbb3e8f34954abf23e0cfb5e7f58 : SUCCESS in 9m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e38a596e04df47ca894662a1f43b2637 : FAILURE in 19m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053d59f0d7614453a417739489ef98b6 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1978e33eca7d46e1854191ad9278d725 : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3211b99e02b5426e9b22153877ead006 : SUCCESS in 10m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff92aa619b934d57b916021a0f7a27f4 : SUCCESS in 12m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96d0244680b74d15a82c74d97ba83712 : SUCCESS in 5m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/58ec18acc6af446cb80778276610f40e : SUCCESS in 31m 35s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/1f580ff3cc254373bccd3561cb18530a : SUCCESS in 1h 09m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d7b054d2f8ad46cf9da32e23b1c3eba5 : SUCCESS in 2h 10m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/129bd9052b854ad49b9a0d2c711a23bf : FAILURE in 1h 08m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/22dd22d7c605427988a19de5f2fe2b87 : SUCCESS in 28m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f6df3c1fdc0b486b9545fb82977bcdde : FAILURE in 1h 34m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b6894ca057384426a3229372b2e7b7c3 : SUCCESS in 35m 43s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f8944cdc5a0d44db8c92c9426a18e0f8 : SUCCESS in 5m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f3c7c6ac9004fc496fb7df9f9be0d9e : SUCCESS in 4m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/699f9ba61fcc47e2b47af1302b218fef : SUCCESS in 37m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0da2d2210ead427689370c4603280ad9 : SUCCESS in 29m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9283c03895e34f4d93a09490f6f02950 : FAILURE in 50m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/7caeebb1bab94414b71f7c58453e40a2 : SUCCESS in 50m 06s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3bea84b542ce41f9bdcfde9f55324434 : SUCCESS in 53m 20s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33667451ed8546979eb357fd13fcc96a : SUCCESS in 38m 07s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e66ce472dbd8428b890969b8415fe617 : FAILURE in 9m 25s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/557e4fde985145e4b0ddfe017fc110c4 : FAILURE in 9m 49s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/5b86964d45f54c43b1f5f8452ce4b685 : FAILURE in 7m 23s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/bfeaba0fcabf494286b836f1c60f55b0 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb781b7a6f883608c4e228adc314f0300a765f28","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-10 16:27:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest manila-tempest-plugin-ontap-dhss-rest : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"db10a51f35138c9e21bb142871c85f473777760e","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2025-04-11 11:51:36.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3253a926e72ac269815957836405698f568d791d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-16 23:13:21.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ec21c9558e0f7704b076d8731a38afe600147f5","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-04-24 11:31:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"aee7dcfcac8266089866f86a0ec0b26f040fe4da","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-04-24 11:33:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"260d730a5189e42bb527b16c9a36dde3b61d928c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-04-24 11:40:47.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9e2a87db7e8e188fb513d3e0579fce0e0a92e1fe","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-08 21:07:35.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bfc67126d7209f23057e9ec38509f682ca89c6a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-05-09 13:27:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6f39460c95223001b24df817a7a57e1bac8d6b92","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-05-09 13:27:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54abb94507a5d4298888a531ef37c9b064a71d45","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-09 13:46:52.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9f1542cf23ff495aba9fc2928c21e535\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e7b8ba7b33df48789a62f5ee03ebaef4 : FAILURE in 12m 19s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2469ecabf06243c7af3b4a29471069db : FAILURE in 19m 16s","accounts_in_message":[],"_revision_number":4},{"id":"8dd8fc0ce3c43b9b9a90f3d38fd322f8d32dc766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 14:25:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba209aab44b84064bf646dee35376c33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef4beba2a07405b904e3882f3855afc : SUCCESS in 12m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/916f647e7b3240b8ae88bf3a7dc6304b : SUCCESS in 29m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6afdd323c1b499fb40971d0bd47993b : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/553632b5b0c241fda4c8d831674e12e2 : SUCCESS in 9m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3348220c85384554b10e7e8b49d126b3 : SUCCESS in 8m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7cafbadc6f29456ca314bef82ff7ae4c : SUCCESS in 13m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e891e85fd6a4b1abbf66314082ba3e2 : SUCCESS in 5m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a180d50f0a994506beaf9a08d7c3be59 : FAILURE in 23m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a76516fe9096409c96103c9f4fb17c9d : FAILURE in 27m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e15727b2984747a3abd8c008e1d5628b : FAILURE in 24m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/84ce72eec25540e7a17184e74151d70b : FAILURE in 17m 05s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4cf29cabe6ec46719e42bb87be6d4074 : FAILURE in 28m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dd2363a8e1664e7fbe476748da49e732 : FAILURE in 19m 14s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9c75bec8da5843dc8e7793ef1370840b : SUCCESS in 4m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef4f2388fa694efa9249c71b5f9f6318 : SUCCESS in 5m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/73807d34726345c1995024d1e1916a7a : FAILURE in 7m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a941083669c45949e4e7b979503675a : FAILURE in 12m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d42ea9f8672d4b2ab7d0508ce77c9f96 : FAILURE in 18m 34s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/37ba5280c93a4f26a85efe39a4ac4ba8 : FAILURE in 20m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c3d8f95f8ece4e49aa4536b9331bbf1d : FAILURE in 20m 03s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a539e9af343a4f1a9c6c50ea6bff5781 : SUCCESS in 52m 49s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/94d4339f073e43ef99b038f496b785c1 : FAILURE in 9m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fbcfbd636c3148a79d645cd8b4fac429 : FAILURE in 6m 20s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/ed74cdf6d7ce447b9fdfb4528c4ca0bb : FAILURE in 4m 39s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ae76d3d1710a49e9a373ab6d218897dc : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"12476d050cf4cfb5ded3f3b19f2710cd266de416","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-05-12 08:24:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"96f104a6c4d7eda3de432990091d7c2991500af4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-12 09:01:35.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/782f88a7d62a4bb59b7bd002292689fd\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/62cb2ff2c3564c2396a8fce837bf2452 : FAILURE in 32m 49s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/08d5448dcc634151ba12e849dfe5e23d : FAILURE in 36m 37s","accounts_in_message":[],"_revision_number":4},{"id":"602ea013af90724bcdffbfdda74cbadfd89aee2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 10:09:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1c2dcd51d3434991735e1b1b42f934\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148675fdf582452eb6b69ba08a1dca0d : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c539717e1fce47218767d80a31950171 : SUCCESS in 22m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f0ef6f3ac94d3fa9b060c1b4582f11 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7dda4d82e214e1bb0f0d5973c3a84d5 : SUCCESS in 7m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e38482e67644a679f51f2bccafc41c3 : SUCCESS in 6m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76dfaba6e4794f9d89a6ffb3b0818ae6 : SUCCESS in 11m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39251e7286fd41659fb20da2a47cc1fa : SUCCESS in 6m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d5952577e86d428d86e239d0175fdae8 : SUCCESS in 1h 01m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/58791d03871d40139182c389b135a98a : SUCCESS in 1h 44m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1f388103e51d4c63a575f336f0f7f629 : FAILURE in 46m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/79ff671a0268452aa8f1d5f92f741e0e : SUCCESS in 1h 08m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a8b2d063cc8b4ccaa29992dc9fa15f7d : SUCCESS in 1h 02m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8c3d537ccaa485c832e6dd66cc62a98 : SUCCESS in 1h 09m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c6f1cbc3d405419e8a0cc71fc04a7948 : SUCCESS in 4m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bc1c7eb2a4834974bc8107af85da30ae : SUCCESS in 4m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eb5b8aad57b14314a9f3068c0e5a09a7 : SUCCESS in 31m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b9aa8ba4e30b4cf6817e19a001c1d3ac : SUCCESS in 36m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d2cc3aa2026241b5afb5b46730f077e9 : FAILURE in 49m 29s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a53e5d6869174b878f308d9508945f2d : FAILURE in 13m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/050361f7182f44ef8237baeef99fbfb7 : SUCCESS in 53m 03s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4dddd57c3d6e478fa25a00a0412eba16 : SUCCESS in 53m 22s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4e99f6eff12b45d58403325c4822e566 : FAILURE in 4m 19s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/84d9544d70bb424f84546681c4804b23 : FAILURE in 7m 40s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/1198bab5298246d0b2b5f5917482fbcc : FAILURE in 3m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1d2555788b4446118cb16914686878de : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b24b0d13cbf4eb7a8937106ecc42ea7cb08dc11f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-12 13:11:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest manila-tempest-plugin-ontap-dhss-rest : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"53385b6eba25b0fbe577f196f04e7219076af49c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-14 21:33:12.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"87ae87747099470941d5a317bce4a4c0c03426c2","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-05-21 08:35:39.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3c9cd429a7f4d767a969f6f130de8d6250117cd9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-05-22 15:37:17.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c39866b10381e77b8367870b0035cf411e0e0a8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-22 18:45:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2655a42739819a3b050911fc808d20a2ca05de4d","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-05-22 18:45:44.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d64b58b7f504c361cd268111ab14f1f8f98ad8d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:45:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"044edb79132574e8b98567ab6295e27af591112f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 19:21:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2582917203c1405a85e7eeb79f48c24d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0600c022f14982822c907feaf51aec : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9233580cd7843749f2f693330b897da : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9663f73ab9604f7bb5908e5c317c04c2 : SUCCESS in 9m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7916d4be67c14bafb1e93887cbf622a8 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb56fe65f3e544d4be39b192d17738d9 : SUCCESS in 8m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dd5c507b97064ec0b270480261a9c4a2 : SUCCESS in 31m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6d6aa43420ec4063ae435d44e4e391d2 : SUCCESS in 30m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/92d17070cca942f1a31a22805162073c : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":4},{"id":"ffbe6e24c308f71339f61a33837a4e1e2bddef6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 19:21:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bbc4569c9b2be01c6ed75cdb1fe6046920d76ed5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 19:22:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e81ab1d75f4bce80f430001205fc87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d9872faf5b4e0ba6df389052a4414c : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99a622b0568f4f7aba0c1cbf88e297c1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4},{"id":"7d92a4ab3a8ab4b0d45abd23972672e4331186ac","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-05-22 23:27:49.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/d6727403877e44da8f750823a2bb6a5d\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/443e906ba6ff42bd9cb05807faf0d09f : FAILURE in 4h 42m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0130010829bfc96b337088ae835a6dc75495730c","revisions":{"bc7fe368ffaf8a3afc32f53861356b4e7714a4e8":{"kind":"REWORK","_number":1,"created":"2025-03-26 10:32:37.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/52/945552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/945552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/945552/1"}}},"commit":{"parents":[{"commit":"fd8982e9c0134eccd69f6a4325aa3627173b2c77","subject":"Pure Storage - bump version number for 2025.2 (Flamingo)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fd8982e9c0134eccd69f6a4325aa3627173b2c77"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-03-25 17:07:27.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-03-25 17:07:27.000000000","tz":0},"subject":"NetApp: Fix export location during replica promote","message":"NetApp: Fix export location during replica promote\n\nDuring replica promote, the export location is retrieved using the\nshare replica vserver_client of replica under promotion instead of\nactive replica. This invalid vserver_client is causing NetApp API\nerror during _get_aggregate_node(). Fixed it.\n\nChange-Id: I6b809e093413a9a5d574023db73bbb01023f0e9f\nCloses-bug: #2104153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bc7fe368ffaf8a3afc32f53861356b4e7714a4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bc7fe368ffaf8a3afc32f53861356b4e7714a4e8"}]},"branch":"refs/heads/master"},"90e85b3d853048689903b1f2675b2bee01fb4ff9":{"kind":"REWORK","_number":2,"created":"2025-03-31 05:42:44.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/52/945552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/945552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/945552/2"}}},"commit":{"parents":[{"commit":"fd8982e9c0134eccd69f6a4325aa3627173b2c77","subject":"Pure Storage - bump version number for 2025.2 (Flamingo)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fd8982e9c0134eccd69f6a4325aa3627173b2c77"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-03-25 17:07:27.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-03-31 05:42:38.000000000","tz":0},"subject":"NetApp: Fix export location during replica promote","message":"NetApp: Fix export location during replica promote\n\nDuring replica promote, the export location is retrieved using the\nshare replica vserver_client of replica under promotion instead of\nactive replica. This invalid vserver_client is causing NetApp API\nerror during _get_aggregate_node(). Fixed it.\n\nChange-Id: I6b809e093413a9a5d574023db73bbb01023f0e9f\nCloses-bug: #2104153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/90e85b3d853048689903b1f2675b2bee01fb4ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/90e85b3d853048689903b1f2675b2bee01fb4ff9"}]},"branch":"refs/heads/master"},"be50c8dc2899bc5a06d8f1e33738d7014962818d":{"kind":"REWORK","_number":3,"created":"2025-04-10 11:05:30.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/52/945552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/945552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/945552/3"}}},"commit":{"parents":[{"commit":"f183a05976036219b142e787b31984cdd575f34a","subject":"Merge \"Update CLI commands in share basic operations documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f183a05976036219b142e787b31984cdd575f34a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-03-25 17:07:27.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-04-10 11:03:10.000000000","tz":0},"subject":"NetApp: Fix export location during replica promote","message":"NetApp: Fix export location during replica promote\n\nDuring replica promote, the export location is retrieved using the\nshare replica cluster_client of replica under promotion instead of\nactive replica. This invalid cluster_client is causing NetApp API\nerror during _get_aggregate_node(). Fixed it.\n\nChange-Id: I6b809e093413a9a5d574023db73bbb01023f0e9f\nCloses-bug: #2104153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be50c8dc2899bc5a06d8f1e33738d7014962818d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be50c8dc2899bc5a06d8f1e33738d7014962818d"}]},"branch":"refs/heads/master"},"0130010829bfc96b337088ae835a6dc75495730c":{"kind":"REWORK","_number":4,"created":"2025-05-09 13:27:04.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/52/945552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/945552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/945552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/945552/4"}}},"commit":{"parents":[{"commit":"e1a0e856871669790c88ab3deb000ec638253904","subject":"Merge \"Fix CLI Commands in async failures guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e1a0e856871669790c88ab3deb000ec638253904"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-03-25 17:07:27.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2025-05-09 13:26:38.000000000","tz":0},"subject":"NetApp: Fix export location during replica promote","message":"NetApp: Fix export location during replica promote\n\nDuring replica promote, the export location is retrieved using the\nshare replica cluster_client of replica under promotion instead of\nactive replica. This invalid cluster_client is causing NetApp API\nerror during _get_aggregate_node(). Fixed it.\n\nChange-Id: I6b809e093413a9a5d574023db73bbb01023f0e9f\nCloses-bug: #2104153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0130010829bfc96b337088ae835a6dc75495730c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0130010829bfc96b337088ae835a6dc75495730c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
