)]}'
{"id":"openstack%2Fmanila~945656","triplet_id":"openstack%2Fmanila~master~I25cb02fa5e4b87e9adf28f51157522f301b87e32","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-01-21 00:13:23.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-01-08 15:27:30.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-01-21 00:13:23.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-01-21 00:13:23.000000000","reason":"Change was submitted"},"36761":{"account":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"last_update":"2026-01-21 00:13:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25cb02fa5e4b87e9adf28f51157522f301b87e32","subject":"Stop allowing generic backups for non-NFS protocols","status":"MERGED","created":"2025-03-26 21:24:30.000000000","updated":"2026-01-21 00:14:24.000000000","submitted":"2026-01-21 00:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"945656","meta_rev_id":"03391b010e5ab90f65cb8f99ec87367af9486607","_number":945656,"virtual_id_number":945656,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 00:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-09-12 20:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-12-18 08:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":1,"date":"2025-09-18 15:55:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-01-08 15:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 22:35:57.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-03-27 00:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-27 03:42:10.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-05-27 15:27:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-05-27 15:27:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2025-05-27 15:27:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-09-10 21:37:09.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"state":"REVIEWER"}],"messages":[{"id":"ab8d488c81e8f02511eefbe5183460d5902a37f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-26 21:24:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbc9182dbe4ad63e6fc09b58885e8ba24b261bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-26 21:32:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2d458034f193c08677044af83195ee510e57dfd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-26 21:33:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a07c03ec2c62881cc02961aa692e897355a6eec","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-26 22:35:57.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2648a27d883a425a84be5a36c98f646e\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eab0556366af4dc89a0a4951b3863f02 : FAILURE in 28m 23s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ca70a2f74704c1c9a394434f69ac827 : RETRY_LIMIT in 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d5bb3149b91f93bf6ab6c2c34db8d73eaff7afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 00:40:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06060bc5302a4f6ca4f3b753bdcc9306\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa0cba5f560472ea8feb9f2acd19f86 : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d286f5cf7d99435dacc23354cf1f7605 : SUCCESS in 22m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c11cba12ee44812a0adda47c308d22b : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d1dab5157204b21a9bab335cf5a0e5f : SUCCESS in 9m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a95e23776364bb28bb39f3142bc3adc : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/234c580ed1f040989544f4ef882f55a5 : SUCCESS in 4m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/887471bbd0ab4bcc98043466855924f4 : SUCCESS in 1h 22m 06s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/8a8d7ab147f44757be3f23aa1f612c47 : SUCCESS in 1h 05m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/33a99bf7bf4a4545a24be85d9498408a : TIMED_OUT in 3h 03m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b73c3e9921de4d75a6ea53db457f0c80 : FAILURE in 24m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b13e5616d9f1401eb057e6b2eafc4a20 : SUCCESS in 1h 00m 54s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f0e3ca69e2c545a8bdfdf6d84d7941aa : FAILURE in 1h 12m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/58ceed39e5664dec923bd4ae1dc6df27 : SUCCESS in 1h 10m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2bae4eafef2d4314ba43d26aee1ff536 : SUCCESS in 5m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9aa76296a11541a5a65e51dd6b6a8142 : SUCCESS in 5m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c921e5d4625f4fa28eab13dbf4264474 : SUCCESS in 28m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5922178499f64025932915480ae11b28 : SUCCESS in 34m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/873d2bcd94ba4413bd6e2953aa8b9b33 : FAILURE in 42m 47s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6a82e1e949444eb7bdf42f901e1a1cd0 : SUCCESS in 59m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a1e989f450a148418566058f12813cb4 : SUCCESS in 58m 34s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b01f98ecdc93463da1448230cace9a1d : SUCCESS in 51m 40s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b00eebeab6ad40ada4a4effae8c9276e : FAILURE in 6m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/08ea28853b4b45c6939541b93a213a29 : FAILURE in 6m 16s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/7d9bf747d04947f1903f98b8e9e36d17 : FAILURE in 5m 13s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c02ac10a2dd2498092b87e41a210fb03 : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b4890d492d47ea65efc9023216a45ff1cd43e53","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-27 03:42:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2e35105b8b9c8fc7c64bf4ea13f94cb73ebb45e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-28 17:36:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d6fdbc439e1bfb0127ec13dcf7ba5242455811a9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-28 18:09:04.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9229d6c55b90479aaa65ff4f3371f8c7\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84d857e63f5c4890bb5096fead652ee9 : FAILURE in 31m 49s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fce2de0045694de58bb44e5d87450bbf : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d305c83153378694e6c04d084a2e5699bd1e9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 20:18:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20cd236d4f424e88bd5d84b08872526d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/316350d1af964f4bb04421875fe39f68 : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77749dd26a464df4bea68cb13a9c016d : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65c69df34d70431e94a4f4a8f7bad1e4 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5cc8bad5cd24a8cb1d7c7a452b7b5a5 : SUCCESS in 10m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/962d34e837284bb68372846b6de41aa5 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fd4e4c315724ded9a82af6138d8f9fc : SUCCESS in 7m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dbec3ae2ce304f48b728179d3b0d5674 : SUCCESS in 1h 06m 29s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/c1c62fff75eb4975ac69c8fc10519a33 : SUCCESS in 1h 04m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b5b4f69891b04d4cb61c3f346f3cb964 : FAILURE in 2h 14m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d285dbf7be9a4f90b9ed71235509b77c : FAILURE in 34m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dcd166455473494bb8cdd2718510d857 : SUCCESS in 59m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b69610473d8143268a9c26a666b30fcb : SUCCESS in 44m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/66f7e4c7da7d431b9b806c8e2e4e4430 : SUCCESS in 1h 26m 40s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1d4821dd61784919a51b8e0d20672fdb : SUCCESS in 4m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83991f9348c24a6ea5417e3e2b91caee : SUCCESS in 4m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e07b8c9ee4ca4748a696fc37a5e2237b : FAILURE in 29m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8bfc3938c92d489381fd3d17bd47216b : FAILURE in 31m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d94f9f2833754183805c7a6a7f1cc7cb : FAILURE in 42m 36s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ea576fd9376748349d8701c3da5f70fa : SUCCESS in 1h 05m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0f6b10f30b774a51946f86766f3746f7 : SUCCESS in 33m 49s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5dc8f39204064fd3b7078114754e3ed0 : SUCCESS in 44m 47s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/fd5db20158014e0f99ab57c2f48ffa2b : FAILURE in 9m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/061c6023223c4f0c8b831dc4705a4041 : FAILURE in 8m 37s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c13625dd518f46479213202e800ce32d : FAILURE in 4m 20s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f5640f498df340a7a1087c9b9a073c4f : SUCCESS in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4cc472a1230281fb3741f98a2b4d848111438a24","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-04-01 14:54:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e53842a56ab6766f2072aa6a1445d3d24325098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 17:10:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deddd2d68da74457bbd067c9270af1ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7f1e2cf33847528772f83dea98ca9a : SUCCESS in 12m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ecf22b0ca354a0b9823d835799aa1ee : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0723923fe98a4e638aafc665b80dea32 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dee58d95b87f46a2aea6f55ab15e2781 : SUCCESS in 8m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca1f4174e7624356abe496fbbb4007d7 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d4e9ade5f1645f8952e0969eeeb2d71 : SUCCESS in 6m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9b38bf6dd7f84a328bc58b9cd6d60f08 : SUCCESS in 1h 10m 01s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/6a5fe5ec51f04a3e81b773d09ec77a6c : SUCCESS in 1h 01m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/792d25552c924ac3b9601ce218b91171 : SUCCESS in 2h 13m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5dac4e29d8224cca918e4cc1382e219a : FAILURE in 47m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/953682bbbca34dfc80b99fb087fe90b3 : SUCCESS in 1h 03m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/77412af4502a4211a88890ad974e8d83 : SUCCESS in 1h 23m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/746f140196fb436fa63bbbbbda10a46a : SUCCESS in 1h 11m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9fd49ba320924654848e09520127f0aa : SUCCESS in 4m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/336fcb760b494c4db289f25e50464210 : SUCCESS in 4m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5572f1ebed214e94ab9dbddbe1afe6d8 : SUCCESS in 33m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/319cfd7f6df54f779c427f21e4431f86 : SUCCESS in 27m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/854280b6454141159a30dfacbd9e8250 : FAILURE in 45m 59s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b57e259a5ada43c4aadf6a30756d4f7a : SUCCESS in 1h 07m 53s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/50f7e85dc5dd4fb2a3713601ecf8accc : SUCCESS in 41m 23s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/618e2d30a9eb42018ccf359d6441502e : SUCCESS in 56m 19s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/02df67911345411c8716513cab8babe8 : FAILURE in 11m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/28a7383e59de42aab37ce0bd793e57d6 : FAILURE in 8m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c46d55bd2f3f488cbf5a25d1323ee422 : FAILURE in 4m 10s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c3fc30e68cd843198dbd6dcd87d92fbd : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d74758296a20817f485d534a2c37c759cbae578b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-01 20:12:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"1947cc851ccfe8d1cf445d28634136938b8259d5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-10 21:37:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb9934a78e41665dcb638bfb7a0a7cef63d4bc5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-09-12 20:17:45.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"222bb0eaaa86dc7990dfb9436fc4e7c8e97825ab","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-09-12 20:18:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab23af4f4cd1e7b8a7ed9f2fb11b13396f1a0af0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-12 20:19:15.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"f2a63147d616f861e7dc0a559e88c7bfbb95fa6e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-12 20:19:58.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0c94874d060f6e69dcf96545011074241b7a137b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 22:59:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b79f8258858495aa681af6f5560bd57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178142c5e7ae498bada92c3517fb0b04 : SUCCESS in 11m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b68875689f0411db78f9fc68601c9ce : FAILURE in 24m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9d8a4580fc49568449dc7e0e5c02fc : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5e21f52de304d2984e03df62754abb1 : SUCCESS in 7m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/120e7e1c82774850b211766fb93659ae : SUCCESS in 11m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/724cb47303ca40a0a0a051d1eb5e372d : SUCCESS in 15m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/954f78a2014e48b7b52fa323783a126f : SUCCESS in 8m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/493eb7dfa9f14228bfbd36d42b389d0b : SUCCESS in 40m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e2f5c11ae73549f1b78cc0d3ee72d43e : FAILURE in 2h 36m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9618d53892b44867b655e346c62c79d7 : FAILURE in 50m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/94bc6417d16d43ce8b0a1e828be6ef87 : SUCCESS in 44m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ab557c739c784ddca75ba60d58029f76 : SUCCESS in 52m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d5e693dd2a18457188c09fd47f93ebf2 : FAILURE in 22m 40s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e0953bff651a41b09a5461100b386fdb : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/99ab6437a0714de7a4d6acba8969129c : SUCCESS in 3m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fef219c621784913bbe4bca4d6069f0d : SUCCESS in 34m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/45537c5ef6e1498c82a1d3072f39e7e7 : SUCCESS in 38m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2c91bf5dc31040d3aa7ab868eb3e4032 : FAILURE in 24m 21s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/3be73b4759914eada8a1ec911788c05d : FAILURE in 18m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c6467fd95bc84ef9a7215218a6d5bcb5 : SUCCESS in 38m 12s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16c8a0a9c11a4ea89ac6d20c61ba6c4b : FAILURE in 24m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/43562f9df39243d29a039ddee6aae928 : FAILURE in 4m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d03268a2952441ab895cff38c90ba0f3 : FAILURE in 4m 52s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c0fa2a43ec0a494ca060332fa8fdb68f : FAILURE in 4m 41s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4109ea0588dd4480b0efbaad35aeabd2 : SUCCESS in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2c94b97c2016bab83a541c23d9a3b8dec7092466","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-09-18 15:55:41.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e165916170747591913242469feccfa320885472","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-16 15:34:21.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9753d61de46df1fca004d309330a088b0c64ad33","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-12-18 08:01:05.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9e9524722ff6bb60418b9761d35574d670d8350c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-19 18:16:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bd520cd4c5d0695304024e9a352002cb511b2cdd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-19 18:16:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6d60c025feb0f1674e186d7e2f4f327e3b46b468","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-08 15:27:30.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c668b0af363f61cbc77a7832c2d04f5888c883d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:28:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"70242738129a9b899e3307b72e1ea895a0f0ca80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:08:13.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c38367c652da48b8abdae47360943a79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db77c24cba0467aa89aba27d7949dd4 : POST_FAILURE in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc4454c3a754be69ef84c4930666c12 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bc0a3017a004d348d6c09d8a0ff6df9 : SUCCESS in 8m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2cd9dc7c8dd4b5b8f70950c6f761b27 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/563fcf45713848fe88106dca1bae2e85 : SUCCESS in 8m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ad15b34e1bd945798910cd6c02c2c5d1 : POST_FAILURE in 36m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5ad75d75e0254bc793537276b4f79b56 : POST_FAILURE in 34m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/49b9f314835743cab6f566811361ade1 : SUCCESS in 29m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ce7e32a9040f42dd923fd57b47c13542 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":4},{"id":"626a195ec98a819e07281a578121c69ce3a7bb78","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-08 19:26:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"20e7e11d9d3213f65960bfa201f5c86bd6052791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 22:01:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be2ff690c0974257bd4ac3653ca186a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aac342f0a3f460b97a04057aa4641b7 : SUCCESS in 11m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ce99e88dd1340f2916ef822dd2a1afa : FAILURE in 32m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928f571ab6af41a28abadc0ac911228f : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9f11959b0bf4061b92d4d14ca311ee2 : SUCCESS in 9m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5694aa3a011646cd97b5cdfcaa6fb06c : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26ee24dccbcf428c8576ca5a33e2206c : SUCCESS in 8m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cdc71eea064e4c79b3b967201ad7696b : FAILURE in 1h 45m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/42cdd7799f7348e891d171fd6675fa03 : SUCCESS in 2h 13m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/041886028545434da55603de9ddb7e89 : FAILURE in 51m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9c011fa1697841b399de1f8dece606b2 : FAILURE in 13m 31s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f2e85e8f99f84821a0cc61d23d32d601 : SUCCESS in 1h 27m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f31f6cf31cd5482aa487ed6f190454c0 : FAILURE in 1h 28m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ba49c377d74c4ca5b194f8a2bcb4b703 : SUCCESS in 4m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b0f41e8fdf714e85b263804dedb86c26 : SUCCESS in 4m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8db6d128bbbe497ab39337fd30c18c43 : SUCCESS in 37m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a53eeccf618542cb9004eb8dbe551412 : SUCCESS in 32m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af45494e39a04ba99957c2665239cf83 : SUCCESS in 48m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/233ca9a768d041e9b13a019988d20ce4 : FAILURE in 56m 18s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/72b2362c4d7e4253b129cfd17f4d0d36 : SUCCESS in 1h 00m 56s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0807b112183544fe9cc1a06d5c3a8796 : SUCCESS in 48m 12s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f14026f31564c88a0f721c8b42527f3 : SUCCESS in 59m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ffb62b4b38464c72bfe467db720432d5 : SUCCESS in 30m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c1a027b0c7cb424b9fa285f326df7ab6 : SUCCESS in 26m 56s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/83bd4cc9050c477f80ed0cc119fe1ed2 : FAILURE in 3m 46s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/fdbad0e6a5a6486d84ee5b727df687d0 : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f067c1cd377e2b164247e555823fe521c2f260ee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-20 20:44:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cadeb0fb4b4ac013c3129ddc568f67f9044a9fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 22:58:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/225aa5a34e3b447db338b3d5f6e8dc0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca28b4ba27444b8ac8257d9c41b57de : SUCCESS in 12m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f70257403a0a40768429cb07cf630e5a : SUCCESS in 32m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a452e4864614e3bb20a5891f26c59fb : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40d7648b25d74d85ba17e9af9941f0b1 : SUCCESS in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f32b44a310f94d73bc22cea6c4028505 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1ab8087959d418b9ec9f4774960876c : SUCCESS in 9m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b72a7ed5167a4cf1aa7d259a39d07925 : SUCCESS in 1h 08m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cf5c7c98c82d4423b3a74aa88fffe1c7 : SUCCESS in 2h 08m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ee29861a4ff44019bfc3fef9a5a074a7 : FAILURE in 58m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/08d00466fb1d441e90511460f735b341 : FAILURE in 23m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6c8fe1abdd4b4e0083fc0f84f036cb4c : SUCCESS in 1h 35m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3df491d1ef044e0ea81ad791950573a1 : FAILURE in 1h 40m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f106d4bd7b0945ff9d903b669fd80ed6 : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4deb807c2c124ed482ef50bd2df190d7 : SUCCESS in 4m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e7b47005b7754173948bc7a767ba269b : SUCCESS in 32m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/93867693bec44560a89350fc62f898c1 : SUCCESS in 23m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0d1ff0e8e06d4cd08ede89571cf78da8 : SUCCESS in 47m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b8131a71dcdd4b15935ab7f38951dddd : FAILURE in 1h 10m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/7b736bde8e91441aa31c849162e66c48 : SUCCESS in 1h 08m 57s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8b8d767c6f524f46b09630e36d246022 : SUCCESS in 59m 27s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0cb867118f144dfc8d32f7731d7857c3 : SUCCESS in 56m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/15b85ed520464ca0bead8858875f9f40 : SUCCESS in 31m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/55ee6f4b63654c599fd153a13f0e3650 : SUCCESS in 24m 21s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f8d51dabbbd9456ead724408c4636a20 : FAILURE in 4m 47s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/76734b883a294c81a0c42c4bc6f25e98 : SUCCESS in 3m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"714dcac25ebed69b9ac4f9662584a034c4311ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 22:59:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"398f3f10fa9cdbce30f7250c734f344c1e0d6bd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 00:13:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1c0091cc0f24e9891f94a4c34a96862\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40aeb1eb3bb47e386fa31ceb44476f6 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428b561fb7e64f6092e854295c60bc94 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21b40164c6764a7da1b22116835dd428 : SUCCESS in 8m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06a4b303403545379fd989f6a2c6afdb : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29b2597a7e774d9f84acc48d1eae6d84 : SUCCESS in 8m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5277526cf4914241b7c88f9d16b839fd : SUCCESS in 1h 13m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/275b8219d73a48d58699c970decd6ebb : SUCCESS in 41m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4da8356545714c8c972424e83a562946 : SUCCESS in 34m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fb4b3f59da53432bb49e2c1744497153 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":4},{"id":"7d5831ce0f48cb1e1b132c4ae789b4b5335424d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 00:13:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"03391b010e5ab90f65cb8f99ec87367af9486607","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 00:14:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c08b1d8f3c74f808269f5f99e72985b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2fdead186b6427291b0ebc70a9a162f : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51a5744a4b7d4fffbe8b091bd6ecd2a0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4de1de0f05236dddf54afe4fea509d97b09ac75c","revisions":{"d82475f5eaae8df21923ff72a7e4426eb8e39240":{"kind":"REWORK","_number":1,"created":"2025-03-26 21:24:30.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/56/945656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/945656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/945656/1"}}},"commit":{"parents":[{"commit":"5a93b39467af61546e7382086cf9b6def5a3abea","subject":"Merge \"Skip installation to speed up pep8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a93b39467af61546e7382086cf9b6def5a3abea"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-26 21:21:21.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-26 21:21:21.000000000","tz":-180},"subject":"Stop allowing generic backups for non-NFS protocols","message":"Stop allowing generic backups for non-NFS protocols\n\nThe share backups feature has a generic approach for share backups\nusing the manila-data service. The generic approach will mount the\nshares and transfer the data over from one share to another, and\nin this case, this is done with backups.\n\nHowever, we currently only coded support for mounting NFS shares,\nbut we were not checking for the share protocol prior to attempting\nthe mount, causing the mount operation to fail.\n\nWe fix that by checking if hints for third party backup solutions\nwere added to the backup information. If not, it will use the\ngeneric backup approach and we need to ensure that only NFS shares\nare allowed.\n\nCloses-Bug: #2088063\nChange-Id: I25cb02fa5e4b87e9adf28f51157522f301b87e32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d82475f5eaae8df21923ff72a7e4426eb8e39240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d82475f5eaae8df21923ff72a7e4426eb8e39240"}]},"branch":"refs/heads/master"},"b98a4637b624f14f6650f0ba0645f8bf4be634d2":{"kind":"REWORK","_number":2,"created":"2025-03-26 21:32:02.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/56/945656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/945656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/945656/2"}}},"commit":{"parents":[{"commit":"5a93b39467af61546e7382086cf9b6def5a3abea","subject":"Merge \"Skip installation to speed up pep8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a93b39467af61546e7382086cf9b6def5a3abea"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-26 21:21:21.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-26 21:31:46.000000000","tz":-180},"subject":"Stop allowing generic backups for non-NFS protocols","message":"Stop allowing generic backups for non-NFS protocols\n\nThe share backups feature has a generic approach for share backups\nusing the manila-data service. The generic approach will mount the\nshares and transfer the data over from one share to another, and\nin this case, this is done with backups.\n\nHowever, we currently only coded support for mounting NFS shares,\nbut we were not checking for the share protocol prior to attempting\nthe mount, causing the mount operation to fail.\n\nWe fix that by checking if hints for third party backup solutions\nwere added to the backup information. If not, it will use the\ngeneric backup approach and we need to ensure that only NFS shares\nare allowed.\n\nCloses-Bug: #2088063\nChange-Id: I25cb02fa5e4b87e9adf28f51157522f301b87e32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b98a4637b624f14f6650f0ba0645f8bf4be634d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b98a4637b624f14f6650f0ba0645f8bf4be634d2"}]},"branch":"refs/heads/master"},"ae8d5bc1eeb2aebabf1073c065d97c97eca4304e":{"kind":"REWORK","_number":3,"created":"2025-03-28 17:36:45.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/56/945656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/945656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/945656/3"}}},"commit":{"parents":[{"commit":"5a93b39467af61546e7382086cf9b6def5a3abea","subject":"Merge \"Skip installation to speed up pep8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a93b39467af61546e7382086cf9b6def5a3abea"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-26 21:21:21.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-28 17:36:28.000000000","tz":-180},"subject":"Stop allowing generic backups for non-NFS protocols","message":"Stop allowing generic backups for non-NFS protocols\n\nThe share backups feature has a generic approach for share backups\nusing the manila-data service. The generic approach will mount the\nshares and transfer the data over from one share to another, and\nin this case, this is done with backups.\n\nHowever, we currently only coded support for mounting NFS shares,\nbut we were not checking for the share protocol prior to attempting\nthe mount, causing the mount operation to fail.\n\nWe fix that by checking if hints for third party backup solutions\nwere added to the backup information. If not, it will use the\ngeneric backup approach and we need to ensure that only NFS shares\nare allowed.\n\nCloses-Bug: #2088063\nChange-Id: I25cb02fa5e4b87e9adf28f51157522f301b87e32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ae8d5bc1eeb2aebabf1073c065d97c97eca4304e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ae8d5bc1eeb2aebabf1073c065d97c97eca4304e"}]},"branch":"refs/heads/master"},"4de1de0f05236dddf54afe4fea509d97b09ac75c":{"kind":"REWORK","_number":4,"created":"2025-09-12 20:17:45.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/56/945656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/945656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/945656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/945656/4"}}},"commit":{"parents":[{"commit":"5a93b39467af61546e7382086cf9b6def5a3abea","subject":"Merge \"Skip installation to speed up pep8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a93b39467af61546e7382086cf9b6def5a3abea"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-26 21:21:21.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2025-09-12 20:17:44.000000000","tz":0},"subject":"Stop allowing generic backups for non-NFS protocols","message":"Stop allowing generic backups for non-NFS protocols\n\nThe share backups feature has a generic approach for share backups\nusing the manila-data service. The generic approach will mount the\nshares and transfer the data over from one share to another, and\nin this case, this is done with backups.\n\nHowever, we currently only coded support for mounting NFS shares,\nbut we were not checking for the share protocol prior to attempting\nthe mount, causing the mount operation to fail.\n\nWe fix that by checking if hints for third party backup solutions\nwere added to the backup information. If not, it will use the\ngeneric backup approach and we need to ensure that only NFS shares\nare allowed.\n\nCloses-Bug: #2088063\nChange-Id: I25cb02fa5e4b87e9adf28f51157522f301b87e32\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4de1de0f05236dddf54afe4fea509d97b09ac75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4de1de0f05236dddf54afe4fea509d97b09ac75c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
