)]}'
{"id":"openstack%2Fmanila~949613","triplet_id":"openstack%2Fmanila~stable%2F2024.1~Ie14fb4db0da3b20daadfd55647c8563e65a338d2","project":"openstack/manila","branch":"stable/2024.1","topic":"bug/2109645","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-15 11:41:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie14fb4db0da3b20daadfd55647c8563e65a338d2","subject":"Add pyproject.toml","status":"ABANDONED","created":"2025-05-13 11:46:13.000000000","updated":"2025-05-15 11:41:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"67d36999764921abdd9bcf887379cf4630439547","_number":949613,"virtual_id_number":949613,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-13 13:27:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 11:46:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2025-05-13 12:02:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-13 13:27:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea6bb342b65c21f557e7acf5f67a2416d94befe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:46:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"877430dba92c789ee035a17aa1021e2a8f92c5ef","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-13 12:02:04.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b5176bb2765545748a76cc9004b6e143\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ccfd14a109f7434cacdbbf7d70b6d0c0 : FAILURE in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"083c388665f3cbc71e28950ff1a598e2f3d205d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:27:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/680cd672db124afbb4374aa61c9b6c99\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c1abcc9cb6814cf78c2394297b121df3 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572870dfb42e410a8176d34b7dc69d14 : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17c1f51807884b869c806a9eeff80293 : SUCCESS in 23m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ed6873e965466e8d0123d6bdebd642 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5ff82d627514fd2a9723f822cad44c4 : SUCCESS in 9m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b377a86af68f4e2caad230f58f9f6f71 : SUCCESS in 7m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d502924a58cc4fa7826a81423ec74949 : FAILURE in 22m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e8f7b8f591ae484d992581f28ace7d95 : FAILURE in 26m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c4f215b2a4c649ac820365fc99bc7476 : FAILURE in 19m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/881ed8a8b24446bbbeae2ccb1a519fbc : FAILURE in 31m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aaf939ef56f048a5a42738ef3993968e : FAILURE in 23m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6240235d7c99499689c34046fdedda1a : SUCCESS in 4m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34e9871022844fc090ea9b2d03e2b703 : SUCCESS in 3m 18s (non-voting)\n- manila-tox-py311-with-sqlalchemy-20 https://zuul.opendev.org/t/openstack/build/66e45c4620aa4d52a120992f5d4261a5 : SUCCESS in 10m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1b876973677d439d9238a1933b5c98be : FAILURE in 11m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0f70326ac72f4b13816c44ab835f1651 : FAILURE in 15m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fbd84dbabc894c978e5b011fe58d78f7 : FAILURE in 23m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/56c70ff3289449dfab851387dd4c8b6e : FAILURE in 36m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/25bc65eee49c41bab5278d161fe032d6 : SUCCESS in 35m 41s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69135954d6fc47aaa707cf94a28e289f : SUCCESS in 1h 09m 56s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/76001dad019146d38b8240fea80c8602 : SUCCESS in 32m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/29acf0bfab1e4233936ee35979a8dd38 : SUCCESS in 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67d36999764921abdd9bcf887379cf4630439547","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-15 11:41:57.000000000","message":"Abandoned\n\nWe don\u0027t actually need this here right now. I suspect things will break come October (when setuptools removes setup.py install support) but we\u0027ll have to come up with a larger, OpenStack-wide fix when it does.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d75ab3c1fbf1f2b77c757bf3558d7ffd6d6250b7","revisions":{"d75ab3c1fbf1f2b77c757bf3558d7ffd6d6250b7":{"kind":"REWORK","_number":1,"created":"2025-05-13 11:46:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/949613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/949613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/949613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/949613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/949613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/949613/1"}}},"commit":{"parents":[{"commit":"cc0f148566039d0c3577be22b6ba24e05c8631f4","subject":"Merge \"Fix access rule visibility locks\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc0f148566039d0c3577be22b6ba24e05c8631f4"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-04-30 20:24:50.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:46:05.000000000","tz":60},"subject":"Add pyproject.toml","message":"Add pyproject.toml\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml. This change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\n[1] https://docs.openstack.org/pbr/latest/user/using.html#pyproject-toml\n\nChange-Id: Ie14fb4db0da3b20daadfd55647c8563e65a338d2\nCloses-Bug: #2109645\n(cherry picked from commit db97a9f322a136604bdcf6846a2c4020cecdc868)\n(cherry picked from commit b4168f42dfe7ddc23a81a00b4f471f803de465cb)\n(cherry picked from commit 5010ff01234316612d3d9ede859f029dfeafc8e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d75ab3c1fbf1f2b77c757bf3558d7ffd6d6250b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d75ab3c1fbf1f2b77c757bf3558d7ffd6d6250b7"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
