)]}'
{"id":"openstack%2Fmanila~952736","triplet_id":"openstack%2Fmanila~master~Ic377d336972c55ce3cd59549c6b1c034e062bc52","project":"openstack/manila","branch":"master","topic":"bp-rename-isilon-to-powerscale","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-08-12 13:59:36.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-12 15:57:44.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-08-12 15:57:44.000000000","reason":"Change was submitted"},"36132":{"account":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"last_update":"2025-08-12 15:57:44.000000000","reason":"Change was submitted"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2025-07-30 13:59:39.000000000","reason":"\u003cGERRIT_ACCOUNT_36725\u003e replied on the change","reason_account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2025-08-12 15:57:44.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-08-12 15:57:44.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-08-12 15:57:44.000000000","reason":"Change was submitted"},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2025-08-12 15:57:44.000000000","reason":"Change was submitted"}},"hashtags":["powerscale-rename-manila"],"change_id":"Ic377d336972c55ce3cd59549c6b1c034e062bc52","subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","status":"MERGED","created":"2025-06-17 07:32:40.000000000","updated":"2025-08-15 09:08:17.000000000","submitted":"2025-08-12 15:57:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":70,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952736-bp-rename-isilon-to-powerscale","meta_rev_id":"3141daa74656c1fef1478352e122a89c43da2221","_number":952736,"virtual_id_number":952736,"owner":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-12 15:57:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-08-12 13:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-08-12 05:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2025-08-12 13:59:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}],"CC":[{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 08:09:38.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-06-17 09:48:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-19 01:53:16.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2025-06-23 06:47:31.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"CC"},{"updated":"2025-06-23 13:22:29.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2025-07-01 06:15:47.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"state":"REVIEWER"},{"updated":"2025-07-01 06:15:47.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"REVIEWER"},{"updated":"2025-07-01 06:15:47.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2025-07-01 06:15:47.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-07-01 06:15:47.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-07-09 08:50:22.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2025-07-09 13:47:30.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-07-09 13:47:30.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-07-14 07:24:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-08-12 22:43:06.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"4936e418505d9ac223b97df43b7fd6c45f89081f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-17 07:32:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9634f490e2658a795b96e26d79affd2d1823e937","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-17 08:09:38.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/09543094ed6645188fb11b41ba45617c\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/91cf2e5a6bf04beba3ee6fdc7fc4b8e5 : FAILURE in 32m 00s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/654ed2efe60b4d79a59897c378b574cc : FAILURE in 36m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fe0c60aacb08c9dee8830dd50b2b031afb98ef00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 09:48:08.000000000","message":"Patch Set 1: Verified-1\n\n(59 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31d64f03bd19420195090f6fe8097d52\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a0f55104beed40b39c08005fb4b63ea0 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2889e466c784b15aa47e6660e3c3a1b : FAILURE in 9m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bea74840a5fb40378293a2acc2d449d5 : SUCCESS in 30m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6540809b7fce4f9397e56a9d2563b983 : FAILURE in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7710ceaa00a345e99da9453357fed40a : SUCCESS in 10m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c54f38303f4849198ef2b921dc117848 : SUCCESS in 9m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f91fe22a1aae4f2f9577799c67f912c9 : SUCCESS in 15m 06s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/88563d2aa2214911b722d9265655d884 : SUCCESS in 57m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6fcf10fc373d4016ad7dfc6293d9a83c : FAILURE in 2h 11m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/694dd2fe0ff44f91ab83f2eefd1a9cf6 : FAILURE in 1h 09m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3c8100ca0632475ca2f54c56377023f0 : SUCCESS in 1h 13m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/06ab1da06ac5493ca836f1578ab816bc : SUCCESS in 1h 14m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ab2216d79b4949f1aaaa561765ac4b66 : SUCCESS in 1h 05m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3c44b06627f2471fab336a383e675062 : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d09ba7da7ae440092f1e5d9b476b14e : SUCCESS in 4m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/33f46df21deb43bba6f0445609c5724d : SUCCESS in 23m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d5c375700fd14fd89e146d980a8617d1 : SUCCESS in 39m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1e681e773c2145f58141bbc2b5a77ccd : FAILURE in 43m 12s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/881eb18461d64415bf3560564f96709b : FAILURE in 12m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e54a635fc9674925a6c24532feef34bb : SUCCESS in 52m 55s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aafd6576c15b4e6c89da0284e57c1ccb : FAILURE in 16m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/008ee2cc9e7b4e9f92bbed0022420d17 : FAILURE in 7m 49s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e040dc0fcfb14beab10e5e0ac39ab08e : FAILURE in 6m 28s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/5b7b151f56364694889758a393311e94 : FAILURE in 3m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/49b5292ea65e403c9a83966bffbe274f : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89ba4db41ccfac677359bb13287dd67365942053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-17 09:58:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d4a1719e48ab1761f8bcb7e58183eb87950daa33","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-17 10:41:23.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/41425ba81ae541b2aff233ee8c31931a\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c6684605d5b446e99cb9d5b778fdaaf : FAILURE in 36m 19s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/143d55ee921b434f8feaaf9435076e05 : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":2},{"id":"bfd69bda7a026b9be3bb8c7f9572a7289e3e77f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 11:53:24.000000000","message":"Patch Set 2: Verified-1\n\n(38 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/251bbc2c49d94fbbbb71bf8c455a00dc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aee28cab17a54d17b5c0424df9eac35b : SUCCESS in 5m 03s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ce4d896953a448d8dfb07b37318aea5 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93139b462f04a638173506114b3a7fb : SUCCESS in 11m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab7eb4636f848eb8bd8c11d8309da6a : SUCCESS in 31m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a13656bd688437683f7ff74646866f5 : FAILURE in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ff1e961d42e46d3918e1b237ef97374 : SUCCESS in 10m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/974c91a5c06f476ba2092cfa959dd56d : SUCCESS in 9m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e297afbba4b491787d06e95016ee00e : SUCCESS in 12m 02s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b3a01577ea94f95b80a7d8a956f7110 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/413fe9d5859e441fbcc82eaef09a3239 : SUCCESS in 7m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aedcfff9a94344659e6dc0355b09ec1f : SUCCESS in 1h 08m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/87a600dcd8bd41d4b2c115f9ff8c3d93 : SUCCESS in 1h 53m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ec4b3dab6a3b433cb85d04b712adcb08 : FAILURE in 1h 10m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/39addca45d0f4786866e7286c2502ea2 : SUCCESS in 1h 00m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a6a3a7047de444fca22a491e97a30b04 : SUCCESS in 1h 24m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b0aec5a9f806407dae73a19a0fed2049 : SUCCESS in 58m 24s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/298125f8274c4d34b347d494dd5a0033 : SUCCESS in 4m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e58746e5ead94df1805bf7d6e9ad433f : SUCCESS in 4m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/80f53a0c3a0049708c1f6cfebaf1ce7a : SUCCESS in 27m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/89bad3c3e2d946bc9e591bd111bb9be9 : SUCCESS in 34m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a9f49ac27f1e425f9452e3dc1ae1d9bd : FAILURE in 39m 47s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/948b72489f3346ba8729ace968090a27 : FAILURE in 12m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/84515237dea24001adeff40de755a59c : SUCCESS in 37m 24s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4f71b0863b04ec8af5681b9debe237e : FAILURE in 18m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/140346582303410da04ad2dcef7f594d : FAILURE in 6m 05s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/20768122eef94a7d93761a0e7e335dd6 : FAILURE in 5m 46s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f80e08c8394345a293ef51c77710d105 : FAILURE in 4m 24s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3d10e7bc339d420392735e7603996e86 : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87df18673fea62dc0eee1ea60e30d5a96077e938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-18 02:49:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"28bbe4cdb0fcb76c72e339f472744990472d6d0f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-18 03:31:36.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5213c27496c8499db53e563639bfa0df\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/221690c3dad44230886aca3ded21d034 : FAILURE in 34m 55s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c9068391dc02416fad4ff63cf5ff9232 : FAILURE in 41m 02s","accounts_in_message":[],"_revision_number":3},{"id":"c3f89f9e8efb5e47fb477477e3b8e4896c9067ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 04:58:09.000000000","message":"Patch Set 3: Verified-1\n\n(34 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a401346e54d4b60b66f547fb1a97a51\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2f0f4215eb4949d29af32e64f4d55a56 : SUCCESS in 5m 29s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f24d2dbec96470f9874544e690f0cd5 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e0638639504b53a11cccf3a5a69934 : SUCCESS in 12m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d660d6b538403c806b114b98d14bcd : SUCCESS in 29m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd554d72b63a41bdbf0d52e5b04e310d : FAILURE in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3006badd10304fa18ab2e1189ad6cd4e : SUCCESS in 9m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4078dcbb81f34f6385cfc129e9f1fdcb : SUCCESS in 10m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c00fc8d74db648609818b146c6adc27e : SUCCESS in 14m 00s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c890f453e814092b2e54fcdded480b8 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c6dacdee84450a9067e12b242b1d15 : SUCCESS in 9m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cc65691650d341c8b0027af37004832e : SUCCESS in 1h 36m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4b09da3b8d52436897ff9358600c015c : SUCCESS in 2h 04m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1b39d7d227a5417bba923f8bd6555ebd : FAILURE in 1h 06m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d12bfa1da8a64d2aa05c3523e82fe2db : SUCCESS in 59m 33s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f1d563c0ea33485998391bcf7befe55c : FAILURE in 1h 30m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/224e5b3bc466459e8f949a458f1d724a : SUCCESS in 1h 05m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3535fd0264be4d4986b152d46be62be3 : SUCCESS in 6m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f42f94173ab46a59aff0a5b27617df2 : SUCCESS in 6m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/08bbe0820e4b4f84bf6cf74e7b3e5827 : SUCCESS in 19m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1977258cd3194f4a8ec489e30802b522 : SUCCESS in 39m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f59c9eeb026e454a9b9150df0b2b5f74 : FAILURE in 46m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/3dde847f990849dbb63388fd902daaf4 : FAILURE in 20m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/245c2ae0c75f400591e2074b84b7e3b3 : SUCCESS in 50m 43s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/acb7d82c81194816b3ab4817d7cdbc14 : FAILURE in 21m 49s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4495cc47201249e9bec49402e92272fa : FAILURE in 9m 45s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4c0a5270eacb4b97bb0d91e474963ea0 : FAILURE in 10m 24s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/e89aa906674b4ca295d886299a0e1107 : FAILURE in 4m 52s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/cbf512c7f4b547aeaa6b4b6ddf5ecabd : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b684e52df4d2f28ab423b15d8b3e56341964feea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-18 08:02:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"eb4123120846b7b8156771edd38f8e857a480b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-18 08:05:27.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"1a951d2679913229ed6277241228775cc81a5d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-18 08:36:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"333df96ef66801d914ed945771414655fbfd3132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-18 08:58:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fb9cbdc072e3ea455317eedcd82da22bbfadd8bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-18 09:22:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d99c3ce8b42e890bbd518762f088f371bbce1c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-18 09:34:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8e54f293c32d5caa59c76f469e9f8945bc2caf32","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-18 10:20:41.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5eac2bed1e354ad39866aacb2c885525\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2861273b495b415a99de2aeae34b8064 : FAILURE in 34m 41s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ec291eaedd5442fae03c6e8a5c554d9 : FAILURE in 39m 31s","accounts_in_message":[],"_revision_number":9},{"id":"fae02a1e6b47cc18749fe18982f31e577fd712ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 11:58:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d54792e0bb844a191f4be1506dd3270\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d67d1b33743842beb2f0fb1a689ab398 : SUCCESS in 4m 29s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c3b7802347c4814b5043241699cee55 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae087eb5d4ae41fab8a0c5a854778b3b : SUCCESS in 11m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50090a919077478da3342d7a5c9c441a : SUCCESS in 26m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4615dfa714fd4544a29ce66ded0531e6 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dba4615d54ae44b0bd74591406525bf4 : SUCCESS in 13m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/492f3a349ba54a559f158912d3ba15af : SUCCESS in 12m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41015a64575647e58bd27419d297a191 : SUCCESS in 16m 07s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a7bacaeb10a4b01a7502f35085eed4c : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d73db8d3ed8e46738bab58f08a5ef258 : SUCCESS in 8m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a49ee936100646e8ab562076d4f86491 : SUCCESS in 58m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0a010f2150a645c7bba0248e823083db : FAILURE in 2h 21m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2576219afb74460fa1d0ef9d31f1446e : FAILURE in 55m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7b4897846fa04d318ca682569c227be5 : SUCCESS in 1h 07m 33s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1ee7826b8f5f4fce80b6daf58982f441 : SUCCESS in 1h 17m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a26d0c0a86294d53b912e95b4926bdfd : FAILURE in 1h 12m 24s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/927d61cbd51940c798bb563ee686098d : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5bcae49c3f34fe9804ed77b736945db : SUCCESS in 4m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bf2fdecf1f3340cbafee3fc6b7de9d23 : SUCCESS in 32m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1a564815c03e47c2ab8a269b7622b8ec : SUCCESS in 29m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7295211858f94754b318c04161c5ea46 : FAILURE in 41m 18s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/401339dada8142168037564c82374d76 : FAILURE in 13m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0475a2c224e640d8b1fbfe33bb812cae : SUCCESS in 46m 23s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5130d43b20514b68ac9fc2aa4d43c1f0 : FAILURE in 17m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/94e9139580a1491185fffffc8008eb82 : FAILURE in 6m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6a05add752b24a06afa8dfcb7fc55bb8 : FAILURE in 6m 03s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a9375e40cac34531b286313aac7b05e7 : FAILURE in 4m 48s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7324d309d3614af2afa0fa64f4f1f3e0 : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"31d358554bd783fb8eebacf64f34a8aa1f8fb4d6","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-19 01:19:42.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"69d66ad9e9f0f62fdb51846090f85ae95b3bfb97","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-19 01:53:16.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/e2f2c22c2ff44572af6c959e44f17bac\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/59736b689bfb40ee816dfdfce3d58b80 : FAILURE in 33m 06s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"320c4207e601c8af72a8b9ea88827fadc43df921","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-19 07:54:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6f9d8104ee33135c4d74d9a1ea132b5ccc5f06d8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-19 08:26:17.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/c01725a5d1f44354ba4540d62cf2997e\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/3ebeaaad685a40f3a4a9c1e44ab24a78 : FAILURE in 31m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"b5ffac3b968ad45aac04b43c5e2ea43c73bdb827","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-19 10:54:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"59228a110e497c5a6d0fa6a372258c020de6f9df","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-19 11:26:47.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/6886d2bddff448b28704ad6f583aba2c\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/ad31f9c9a69548d5b68cf9a80e90fa6f : FAILURE in 31m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"abec58f88f575aee070613c4685d7b99bb412902","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-20 05:48:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9b5df9227559e67a39e0ec7fd8cb7ccec080fff5","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-20 06:20:28.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/6f01f91f90a344b8a36b08bd124c2805\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/2ed8955c58af4abfb65b92e471dee992 : FAILURE in 31m 14s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"0ecb0e06a73d36601a97f719c7a60b23b69d2785","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-23 01:39:58.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0400a06d887927117b79217ab8e6657a8cbbef4f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-23 02:12:51.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/5b8bc277444247dcb764ee5b2e4e51b3\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/29bfc36a19354125baf6fb99af5905c3 : FAILURE in 32m 24s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"fe1af98f9fd26a86636fefe55e8cecd075895564","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-06-23 06:47:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"926cd2a3506f38fc813acf2fcea5d1ce18f16fe6","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-23 07:19:02.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/39e86912a4a646b8ad34a71b4b34edf6\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/38ca9966bbe7472ba58bceafdb5d4cd6 : FAILURE in 31m 07s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"0a45d4a367a9c93be7690f7ac1b3dd36fda4ae9f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-06-23 07:38:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"330edf49390c9bab62de6c147327c31bcf0a6a2c","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-23 08:10:11.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/d9a4796696f547eab3c03d85fdbef59a\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/7307d80c25d445fd84e75ef33a48425a : FAILURE in 31m 39s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"d1122c17befe570d71e99df3f6f6dac31cf07e79","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-06-23 13:22:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"772c947ce046ecf5cbff2979514c95ee37982152","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-06-25 09:30:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"77eca5a64ae6d56ff6158f7e78b2faae8b1f6591","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-01 06:17:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f6db131d0d946ddc26ae53d30f6bd99919041c09","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-01 06:53:27.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/18d70a0a78454fbf954048646516a770\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/6ee9c027f32d41ef9ca3d1b41f3ca4c5 : FAILURE in 35m 53s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"93ec2b08caf24bff1c18e0592957948dbc3b7c46","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-07-09 08:02:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ac9f74ed306bde645296d7ca97cfb54130efadf2","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-09 08:36:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/03de76630ae34b66a51b0a20dc2eabe7\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/f6e80b49f408405491e6f9c1d13218a1 : SUCCESS in 34m 04s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"56e0cecb8a2db1228710cfede50287a584c46236","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-09 08:50:22.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"13e3684e304daf6dacfd48c316e1b7c539b80273","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-09 08:53:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9a1bbf9b58a762be494c3782ecf7f1280e6d8a52","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-09 09:24:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/4e7f8951b09e49748100dda428a7f2e3\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/47de42cd53b047dda43392f0d7661742 : SUCCESS in 34m 10s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":9},{"id":"1caeb26fee2292e99d830239a50c3c6e8802b3f6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-10 20:49:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"18d95e1df4fa5b264c13e5b2e391fc1f9835d41d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-10 21:49:19.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"04cd752629a0638779ca41fd7defabc51c53dc46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-11 09:52:33.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"84573c32e5d46027e74741dcba074528870da3b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-11 09:56:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"14ffb1292ba5b7e54238c0009250f0385f278ce9","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-11 10:01:00.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c72595f69456abab605d7eb506657358995b641d","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-11 10:01:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ad3c994778d6497a5dcd9825cd50e9a2b2bff1df","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-11 10:42:06.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fcd9bb4c1efa4f0eb7884c21655e2f67\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/42d253b82ac644679eb73a0d7269d2ef : FAILURE in 40m 02s","accounts_in_message":[],"_revision_number":11},{"id":"21bb6684a8ea426946b2b07af7bdf2a0a1aa4b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:55:17.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28d11d4ee8914dc588114520fabe6801\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/533de1d292214940b1f45a7de958b24f : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc10fa33897441cacb27479f2d8ba90 : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2a4a66fad874d8c9996c2f3abf1844b : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6deddaaae4c14399a5ca7711e3899850 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db0f144888244b778b7e76c7f5d6ef05 : SUCCESS in 7m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/379759ec83c04cf59338aeb7c84ebdb0 : SUCCESS in 10m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3cbed4845c9d4bb098d6085aa96f1328 : SUCCESS in 13m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77f6ec8685be4c268b6e77e875457260 : FAILURE in 7m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/13b88aa1050443368e197dcfcddc605c : SUCCESS in 1h 01m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/36fe342f9be94a048fdd6fc3f6150483 : SUCCESS in 1h 57m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77faf51f9b1242ac9d1de8d196a2a3bc : FAILURE in 1h 02m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5d9cfedf58b845b78592a14e4066d7c2 : SUCCESS in 54m 05s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/af0024a1f1064481ab4d9a1a6988a1d6 : SUCCESS in 1h 12m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/76b6c916102c4479a89c7eec697c8bad : FAILURE in 15m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/79b4d66209ef4ed3bc9eba1fb43b174a : SUCCESS in 5m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b3a7715b058f47f3bff71be5244a52e0 : SUCCESS in 5m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b1b7e736c571479aa4f8f73abb8a7d51 : SUCCESS in 24m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6c1688bc42c74484ab42eee2850f33c4 : SUCCESS in 29m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/378a16776b834bd1889454e0a7d47916 : FAILURE in 16m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/994248fd4c034a84baae7e77821b7db6 : FAILURE in 15m 31s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/abee0b7ad1c5470095c3a5c14d732f4c : FAILURE in 10m 43s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7bb6cb4ddc864f6f93081ac3106c0c8f : FAILURE in 18m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5ce2b268eb7a48d6a50649896eeb24bd : FAILURE in 6m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/06d467fb86bf40a9adae4feb4ee4c4b2 : FAILURE in 6m 26s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/fa30bb1f74d44be99fe5e0b3b7e3643a : FAILURE in 5m 32s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/797ccb4089e149adbf7fbbb9a8932771 : SUCCESS in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2c2e5dc5c9226a8036e9fabb4d7b010007b7e024","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-14 03:06:16.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"c15695bd3a265adb266014566081fe44014f740f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-14 03:51:50.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3ee2de2cb1734df8813c559a89b88a50\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/10ad7b4b7c0a4999b8392a96ce3e4a4e : FAILURE in 44m 49s","accounts_in_message":[],"_revision_number":12},{"id":"4cde9a5c0c8aad439ab6365e6e99bae29fd959a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 04:23:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba666d70f8c40a69160588b9bd4efea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/171d0500684141869f477d245b1570f6 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4efddb99465b40ad9b51d60216960270 : SUCCESS in 11m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a72d8149e14422d8f3235373e82024c : FAILURE in 24m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09877266db28432db827d9d12ba6e63b : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68e6efa8a5234261863d8030a82798a0 : SUCCESS in 7m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80a1bacd3ce2496b83ee25c1b206b950 : SUCCESS in 9m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed0b86dfe269435eb671963a4e0a1bed : SUCCESS in 12m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f6d4a2e1a5a4146b3d0551d5007775c : SUCCESS in 7m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/34f496d233a347eaa882d010e959e67c : SUCCESS in 1h 08m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/11a11305ae5b4180b4992a560b1a13d6 : FAILURE in 24m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8aac22b21ee54702bd7b157f53269135 : FAILURE in 1h 03m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b48b49c8e27743589ccb6cc94f4b4ff1 : SUCCESS in 1h 06m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b90a98d588444163ab6e96346b3e8016 : SUCCESS in 1h 14m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aa7a30a96e904d139ff4887d852383bf : FAILURE in 20m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f215d46c2e6f4159b32f2f657a9e8232 : SUCCESS in 2m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c75b51f343d4316baa81dc39824532c : SUCCESS in 2m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f1a77bd9945414299657fbfeac370a1 : SUCCESS in 20m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/33518eca15604a2cb225fe8a888f4a37 : SUCCESS in 41m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9856bf2117d14c4dbf7c2e98bb587a7f : FAILURE in 22m 37s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5dce5b1f89d34695b19a7a3f5c801f53 : FAILURE in 29m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/57298efd8e5843b18f50bb06e3a5aa13 : SUCCESS in 53m 45s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dbd639ecf5c84097ae9564e3edaa28d7 : FAILURE in 11m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5b50f16ecb54402aac87ee79e7396fa4 : FAILURE in 6m 05s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b750b1c1894f48d0a205ed9d5f1c7d4b : FAILURE in 11m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/71c47e05f1e9483c819ba70fd68287d7 : FAILURE in 4m 26s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1645d8111f344ca09e38a81e1880daaf : SUCCESS in 7m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b906ade87169e5b7d1baab5fd509bd0f20582331","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-14 07:24:28.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"a59e173556337920a0ca025ae147f7083ebf1767","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-07-17 08:28:53.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ba43903582200b078e0acfb0b81156d799bc00a8","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-22 06:18:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ce8e8cc3e7fe1041980ca8c89755fd15e76a2bd5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-28 21:11:31.000000000","message":"Patch Set 12: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"17c02a6108b025ff8142bd18df58a84d4a2c1482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-30 05:36:50.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"67f9e43c5c92ed75f1eb9f0218e03b2665b0bed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-30 06:07:05.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"34f8cb4c866891b1dac80a78edf693b7bc0db0c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-30 07:41:24.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"aed5751df233cdcc8a6213b75e464dfa28d4eaed","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-30 08:15:13.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/788c4c217acc4fec997bcf03d7dce3ee\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01b345910b384901a4fa10c909c94ef9 : FAILURE in 33m 06s","accounts_in_message":[],"_revision_number":15},{"id":"d7663d29cc994005fabc5c9814e0d0f56c80b9e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-30 08:27:31.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"d357c6976af8899b1cb4e0f825c7cf468c199005","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-30 09:08:20.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/99c146fae9514a3bb26562bd44783e1b\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7aa654003fca414b9542e895cc18c994 : FAILURE in 40m 05s","accounts_in_message":[],"_revision_number":16},{"id":"e28f690d9f45b458c72d279e509f24a47562fa5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:15:28.000000000","message":"Patch Set 16: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ade5cfa6c37c4f44a5969271a98849d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df29e1041f26432792d937bc2b7653ec : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387acfbbda4e4b418ca58c4ed5171ce6 : SUCCESS in 12m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd04369877c74f8da270070705ba2d3a : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7452263fd5c421787b5a5fcb020871f : FAILURE in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03fa80753f40486bbfa5547b0cd48331 : FAILURE in 9m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02c722f2bdd449b7b38f64cd5ecd7675 : FAILURE in 10m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60d1523caa9946a899a2059b442f5272 : FAILURE in 12m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92aa801799bb441299732eea296482da : FAILURE in 6m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8a2bb4faedfa458bac30c6a2fd93a70a : SUCCESS in 1h 04m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/52142b718e764c388c57182a2ca4703c : FAILURE in 2h 42m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5aa98fc213b64da19836dffacf5e9dda : FAILURE in 1h 11m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bcd3202f0ed1473c884bc6b32132a821 : SUCCESS in 47m 01s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d783c4fbbd514eb18cfca952ccc87acd : SUCCESS in 1h 11m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a235ccafbb6140a381e0054c60d3a34c : FAILURE in 24m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bc32bc3d2e084477a244dcc2bf420d09 : SUCCESS in 3m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dddb8e77f8464225b005e068ace94c18 : FAILURE in 4m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9b02c24d42ce4e49b9f808aea9b7ad79 : SUCCESS in 38m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ffc3d49082a34af18fcd243363a4c477 : SUCCESS in 31m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7de35d60ff7c4d65943e3de8daecfcac : FAILURE in 15m 01s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/742d981bcbff4f24a2040fc9ccf58c76 : FAILURE in 38m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/50dcb5b26c98499a859ec64fcff4f776 : SUCCESS in 59m 40s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b120517a59a84811b04f628321f5788d : FAILURE in 23m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4cb2a747dc854df1b58ccb396d748b49 : FAILURE in 5m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/14279db96b824308bbef29aa4d779e7d : FAILURE in 5m 36s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f4ffd4a37d374ce78e4a42dc01dd71db : FAILURE in 4m 23s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/cb2eebe2f56645528f7e8423bfce56e3 : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9b985ad54b6e46ca854caa2360b17959790a4b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-30 13:51:37.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"fe4fa99fe888c9ec4413900f19feb76b5b3202f9","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-07-30 13:59:39.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"603f8d0517447ac9f4a0e659c8b19991b13721ca","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-30 14:33:32.000000000","message":"Patch Set 17:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7760b3380e514587a716e87610e80352\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a09bc38917ce4eb8b0b7c163da25a023 : FAILURE in 41m 10s","accounts_in_message":[],"_revision_number":17},{"id":"8fcfd466cedf8d09f5e3f7af8db9578a0407b4d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 16:23:35.000000000","message":"Patch Set 17: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e81b212a6609480b91a3620448b0b4eb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ae822c4268884785bb2c38a293c8f1e3 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b0220813464cef92e96647451ee1c5 : SUCCESS in 12m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db7b177f66ca45f58f76c6849cf77335 : FAILURE in 24m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b5b2b5dcd2495d89f993bbe399975f : FAILURE in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70e6e00931a34581aa9911efa8f258d5 : FAILURE in 10m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cf4cad70030408c89102b981a56b7fe : FAILURE in 11m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36c87a8ea78f48b69b33bd13e793133c : FAILURE in 12m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a71039a9b5d49639c0530a68058efc9 : FAILURE in 8m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/95aa338afcf848f1b2937e57a9124dc0 : SUCCESS in 1h 12m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e074910baac34f72bec7dc36479fd8ec : SUCCESS in 2h 26m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6111c4aac6f249b5a9ffa3ca37e18241 : FAILURE in 52m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d156baaddc7a452c80bd0695ea40209b : SUCCESS in 1h 12m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/fcecf5d2689f40da97b4e1644e63417b : SUCCESS in 1h 29m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c37391afd2dc48fdb83f7a86874aacda : FAILURE in 22m 27s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7c50d9943af84d5480b73d109c00c4af : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17ffbb8a63d146729ba063bd4dfc9561 : FAILURE in 6m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0a08bf396e5e4ed49fdd228c985e7f73 : SUCCESS in 18m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2dd1a365d4764a8e9ad077fd95d3beba : SUCCESS in 25m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a049511d67e5468b9bbb31f7af096109 : FAILURE in 26m 03s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/20176c6dc563401c872e87ef35acc528 : FAILURE in 36m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0a9b9577e7de42048c731ce4392f6ba6 : SUCCESS in 59m 25s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2108ed6c86714a788a6318aebbf068bf : FAILURE in 22m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/63bb6d28e7844472a72a7493d879f7dd : FAILURE in 13m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3d707840d2434c288299a8219d0f32d1 : FAILURE in 11m 05s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a2d403f7adb34ed4bac9bb6e3b014563 : FAILURE in 5m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/8784cf0b45ff41a6ae00c157b4d27aa5 : SUCCESS in 3m 16s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"14edb0db437e178ed085643af54251aed28c58b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-31 02:56:27.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"b96fa8c211745cdeabd88ad1ad3f4a536f6f9d9a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-31 03:08:38.000000000","message":"Patch Set 18:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3de2270edf6e4bd489b5df88f5584d68\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/05e25cc81ca545f08c4e85afed5cf6d9 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":18},{"id":"cee41acabc5cc0880fc2a4029e1e967749fe3f98","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-31 05:45:05.000000000","message":"Patch Set 18:\n\n(6 comments)","accounts_in_message":[],"_revision_number":18},{"id":"792e7080169180f7469e09221e5e8b199d4bb8e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-31 07:08:44.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"189ce68f4ed7bcbbbd8774db1e1ec54d66f22386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-31 07:40:30.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"73c789887e02d15ed352d18d44f1fb307269b83f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-31 08:28:17.000000000","message":"Patch Set 20:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cdf9d09693424a36820af2411d5709c0\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d71f11aaa91a4bdc85e045b206451b3d : FAILURE in 46m 59s","accounts_in_message":[],"_revision_number":20},{"id":"3d156ab8ba44d9f0bb4c28bab92e94c6d6e0b736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 10:52:22.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0226c34ee94427aa87116935e9b4f91\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d1fa40afca6a4958b494786563fd8a84 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79b70a5e7554fbbb53314850df85c96 : SUCCESS in 5m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6629d0d73cfd40a1b5cdb7c98f4efc9a : FAILURE in 22m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4070313afe84ab9a6e05fece86fbfef : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/778a43d3c8a949a58b1b4f54c73a72a2 : FAILURE in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f19e76e7c4794584abfcec504aceaafa : FAILURE in 9m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d44652f1c93b4f9c9d774fc1dc91a087 : FAILURE in 15m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f15b81276e834ffbb63fd32b44f239c0 : SUCCESS in 4m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1e95628da0e8457189331c5c74d60d82 : SUCCESS in 1h 22m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/93bd4083947f46c19c70c29324d769b5 : SUCCESS in 1h 20m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5c85f443e85d4b948a4a8e461cec998d : FAILURE in 40m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e0268acf24a54cbe8e3561a95e1206c7 : SUCCESS in 43m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/de872ce76926490e8f75e82b48fb748e : SUCCESS in 52m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/69ddae56c6ef4934a37467f513387b7e : FAILURE in 10m 50s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/510cdcc7e00e4a9a9c8bc3448657029d : SUCCESS in 3m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd90dc8bef4d43f1a291c9733d2e2408 : FAILURE in 3m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eeb0e9076d6b4e7e85b6aca234bacfff : SUCCESS in 33m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/61d42ad6cccc474c81915a9370dafbf9 : SUCCESS in 36m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/48c1d7bbf0a8423c8320e9f653fd3087 : FAILURE in 14m 24s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/626b18806c664e229ed25e207a724c13 : FAILURE in 20m 06s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/35a572a2fb7e4a90a8b12c29f641e1f6 : SUCCESS in 40m 23s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4bc40f71cb27498cb6a9c3054cf2ea0c : FAILURE in 20m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2ebb078f60ab43dca95a5718f4e063f2 : FAILURE in 11m 53s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8127841454e0466ab6df928e765a193a : FAILURE in 6m 50s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/ceb402c553cf4ec5ae5c6227632af1d4 : FAILURE in 5m 32s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e78e62e4950e40b2ab19203a98548c6c : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"31dbd67d6aec11a26ada4ad1455d97fdb1ef4043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-07-31 15:10:27.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"179bfc20aa8590f714ade18eae1a23068dfb43b4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-31 15:44:39.000000000","message":"Patch Set 21:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/27a4cd660ffc493a956d24ecf97afdf7\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c67a398b75924d29af159bdd57c6f9a8 : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":21},{"id":"413a8e28cb7112d6c2a3d75d1775fb29da6074e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 18:14:01.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3359c2e0f5aa4692882c44f5ca11dfc0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/622e31cb1ada4689a9f695a8dd595404 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438d4e68a5584647b0971972e5ff698f : SUCCESS in 13m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc4350aa9cc4967bfd27645dbd19992 : SUCCESS in 35m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/427ed2c8a64348bfae75fa5c947fab13 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11b0ecc7fb514465a0737d8d5c12d729 : SUCCESS in 8m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68cf9dced30f479abc707a9c07170dd6 : SUCCESS in 11m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8efd7ec25b9e458bab4065b27e2655f3 : SUCCESS in 13m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e9a4d79bfda49a38103ab76d16fe105 : SUCCESS in 8m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/20b6a784fdd64d1283fccbdc72fb195b : SUCCESS in 58m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/71581b9d396646a3be95c94ea97c194e : SUCCESS in 2h 30m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9c826f85f334436eaaebfcea955f9eef : FAILURE in 1h 55m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6eb13b3bcfef477a902f0e2e4c0ad7ca : SUCCESS in 26m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/43f1c65b63be4c089b40a2518acbbd0b : SUCCESS in 1h 14m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/46c8873dc6414764a97c19bc11239f35 : FAILURE in 17m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/66b51d9e0e174bac92e2703463949472 : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd4a93391a464d5682a2931e4a8b7691 : SUCCESS in 4m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b90be4e7f9bc4144b19cbd979f04d9db : SUCCESS in 29m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eb9dc6bc034f4e1999c158a179fc4403 : SUCCESS in 39m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5593d27351ef4670b49f75ca7be025df : FAILURE in 18m 30s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/03c1dd74d2a4449799384ce0d96fc74a : FAILURE in 35m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a0a2cb2a1976478d9dd2cb01f15522cf : SUCCESS in 52m 29s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e137d4fd028a4b68ae5660ba21c0a7b9 : FAILURE in 23m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8259f242230a485abd7ccb1cb3292ba8 : FAILURE in 12m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0cefaea44e6a41538ff0ae0e6955ab7d : FAILURE in 12m 15s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/811b686e1dcc44df9a11d612d0e3ebf9 : FAILURE in 4m 14s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4091944c762a41f0b9f8dee9734beffb : SUCCESS in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"49f698cdf15c4eae2f937c2211c1014a173d2b45","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-31 18:21:39.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21},{"id":"1ef9205bed8a83581f92e4fbbdb096cae2913099","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-01 05:07:56.000000000","message":"Patch Set 21: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":21},{"id":"e72436f18537017334ad21c8459f6fbfa00608d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-01 09:50:29.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"4d05abeac973f50b75003e59d626cc2196c03f1b","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-01 09:51:56.000000000","message":"Patch Set 21:\n\n(7 comments)","accounts_in_message":[],"_revision_number":21},{"id":"eb3bb5874afedb920c8b50c7bcf393221c6c21bb","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-01 09:55:17.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"c162735663d09fbe3d2821a73a93788f6ea1c3f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 11:31:48.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5acfc009aa4f6a9b14d3564cd1196d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4e650881a3b34deba8cfd9a24b9e7568 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3e688cfaad4a2cbf0567e75ea0e030 : SUCCESS in 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2d9ed4e79384ceeae18758bcd9ecca4 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e67ae25646647e190c157fbf9fdde06 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7460441f8054a64993caaa267cb47a6 : SUCCESS in 11m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6279b3907bc344608277edf7b204e5f8 : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe984f926d6643f0a5da66e2632394de : SUCCESS in 7m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9650cb985a945ae88207d47d4e66219 : SUCCESS in 4m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/adab02f18ab14e45b519adcae8d9b6d1 : SUCCESS in 1h 33m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7fe2b7b7a155425a950db2d015dedff9 : FAILURE in 1h 05m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4b5d15366e334731b143e50cd432db82 : FAILURE in 57m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/63cc9b9b1cc24e828ae81c203ce84ab0 : SUCCESS in 37m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d20c190258fd4240aeef489b4d8f92a1 : SUCCESS in 1h 33m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d05d2dc1e5e14a5d880b3324427839c6 : FAILURE in 10m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/02d3cb62a2294db1b59ab9f6b94c0bdd : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96e3c865617243c687a20ae77cc8147f : SUCCESS in 2m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8e8d9dc16b1d4f86a48d4c2fa8eac909 : SUCCESS in 22m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f661a8d296934445ac5116a10db82b9a : SUCCESS in 24m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa800006db0a4818be00928df373b3a5 : FAILURE in 10m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/73e411d90fdb497ca164de6c07a538e9 : FAILURE in 35m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/89b93c28bbf94ecdaf7958956742f58f : SUCCESS in 41m 05s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce404537a6bd4de8a334a40b638e627d : FAILURE in 21m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a75a2014f07c4e8f8af42a3207a7dbf4 : FAILURE in 10m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5151bbea814144a4ab819c5040766f96 : FAILURE in 6m 16s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c20eacd97e9d41cba4cb38eec39008cf : FAILURE in 4m 08s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/fc64fcc4adf844d99f807718e4aa9e06 : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"82dfdf6b396392ae2d69261c38fa9f9835327006","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-01 20:28:25.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"dbaee9aa639d8218d192b5f197b00fb219a9e388","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-01 20:29:33.000000000","message":"Patch Set 22: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"73786491eaca7309ac68abbc54022410aa15a5fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-05 05:43:18.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"0444a4cefa739f62bba3a56d29728b046674b6c9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-05 06:24:27.000000000","message":"Patch Set 23:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a23968aaffe94ef1b586db22678d368c\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/99a8f8b073e64e3ba568081fe56ab062 : FAILURE in 40m 16s","accounts_in_message":[],"_revision_number":23},{"id":"6d17a2a2882318e90e45877bf256d39741d2bd43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 08:14:26.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf3db7574d1d402eb73a282d7dae707f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4c7d0af80a64079a7540430e4923cf3 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb55abc32abb477fbb248758cdd69b90 : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cf7a7ce0d4842c795b818c92e04cec5 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b462b9471cb6419ea41824322439097a : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6873e7c04fa74612a16fe7c6cfdd2889 : SUCCESS in 7m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b990640c93f04aea976affdf7bba0feb : SUCCESS in 10m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3036fdf7d5c34c4597de60e8c6d3e3c9 : SUCCESS in 13m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/582bb0db830a41748f7752eb4ce62d89 : SUCCESS in 8m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d2b626dd2a1b4fd1a0c1d3b20d5d9411 : SUCCESS in 26m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/880facf8ba7a44e7a8f94025d7fa10e7 : SUCCESS in 2h 25m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/552c820277ca43c1b5574420e02dd40d : FAILURE in 1h 06m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7c4e6a992d524bed9505e9d14e0eee40 : SUCCESS in 59m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/31327ea87b4a4f498411fd7203a71eeb : SUCCESS in 44m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a7968be689fc4d289ab9ef968d437ea1 : FAILURE in 21m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/49311ffaf5e14719bb80226fb242c001 : SUCCESS in 4m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37b5eb57e52e4b1cb03948d22172bdae : SUCCESS in 4m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cd0eee50f8ad4dc6a443751ce45ab510 : SUCCESS in 29m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0d6942eea2bc4cf7909d0475a2dbfd52 : SUCCESS in 29m 38s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bb960b293ed54c8db7b4583783e4d4f9 : FAILURE in 22m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/01eb3efc0a6d44ccb3e6102949510c40 : FAILURE in 25m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d2bcd0b06ed444109a4510a1970df38d : SUCCESS in 54m 47s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5283b7c8f801478fb7145f011307ce8f : FAILURE in 21m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f5359da957704ff2896091b417a8b4e9 : FAILURE in 11m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/955a9d846f41452aaad8e1deb7a91c08 : FAILURE in 11m 21s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/d0a012ce8d02464cbf620c495761215d : FAILURE in 5m 31s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4fc1562b2edf418cadedbd7a1d844e05 : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"2b7bfda94d6b550686027a3edbc145f440f1ed6f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-05 08:26:52.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/952736/23/upstream-check/manila-tempest-plugin-ontap-dhss/c0f6d3c/ : RETRY_LIMIT in 4m 07s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/952736/23/upstream-check/manila-tempest-plugin-ontap-no-dhss/9880dda/ : RETRY_LIMIT in 4m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"5f7901cdd15937d7fcfce84c0a00e733672afebc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-07 14:20:28.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"688933d1a4f3c440ecd1cd0463c931b5177d3717","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-07 14:20:51.000000000","message":"Hashtag added: powerscale-rename-manila","accounts_in_message":[],"_revision_number":23},{"id":"47f6d7f36319f81e6d9f0cdc43496156dcc637c1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-07 14:26:53.000000000","message":"Patch Set 23: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"568dd252034a4b907af27033aa52c00be5ae42a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-08 05:08:00.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"d953011beb2efa3c2d346575d81573cd6d45fcbd","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-08 05:10:34.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b22c5f7519a96c1ff4cfb58f19c72057d4423eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-08 05:24:01.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"19cf0022d18dd229d8beea33acd5ca8eb0238319","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-08 05:29:33.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"947aed1da4b915dcb4cd9d43c68411ab36760cb0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-08 06:05:54.000000000","message":"Patch Set 25:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5c996fc325f5468188dae41620ee3bf3\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/028f45c8177048bdb10b00aa335f6fa7 : FAILURE in 35m 19s","accounts_in_message":[],"_revision_number":25},{"id":"32c7a8cd87868ec8ae74167601345d58ffdf3b50","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-08-08 06:07:07.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/ffd72e71337d45deb51bf1afd7ab3116\n\n- powerscale-manila-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d52c94a3c8b34a7e8d1f944d3369c764 : SUCCESS in 37m 01s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":25},{"id":"6af4da068e82abd89583381b59262cb7e4b04f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 07:31:01.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d372f7b7fbc4384924efa0ed8cc6732\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f0b538e919a043f7a237130a8db9bf68 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe922854cf64c80b7393aa2b7e87b03 : SUCCESS in 10m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9863233113f145e6936e669d6c70e536 : FAILURE in 30m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b3644a1c534de4b947150f4def3dc7 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/493f1af7fd5d45c9b594402f02b9ae23 : SUCCESS in 8m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd9ab7d8cf2f416d8c090988459a4e0a : SUCCESS in 9m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cda17fd709e4e7b89766251aca493c1 : SUCCESS in 11m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da71a87c5dfc43f0803700846f77af7a : SUCCESS in 8m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/15f6acfc4cc4413596fd14f8aede6ad5 : SUCCESS in 1h 11m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/abdf78ad06fa4c88967aa2a9324a27c1 : SUCCESS in 2h 02m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bfbf564212a041c59b33543eb4fff11b : FAILURE in 1h 12m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4fc3350fe21544d3863e243e8893227a : FAILURE in 1h 13m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/29fd42814c6248fe90308803ee8eb6a1 : FAILURE in 1h 14m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d85660877c394ad6914e85d4d8fab209 : FAILURE in 19m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ce9699037654cbb8d0b693f1866af09 : SUCCESS in 2m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd0263654fc74f5d8b89312f6bf465ca : SUCCESS in 7m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aaefc278b66d4446b47fd59c114cd808 : SUCCESS in 27m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fe25f9e9cbc947029aa46d7636824ae8 : SUCCESS in 36m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8b851ab2b9b74d9d8d3dc637a56a4ceb : FAILURE in 21m 59s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/3606bd0512694b9d821c2a8cf9d7f1cf : FAILURE in 29m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4bafa76150c243fdb9d048e4c3eba621 : SUCCESS in 46m 31s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3e412b10e844f618369639ccb5ab582 : FAILURE in 19m 40s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0ecfc5c4c244486594662f9c31684276 : FAILURE in 5m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b9c1100f1c0d4f64bf7d5944c535cf01 : FAILURE in 10m 33s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0a38c95e48b74376963442b5d811e1c2 : FAILURE in 7m 01s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1424af644f0444139311d1e4a21e4395 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"3a1cdaf2b3e226bd6ea8790d7883b1ce8bfd5b2e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-12 05:14:12.000000000","message":"Patch Set 25: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"7a4847a7efb0458c89af848dac55c89e68f65153","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-08-12 13:59:36.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"cb97d686dda389ee46e2f5341d86088bcfe4736d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 14:24:11.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"a322cee3cce1fb522498f1d157d4768e3ae12b34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 15:57:43.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cfa249d234a48649a1b7e4de4bbe6e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e062a7884e3a4b10b318693a7fcdc533 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40704e18ed794c9c8024a6444e8679c4 : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9fa9c0104844829b294ce91fcc22fca : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35bb189a91f944469b63b742538af47b : SUCCESS in 9m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70fb4382c5ac45b086d34cb708dd860a : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/620cd9d5d7b048c29e6f65033182608c : SUCCESS in 8m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b0c9af287a45428f844c63fe628a2351 : SUCCESS in 1h 19m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cdf3474563b849c6b44173dce49defe9 : SUCCESS in 32m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9d588e8694d04040b171a6a76dc12ae1 : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":25},{"id":"b0e74164dbb3b6afe81cc18128b81e2eeab86f85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 15:57:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25},{"id":"504cb1f359cf8b2a3a00d5de2e832161be07c020","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 15:59:36.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd035a2661184ac8b3b593db72db3271\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4f7e18e5ce48339d5308b1277a4f84 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aa6f32ef9ca4fb0a00f8df293042160 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":25},{"id":"a8796d939f1661d802ad444961d45fd2851dd02a","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-12 22:43:06.000000000","message":"Patch Set 25:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"b7d10fcaf64bcdd38f4ca098729b270acc6da89a","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-14 16:37:33.000000000","message":"Patch Set 25:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/69ada3be85f9489d9f13a5822dbb988f\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/add5a1d008214f6e814c02bf2c756159 : FAILURE in 12h 13m 55s","accounts_in_message":[],"_revision_number":25},{"id":"3141daa74656c1fef1478352e122a89c43da2221","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-08-15 09:08:17.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"677bf27e0cd4c9c175c3b2ef78da510f8a264dbe","revisions":{"83c8729b504ae13ef001a65754c3d59120561ac6":{"kind":"REWORK","_number":1,"created":"2025-06-17 07:32:40.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/1"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/83c8729b504ae13ef001a65754c3d59120561ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/83c8729b504ae13ef001a65754c3d59120561ac6"}]},"branch":"refs/heads/master"},"b9bd8a6800fba86d9fd117d3aae2ed02960392b2":{"kind":"REWORK","_number":2,"created":"2025-06-17 09:58:57.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/2"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 09:58:19.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b9bd8a6800fba86d9fd117d3aae2ed02960392b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b9bd8a6800fba86d9fd117d3aae2ed02960392b2"}]},"branch":"refs/heads/master"},"44e668365a648aac6c86ef84093cf59529487b59":{"kind":"REWORK","_number":3,"created":"2025-06-18 02:49:55.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/3"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-18 02:48:51.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/44e668365a648aac6c86ef84093cf59529487b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/44e668365a648aac6c86ef84093cf59529487b59"}]},"branch":"refs/heads/master"},"ff7bb6ca090d6ef4bec76997f3d2fd09b9504191":{"kind":"REWORK","_number":4,"created":"2025-06-18 08:02:28.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/4"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-18 08:02:07.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ff7bb6ca090d6ef4bec76997f3d2fd09b9504191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ff7bb6ca090d6ef4bec76997f3d2fd09b9504191"}]},"branch":"refs/heads/master"},"e6f7444374b915c92ce2cc32b6de5a20ed411a92":{"kind":"NO_CHANGE","_number":5,"created":"2025-06-18 08:05:27.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/5"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-18 08:05:12.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e6f7444374b915c92ce2cc32b6de5a20ed411a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e6f7444374b915c92ce2cc32b6de5a20ed411a92"}]},"branch":"refs/heads/master"},"07944e337ef1ae3dab0f56267574c0efe313ed15":{"kind":"REWORK","_number":6,"created":"2025-06-18 08:36:31.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/6"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-18 08:33:12.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/07944e337ef1ae3dab0f56267574c0efe313ed15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/07944e337ef1ae3dab0f56267574c0efe313ed15"}]},"branch":"refs/heads/master"},"50533a7e7c2f2cf893217d8488bd1628f6fe8f87":{"kind":"REWORK","_number":7,"created":"2025-06-18 08:58:41.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/7"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-18 08:58:27.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/50533a7e7c2f2cf893217d8488bd1628f6fe8f87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/50533a7e7c2f2cf893217d8488bd1628f6fe8f87"}]},"branch":"refs/heads/master"},"fda5fdcf959bf570891d0353d3956903ca57792c":{"kind":"REWORK","_number":8,"created":"2025-06-18 09:22:34.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/8"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-18 09:22:05.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fda5fdcf959bf570891d0353d3956903ca57792c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fda5fdcf959bf570891d0353d3956903ca57792c"}]},"branch":"refs/heads/master"},"88b10365b6ca86023ab32ca8b40b0caeb61329ad":{"kind":"REWORK","_number":9,"created":"2025-06-18 09:34:53.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/9"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-18 09:34:35.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/88b10365b6ca86023ab32ca8b40b0caeb61329ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/88b10365b6ca86023ab32ca8b40b0caeb61329ad"}]},"branch":"refs/heads/master"},"d3a0939145c83e704b9253c111fe0bf8173f10ba":{"kind":"REWORK","_number":10,"created":"2025-07-11 09:52:33.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/10"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-11 09:51:56.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d3a0939145c83e704b9253c111fe0bf8173f10ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d3a0939145c83e704b9253c111fe0bf8173f10ba"}]},"branch":"refs/heads/master"},"b39d76839e862c74ffe67c32ef6e9fc0bb788247":{"kind":"REWORK","_number":11,"created":"2025-07-11 09:56:50.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/11"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-11 09:56:37.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b39d76839e862c74ffe67c32ef6e9fc0bb788247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b39d76839e862c74ffe67c32ef6e9fc0bb788247"}]},"branch":"refs/heads/master"},"a9ef56d17495f557a8832615bc17b0d88b814fc6":{"kind":"REWORK","_number":12,"created":"2025-07-14 03:06:16.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/12"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-14 03:06:01.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9ef56d17495f557a8832615bc17b0d88b814fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9ef56d17495f557a8832615bc17b0d88b814fc6"}]},"branch":"refs/heads/master"},"a1541aee4d64b49518d3541f6889d345382a5b55":{"kind":"REWORK","_number":13,"created":"2025-07-30 05:36:50.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/13"}}},"commit":{"parents":[{"commit":"8de57938d745b23db8de72294209ec625c5e2b33","subject":"Dell PowerScale: add support for ensure shares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8de57938d745b23db8de72294209ec625c5e2b33"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-30 05:32:24.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-30 05:36:34.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a1541aee4d64b49518d3541f6889d345382a5b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a1541aee4d64b49518d3541f6889d345382a5b55"}]},"branch":"refs/heads/master"},"4f14c4be295542a1a6cc4ee10e78915d5e19f0b0":{"kind":"REWORK","_number":14,"created":"2025-07-30 06:07:05.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/14"}}},"commit":{"parents":[{"commit":"6209ce3ebaed952925d0ef0da4ffa10faa15ee2a","subject":"Dell PowerScale: Http auth issue, SSL verification error and etc.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6209ce3ebaed952925d0ef0da4ffa10faa15ee2a"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-30 06:03:37.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4f14c4be295542a1a6cc4ee10e78915d5e19f0b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4f14c4be295542a1a6cc4ee10e78915d5e19f0b0"}]},"branch":"refs/heads/master"},"e4cdac1ac6b7e657a5afe356aa20aef110ddca54":{"kind":"REWORK","_number":15,"created":"2025-07-30 07:41:24.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/15"}}},"commit":{"parents":[{"commit":"8de57938d745b23db8de72294209ec625c5e2b33","subject":"Dell PowerScale: add support for ensure shares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8de57938d745b23db8de72294209ec625c5e2b33"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-30 07:40:49.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e4cdac1ac6b7e657a5afe356aa20aef110ddca54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e4cdac1ac6b7e657a5afe356aa20aef110ddca54"}]},"branch":"refs/heads/master"},"ab20198b5f8977f2479430a527259242c31d54cf":{"kind":"REWORK","_number":16,"created":"2025-07-30 08:27:31.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/16"}}},"commit":{"parents":[{"commit":"8de57938d745b23db8de72294209ec625c5e2b33","subject":"Dell PowerScale: add support for ensure shares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8de57938d745b23db8de72294209ec625c5e2b33"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-30 08:26:17.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ab20198b5f8977f2479430a527259242c31d54cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ab20198b5f8977f2479430a527259242c31d54cf"}]},"branch":"refs/heads/master"},"6650c088280f0f8a9d26510ba0dc82d311913532":{"kind":"REWORK","_number":17,"created":"2025-07-30 13:51:37.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/17"}}},"commit":{"parents":[{"commit":"8de57938d745b23db8de72294209ec625c5e2b33","subject":"Dell PowerScale: add support for ensure shares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8de57938d745b23db8de72294209ec625c5e2b33"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-30 13:50:42.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6650c088280f0f8a9d26510ba0dc82d311913532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6650c088280f0f8a9d26510ba0dc82d311913532"}]},"branch":"refs/heads/master"},"96d5681ac78b5095232265d5ba2f6867c0d444d7":{"kind":"REWORK","_number":18,"created":"2025-07-31 02:56:27.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/18"}}},"commit":{"parents":[{"commit":"8de57938d745b23db8de72294209ec625c5e2b33","subject":"Dell PowerScale: add support for ensure shares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8de57938d745b23db8de72294209ec625c5e2b33"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-31 02:55:53.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/96d5681ac78b5095232265d5ba2f6867c0d444d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/96d5681ac78b5095232265d5ba2f6867c0d444d7"}]},"branch":"refs/heads/master"},"cf3b7269482393b9636da8de7d571f5f175e9b88":{"kind":"REWORK","_number":19,"created":"2025-07-31 07:08:44.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/19"}}},"commit":{"parents":[{"commit":"d74f72cf2d6cc881e75162d0b5a4d253bf3b6772","subject":"Dell PowerScale: Added support of thin provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d74f72cf2d6cc881e75162d0b5a4d253bf3b6772"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-31 07:08:13.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf3b7269482393b9636da8de7d571f5f175e9b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf3b7269482393b9636da8de7d571f5f175e9b88"}]},"branch":"refs/heads/master"},"99f7d07416c727c7f4e780a952bb022bebb6481f":{"kind":"REWORK","_number":20,"created":"2025-07-31 07:40:30.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/20"}}},"commit":{"parents":[{"commit":"d74f72cf2d6cc881e75162d0b5a4d253bf3b6772","subject":"Dell PowerScale: Added support of thin provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d74f72cf2d6cc881e75162d0b5a4d253bf3b6772"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-31 07:39:32.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/99f7d07416c727c7f4e780a952bb022bebb6481f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/99f7d07416c727c7f4e780a952bb022bebb6481f"}]},"branch":"refs/heads/master"},"8d2d1b631b352c96fe9fc7455e3d36a400a3fa1b":{"kind":"REWORK","_number":21,"created":"2025-07-31 15:10:27.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/21"}}},"commit":{"parents":[{"commit":"d74f72cf2d6cc881e75162d0b5a4d253bf3b6772","subject":"Dell PowerScale: Added support of thin provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d74f72cf2d6cc881e75162d0b5a4d253bf3b6772"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-07-31 15:09:50.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8d2d1b631b352c96fe9fc7455e3d36a400a3fa1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8d2d1b631b352c96fe9fc7455e3d36a400a3fa1b"}]},"branch":"refs/heads/master"},"f72131281519d21b162ad0247e8bc6973f8783cd":{"kind":"REWORK","_number":22,"created":"2025-08-01 09:50:29.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/22"}}},"commit":{"parents":[{"commit":"d74f72cf2d6cc881e75162d0b5a4d253bf3b6772","subject":"Dell PowerScale: Added support of thin provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d74f72cf2d6cc881e75162d0b5a4d253bf3b6772"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-01 09:49:57.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f72131281519d21b162ad0247e8bc6973f8783cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f72131281519d21b162ad0247e8bc6973f8783cd"}]},"branch":"refs/heads/master"},"e25c33da5c2a1b372e48dc2d4e3bdb12aff1af0f":{"kind":"REWORK","_number":23,"created":"2025-08-05 05:43:18.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/23"}}},"commit":{"parents":[{"commit":"e9386b67d9c99ec4b0d2e02440ae07b4ec1d0b74","subject":"Dell PowerScale: Added support of thin provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e9386b67d9c99ec4b0d2e02440ae07b4ec1d0b74"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-05 05:42:51.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e25c33da5c2a1b372e48dc2d4e3bdb12aff1af0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e25c33da5c2a1b372e48dc2d4e3bdb12aff1af0f"}]},"branch":"refs/heads/master"},"bad7036dcd7bc9c64d4268ee216eefdafa6cb361":{"kind":"REWORK","_number":24,"created":"2025-08-08 05:08:00.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/24"}}},"commit":{"parents":[{"commit":"e9386b67d9c99ec4b0d2e02440ae07b4ec1d0b74","subject":"Dell PowerScale: Added support of thin provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e9386b67d9c99ec4b0d2e02440ae07b4ec1d0b74"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-08 05:07:13.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bad7036dcd7bc9c64d4268ee216eefdafa6cb361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bad7036dcd7bc9c64d4268ee216eefdafa6cb361"}]},"branch":"refs/heads/master"},"677bf27e0cd4c9c175c3b2ef78da510f8a264dbe":{"kind":"REWORK","_number":25,"created":"2025-08-08 05:24:01.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/36/952736/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/952736/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/952736/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/952736/25"}}},"commit":{"parents":[{"commit":"e9386b67d9c99ec4b0d2e02440ae07b4ec1d0b74","subject":"Dell PowerScale: Added support of thin provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e9386b67d9c99ec4b0d2e02440ae07b4ec1d0b74"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-06-17 07:31:25.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-08 05:23:03.000000000","tz":0},"subject":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver","message":"Dell PowerScale: Rename Isilon to PowerScale in Manila Driver\n\nBecause the Dell EMC Isilon driver has now been renamed to PowerScale,\nwe should change all the relevant parts of the code.\n\nImplements: blueprint powerscale-rename-manila\nChange-Id: Ic377d336972c55ce3cd59549c6b1c034e062bc52\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/677bf27e0cd4c9c175c3b2ef78da510f8a264dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/677bf27e0cd4c9c175c3b2ef78da510f8a264dbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
