)]}'
{"id":"openstack%2Fmanila~958813","triplet_id":"openstack%2Fmanila~master~I64a006d5de2fc1194fda90282bd29b5bc533db61","project":"openstack/manila","branch":"master","topic":"bp/share-backup-out-of-place-restore","attention_set":{},"removed_from_attention_set":{"36761":{"account":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"last_update":"2025-08-29 09:53:47.000000000","reason":"\u003cGERRIT_ACCOUNT_36761\u003e replied on the change","reason_account":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"}}},"hashtags":[],"change_id":"I64a006d5de2fc1194fda90282bd29b5bc533db61","subject":"Fix versioning for out of place restore","status":"MERGED","created":"2025-08-28 20:50:05.000000000","updated":"2025-08-29 13:51:11.000000000","submitted":"2025-08-29 13:47:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958813-bp/share-backup-out-of-place-restore","meta_rev_id":"08d4b3d1fd45b21407d27cc7311a1ee09fd8a4a7","_number":958813,"virtual_id_number":958813,"owner":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 13:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-08-28 20:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-08-28 21:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-08-28 21:24:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 20:52:01.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-08-28 20:52:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-08-28 21:24:56.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-08-28 21:25:13.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-08-28 23:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97cee56453195b321d6b5bf7585d94136122450c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-28 20:50:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cc70e7b436b5086b4db98ef3953538230bd48e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-28 20:50:39.000000000","message":"Topic out-of-place-restores-version-patch removed","accounts_in_message":[],"_revision_number":1},{"id":"68430e89b39548edfdb3008f71e0cc48800bfc6b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-28 20:50:50.000000000","message":"Topic set to bp/share-backup-out-of-place-restore","accounts_in_message":[],"_revision_number":1},{"id":"e744022707765928f3a92c8fb2d5d58a0af921cb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-08-28 20:52:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af24259077ed7015dd130835ecbc0c138d2848d6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 20:52:04.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"fde81f237ad63ae9fd60cd6ed8e930a3028a68f0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-28 21:24:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9641172a475c01b05d54675b171596e15445e39","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-28 21:25:13.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"986acb46e3bc130cb6f9f7b5e57a495dd6ecfeef","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-28 21:25:43.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/dafc5340946e4b5185afac12b8607f15\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/5cd1fb9957994aef9b5b8fe0f5836fc6 : FAILURE in 28s","accounts_in_message":[],"_revision_number":1},{"id":"cc342c396331816641c817ae7c7562f2f35316e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 23:26:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc575d68e2e04df9a639477ee022835c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e0719db8434458a9ea4c0fd941bf0a9 : SUCCESS in 10m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f9a8fc8419746598e596de59c223aaa : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9cbac5715874cc3b37ada448aa777bf : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe5e4ed54c5748dcb87ccf005bc74c28 : SUCCESS in 9m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a5f55866fc94ea8859799149bbf8538 : SUCCESS in 12m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7aac0ce571a146348bda9775701cb910 : SUCCESS in 16m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e87ef7d4c449407d9d427ff8169308e1 : SUCCESS in 7m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4acd85470c564af68ce8a67408408a78 : FAILURE in 18m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a00123499b6f47ac818de10260cba35b : SUCCESS in 2h 30m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/150be96d725246738e0963a4748cfadc : FAILURE in 19m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bfa93aaf14ba41329995399dbd0ba9c4 : SUCCESS in 1h 12m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/738cf39fa4e345ef9ef32602d61a7478 : SUCCESS in 1h 16m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e97142e32fc1479cbcff82a0dc5b4a43 : FAILURE in 22m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9e7f1c61de23468196cd956c176e2770 : SUCCESS in 5m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b8244555b0fc4cc2bcd6f7516ef117af : SUCCESS in 6m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a80ec2db60074378b003381739a0ed02 : SUCCESS in 35m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/18ba102551f746e483b5d96aca954a0d : SUCCESS in 28m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4684f82c78c64b0b83a7bbf2312502a3 : FAILURE in 24m 22s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/382f3bfe111641888e600eed715ff941 : SUCCESS in 1h 17m 53s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7c67fe1663d3430494c99bd6564e9017 : SUCCESS in 47m 45s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a93f9363e1424f18b35253f2d2fc3155 : FAILURE in 22m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/03763977f9924373ae9b994116f4becd : FAILURE in 5m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/70b3b7a7a34f42f9b42315c41e5860e0 : FAILURE in 12m 21s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/235a8bab72414610bfe2a95607966540 : FAILURE in 5m 06s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/606a937ef37447cc950eaf91230d1f2c : SUCCESS in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ecb319315acde8680f4768e97dfd32c3527dd7d","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-08-29 09:53:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3be60574d3e24945f6918809e9b4e8e8256204a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:34:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ba285263e94ca5b2f85ac803eb087f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77365babda794387bac355748e596d02 : SUCCESS in 12m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0ee82a9d62c48f2935a50b8cc052ab1 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab64a6098bb404eb82155c0ecaefe1e : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a976b9ae19b148d0a6ae229fd03876e4 : SUCCESS in 13m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5f76b668fb442e0b5df6a99c9e93aeb : SUCCESS in 16m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b0ede64bb6b44a097bc3f158f36fbbd : SUCCESS in 7m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60c0c6825a4642d69c40631a5cf278f9 : SUCCESS in 4m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0b2291bb3af0485f96937ad799120767 : SUCCESS in 36m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c60e28e5965e4c09a534742b123a8e47 : SUCCESS in 2h 05m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/69a1fdc59c434953ac7a782ae7a56327 : FAILURE in 28m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/03f84c0b84c6440c833b5cc5519eb06c : SUCCESS in 1h 04m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6b81ece294824344a474d61bc1939329 : SUCCESS in 1h 30m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e768d0d9f05b41c7934337afbb9104b3 : FAILURE in 16m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9db25825d2454fd3b21175bc69533e4e : SUCCESS in 5m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a71e8e8b3f04d2ab0b24099045b4f6e : SUCCESS in 5m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/47f24500a8f44d43b68c82551b46bf4f : SUCCESS in 28m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/efec3d37b29c4b35aae352db83e0b947 : SUCCESS in 36m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8d5930a9b89440f49380a0e904cb5454 : FAILURE in 26m 58s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/b55cce9dfc8e42189b840d80ab050900 : SUCCESS in 1h 15m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/017a84e71fdc4445bedf325442216c54 : SUCCESS in 36m 28s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/525efedd7f0349e3b422a25b9f48bc51 : FAILURE in 14m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5092e79a8ef14a78a1e3a54aa451d9d3 : FAILURE in 5m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3c09b5bc5a964683b0628d90658309bb : FAILURE in 6m 23s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/eff9ef01010f4610a6cbd3b8c4a772b7 : FAILURE in 2m 53s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1afca90e3de84106a13af360a027c96f : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2bcc6c20be2f27b0ebed7e141dae5c5f4ac41c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:35:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fb176c2cfa02ee0398fdfabee3d9596e8cb8be0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 13:47:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa89572390a748b289bed413214c8427\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ee827b2ec349cca35d70f8443212eb : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338056393abe4e40877ebafee81d925a : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1734ae3784754c96a0a84790c3d88148 : SUCCESS in 8m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f6dbe0c23af45509c2ea50c2f6e99db : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/632d64f66445477d8b6b3ad894fdf47a : SUCCESS in 7m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7968b5f4fe6a4e0bbca69c0fe437c856 : SUCCESS in 1h 08m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bbf80c0f6bb14593b8a3da81b3a094ac : SUCCESS in 35m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/247ed4fdbf3e434f95c945e4df4b2ea6 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":1},{"id":"14a3dfedb11fed34205c887b0f6dc1ccd87bfe63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 13:47:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08d4b3d1fd45b21407d27cc7311a1ee09fd8a4a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 13:51:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9bc223cecae4d13b288dea0c94341fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0653d528a0764e6cbed818e164cc920d : SUCCESS in 2m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35de308709db42d3997d779bb624dd24 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dec9eae1581f2f66a27f4926156b43589ad9f981","revisions":{"dec9eae1581f2f66a27f4926156b43589ad9f981":{"kind":"REWORK","_number":1,"created":"2025-08-28 20:50:05.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/13/958813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/958813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/958813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/958813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/958813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/958813/1"}}},"commit":{"parents":[{"commit":"fb3c376bf6954a333c2690d59c1d046279805fe4","subject":"Merge \"Add support for out of place share backup restores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fb3c376bf6954a333c2690d59c1d046279805fe4"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-28 20:47:38.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-08-28 20:49:57.000000000","tz":0},"subject":"Fix versioning for out of place restore","message":"Fix versioning for out of place restore\n\nCorrect release notes to contain the appropriate microversion (2.91)\nand minimum microversion support for non targeted restore (2.90)\nrestores.\n\nPartially-Implements: blueprint out-of-place-restore\nChange-Id: I64a006d5de2fc1194fda90282bd29b5bc533db61\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dec9eae1581f2f66a27f4926156b43589ad9f981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dec9eae1581f2f66a27f4926156b43589ad9f981"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
