)]}'
{"id":"openstack%2Fmanila~958828","triplet_id":"openstack%2Fmanila~stable%2F2025.1~I171e63b147c37f68f3bad329f4ba4abc290817f1","project":"openstack/manila","branch":"stable/2025.1","topic":"bug/2120291","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-09-11 21:08:59.000000000","reason":"Change was submitted"},"38290":{"account":{"_account_id":38290,"deleted":true},"last_update":"2025-09-11 21:08:59.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-09-11 21:08:59.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-09-03 22:50:45.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-09-11 21:08:59.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-09-11 21:08:59.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2025-09-11 21:08:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I171e63b147c37f68f3bad329f4ba4abc290817f1","subject":"[NetApp-REST] Fixing basic share creation failure","status":"MERGED","created":"2025-08-29 06:35:40.000000000","updated":"2025-10-16 21:19:10.000000000","submitted":"2025-09-11 21:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958828-bug/2120291","cherry_pick_of_change":957045,"cherry_pick_of_patch_set":5,"meta_rev_id":"31d7be2ab382d2d64481ecfd41d4474e8aee6953","_number":958828,"virtual_id_number":958828,"owner":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-11 21:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-09-03 22:50:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-09-03 22:50:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":38290,"deleted":true},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"reviewers":{"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38290,"deleted":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 06:41:07.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38290,"deleted":true},"state":"REVIEWER"},{"updated":"2025-08-29 06:41:07.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2025-08-29 06:41:07.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-08-29 06:41:07.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-08-29 06:41:07.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-08-29 06:41:07.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-08-29 08:12:10.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-08-29 09:29:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2f1f307e4c78ace00b88f0c99681e0d1bdd2570","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-08-29 06:35:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6d6286589cb737a16ae9b21cf753379b02f695cf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-29 08:12:10.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/20b209434f2c4f56a89e3438071665c2\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8482b321e92b4b5194c8b36c567ccec9 : FAILURE in 30m 48s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/235d5f57fc8f4cb99ef72ebe71f98850 : FAILURE in 34m 37s","accounts_in_message":[],"_revision_number":1},{"id":"921ea5ba91093dfa38da3ee449b4523a9649d434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:29:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffe07f771e024ce89d9aa34040f1d348\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275c41610f024bafb4a2ebc9bd87cbaf : SUCCESS in 12m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbc4f1ddd71f49bb8cfcbf90e4bddafc : SUCCESS in 32m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b458ed028474463a3849447c4f89a31 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4395e915bcb9448cbf3e8257306c88cb : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be3baec581224bb3b68b241c1becd5fe : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49fc3ea06e7c4c8db612f85d303e67cf : SUCCESS in 9m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a1a87bfc2fb94f3b99c3ae301a57797f : SUCCESS in 55m 59s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/5c1f6cdab7724cbfb34d87af79e33312 : SUCCESS in 1h 11m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ff1b73cb34cb4079b121b83183a9821f : FAILURE in 2h 44m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1b7290232ca641e3b1d2d02f656ead4d : FAILURE in 1h 13m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9f7225762ea742b18785db66d2aa4fc8 : SUCCESS in 26m 38s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3981d46017de4f9ba17789eabe29667d : FAILURE in 1h 38m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/95d73daf16664d70811ad14136afc46c : SUCCESS in 1h 19m 25s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/30efd0b59cca42a582c1701201e42a98 : SUCCESS in 3m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9363199c703044b7bcb4193e7937a2f0 : SUCCESS in 6m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/82edff92900748b1bf94ca3acfd009f9 : SUCCESS in 41m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/86223317f04d4e6ba6ea55d0ab00f524 : SUCCESS in 29m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4970108a3f764dfa8d6ef0470510d16f : FAILURE in 51m 22s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f1cb31ce0a084235a62b2d561026f999 : SUCCESS in 1h 05m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ffe3e49af21a41ea9c0b644f0d8d35fb : FAILURE in 27m 08s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6df9dec06edb418ca9537cbaf6169311 : SUCCESS in 57m 43s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a59c4c81e8ae45ddb4bb45d09786e254 : FAILURE in 11m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5d01e577349144ae87072b2bd0d81206 : FAILURE in 5m 08s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/431ac15b710246d086926f5e471bed04 : FAILURE in 4m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/212b3921f11d4520ae9bcbff25a63126 : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfac814ee83b5ee717e9f9654c9d7c2034528010","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-08-29 09:30:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"373329b6d02892863f8c8b01f356221e7474e5af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-29 10:23:18.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1e885dbf2e25473dad9d266b04cb1510\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a858b2f77c7048a38dfbca832290cff4 : FAILURE in 35m 20s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c45824aefe64a00b4dd7b54fd61853f : FAILURE in 36m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f702b465caae64247b46b2686e91000c8857e57a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 13:24:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65b8f3a038d48378ac0261625bf4055\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c906a850c24e0f80af43149547d1ca : SUCCESS in 12m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a029284226d4232927aaa7d728704e4 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccb693d5dad64d3fa342ced2a04e18d2 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d233b4a066e4439bba697355a40098ad : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3b1c64f0d684717b68af4f217f7a9e6 : SUCCESS in 26m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5897af68597a444dbff609115cd3eb4b : SUCCESS in 7m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dbd8cd09fc5f4846b0854fb460872f04 : SUCCESS in 1h 05m 42s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/835e85b5bcb24aef9af78cdf443875dd : SUCCESS in 30m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b8cba7cf034d4f7bb84d8e77b23a5720 : FAILURE in 2h 36m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/051cab4d58904e2984a2f4ff4745543b : FAILURE in 40m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7caa0c1bcbb8498cbeded74cd588b618 : SUCCESS in 58m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b86c1fb481294c228644b2980bda9dbd : FAILURE in 1h 07m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/64070aa2fb384f46a6520380cd76228a : SUCCESS in 1h 08m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3e5962f8c98441cda80d85c010d16a7c : SUCCESS in 5m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/07a4df3344334f1c914122975e7a2bf7 : SUCCESS in 3m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/15c3ca0697ae4448b7e61aa6ac4d1fda : SUCCESS in 29m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/215fd966c9644a069017ef2e30d85a8d : SUCCESS in 37m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/79342ae4c7f64b4b845ecc28771d06cd : FAILURE in 50m 39s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/41c8c626286c4d3ca69815cf9eaa9c86 : SUCCESS in 49m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8beb88032556457a9484e377678f2151 : SUCCESS in 56m 54s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6bacdb933bb43d2a8e8ae2d3c0579d8 : SUCCESS in 47m 09s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/766652e76e604f90aa23a58f24174fcc : FAILURE in 5m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/658bcad8f98045658dc6d2b70233a4c8 : FAILURE in 11m 18s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/2e6f2f538a7249eb81ab211be969239a : FAILURE in 6m 03s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/00c523d7444147799d59afb3b4e981ef : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30ed838c5f2078e8fc5fb61d52e846084161f89a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-03 22:50:15.000000000","message":"Topic bug/2120291-stable/2025.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"ed2db5b294c14395d5eba13f536292c6cc4f3f8e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-03 22:50:18.000000000","message":"Topic set to bug/2120291","accounts_in_message":[],"_revision_number":1},{"id":"97379ecf1d8135ca795445eba3e622d2c6498f24","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-03 22:50:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05872658cdb854acf9f23d45cbb4684d79572fa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:51:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5731563df5b80c146e887fa7945a50a8c948d1b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 23:59:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccf3b154499047ad9963fcb7866e134b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b55fbf5b1858470fb8cb5c8924eb4a16 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56694014a0e24ab1b73fa0b22b0800d7 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79ca12a198c24fd9940e2ea805d9a761 : SUCCESS in 8m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10868b394bd448a5a80e6846a8cdfa41 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21da935896d34b13a95c758b375b1ac7 : SUCCESS in 9m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/709f478beeb04d65abe5bb94e27b3316 : SUCCESS in 30m 56s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/e95468f1772f459fab8b402a0d8e600a : SUCCESS in 59m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/28a3ffd13d6044c0a82518a1e435592d : SUCCESS in 34m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a876c45b55694b4c895e9d364ac0380d : FAILURE in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bc77c5a53065e0f005d0829c24a0daa27192a36e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-05 20:11:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4554e711d5f24bc4adf193e26c0b19d25c271fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 21:22:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6aa5fe5fdf041ce9d32313695970b68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507c6670c26744e6b860de6c7bdccad0 : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba1db13b01614292abb896d9f9dd38fc : SUCCESS in 23m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d19752b064447d293169f9df3db7a55 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1151eb8805ff416ba1ad53bc8691fe96 : SUCCESS in 11m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0d310419cac4717be2433ce2f97ce43 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2ed970248cc4c51a209caa7c1ee451e : SUCCESS in 4m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/75d3da6aed9f41d8855bc444108243ad : SUCCESS in 55m 20s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/86f57bef62d246658db494f48b90df48 : SUCCESS in 33m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9ddc07fe469d413ca34a5819c02bc124 : FAILURE in 16m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4d8f3448cd9496985dd36b906688c0e : FAILURE in 25m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/445f9174522f45bc8779022d616225f8 : SUCCESS in 31m 38s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/eda324607fc24e0284f574de07d5de85 : FAILURE in 1h 10m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/03ec250a68b941129573ba368f437c7c : SUCCESS in 33m 33s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/37837b0eff0e4773a6857f81b4d105f7 : SUCCESS in 2m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/efcf083d72e649a397700d599cadab10 : SUCCESS in 5m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/55d622da5a184791a73d928ca6cc7ce7 : SUCCESS in 26m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e9c49909a28549418cffb277c4985cf6 : FAILURE in 7m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a35b486d891342fb86185e22b42b964d : FAILURE in 28m 14s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4d7d12385fdd41ff88b1c5e2e1854d0c : SUCCESS in 56m 37s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1b1a71e87b76486ead97db29655df56b : SUCCESS in 50m 09s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84c1136cde8349d3907874cdcc3529b0 : SUCCESS in 43m 18s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0828e9a855f847df9b060ea9682aef55 : FAILURE in 4m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7471ba8fc7e34eeeb55e35abc76676a2 : FAILURE in 10m 37s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/6278bfbc8bcb40f9a8cd53bdf616e546 : FAILURE in 4m 50s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ae239b91403b4adbb8ada5c18b548cf7 : SUCCESS in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"812bff3b3ac1a9b8f9afb010b0f0a80277e0cd94","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-11 18:02:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af99762d0c9c3837b0b02e0bb9a3242756deb40b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 20:10:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/530e263e952d4a9f8ccce801b286367d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7691d09adf0e4923822ef98ca2a42a46 : SUCCESS in 11m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/882754b22be842ee848cc925b70c9ced : FAILURE in 22m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725dbd7bef48477fad2b5ddb786a661c : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9de1b456123443b288960828d7f19d2a : SUCCESS in 10m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06644df4fc5642e7a5cc856467555917 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48d73fec71b441b8b504f96180fef142 : SUCCESS in 7m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1e4b683fc16743c99a612adc0e562af4 : SUCCESS in 1h 25m 52s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/eaf1046c9de54f2c992a2c6e44453973 : SUCCESS in 57m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/203d6a9275a64c53af50714a6024bd44 : SUCCESS in 2h 05m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ebc843b8da6f46a98fa22b16328def0b : FAILURE in 1h 00m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9a963d02b2924958be2a55b45ffb011f : SUCCESS in 50m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cf24237ad55142e08811cfe56692efb8 : FAILURE in 1h 13m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e5cbe8d2cc9541ad93d9683d65d73c22 : SUCCESS in 1h 06m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4fc42bb2bdb84ada8f4a98c2e4cdbfee : SUCCESS in 5m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f23db37b830345ed8e5e09626cc270ee : SUCCESS in 4m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c461d7164029412fa4b7710613f737d3 : SUCCESS in 31m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec741e3d76664885928b0fe7576ffe67 : SUCCESS in 37m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8c1384a8dbda4cc288379839a2422310 : FAILURE in 53m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5ec3f356bf544be2ab777bc105ff32f6 : SUCCESS in 1h 15m 09s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c4801e9d1aac4a45bf4c45004566ba02 : SUCCESS in 53m 09s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f270c7785e7644719e1128b24fb9ae2b : SUCCESS in 36m 06s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9c53837a52eb48a985425f085244ea77 : FAILURE in 12m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3ec1b5ab5cfd482fac4235fd5f6284ff : FAILURE in 8m 20s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/357db7db732b4811ba302654214714a2 : FAILURE in 6m 19s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/06c76a71cbe848d5aa3a149bc0780c43 : SUCCESS in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1599648d158f3da3fda1754a3945170351d9c7a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 20:11:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb0419b7a53f69c9e1bfc9124a0c6eff1917a699","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 21:08:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e71c0d9f8d4c4320ba4da1ea8478ef7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe297d72bba429790769cefd57eda44 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c834d492074793b7c1f4640337e5d3 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3f5016f859543679bc7c7f0ae973cb1 : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54bbb5e8eb6040759e661c130849f648 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be202a629c8c420f9c9b87d5f9233147 : SUCCESS in 8m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/26fd72918ce74620b112af0c7bf6222e : SUCCESS in 36m 27s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/03eab7dea5134bc7ab6ede203422cd66 : SUCCESS in 50m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/62e29e59994240209162b7fd1d50370d : SUCCESS in 37m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/15ace9b36494436193a5efc803670fd4 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e911049625b6ffc3cd6025cabbdf6d3a85d5961d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 21:08:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb7ee1c0ca8b192b6ff741429038d830ff880827","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 21:09:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841cd978090849c5a3e0a77cd575fe1a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a879bc296948bfb7f18e8187aed213 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c45c6b744fd44ccac19f8dac7d08e41 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"30bba0ff4cfd96a80a810b5784df065d2d8755b8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-15 04:54:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f21ec28a9cab7d9ce03a878a7c657d931c714f92","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-10-15 06:09:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31d7be2ab382d2d64481ecfd41d4474e8aee6953","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 21:19:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfc8ccef6605e4d79e9f7bd44ef0bb32897c5606","revisions":{"cfc8ccef6605e4d79e9f7bd44ef0bb32897c5606":{"kind":"REWORK","_number":1,"created":"2025-08-29 06:35:40.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/28/958828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/28/958828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/958828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/958828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/958828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/28/958828/1"}}},"commit":{"parents":[{"commit":"9994dde8a0fe3da3628a2a2b887d3e90a5c506d6","subject":"Fix server instance subnet cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9994dde8a0fe3da3628a2a2b887d3e90a5c506d6"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-08-11 18:49:00.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-08-29 06:35:40.000000000","tz":0},"subject":"[NetApp-REST] Fixing basic share creation failure","message":"[NetApp-REST] Fixing basic share creation failure\n\nBasic share creation is failing with few errors related to snaplock\ncomplaince, network interfaces get and network port get as mentioned\nin the bug. Fixing those to unblock the share creation workflow\nthrough REST.\n\nCloses-Bug: #2120291\n\nChange-Id: I171e63b147c37f68f3bad329f4ba4abc290817f1\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n(cherry picked from commit 4e694d4780cf2803122a3256b0e4ca1279acb6c5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cfc8ccef6605e4d79e9f7bd44ef0bb32897c5606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cfc8ccef6605e4d79e9f7bd44ef0bb32897c5606"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
