)]}'
{"id":"openstack%2Fmanila~959549","triplet_id":"openstack%2Fmanila~master~Ie7978bff35597d6ce93cd07fec58788a4e81235b","project":"openstack/manila","branch":"master","topic":"remove-v1","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-12 21:15:49.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-12 21:15:49.000000000","reason":"Change was submitted"}},"hashtags":["remove-manila-api-v1"],"change_id":"Ie7978bff35597d6ce93cd07fec58788a4e81235b","subject":"api: Move v1 controllers to v2 (1/2)","status":"MERGED","created":"2025-09-04 15:35:05.000000000","updated":"2026-02-12 21:16:40.000000000","submitted":"2026-02-12 21:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959549-remove-v1","meta_rev_id":"8aea8eaefd9527a7761e8f4820c660df1be8f41e","_number":959549,"virtual_id_number":959549,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 21:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-01-30 15:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-02-12 15:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-02-12 19:56:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 15:37:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-04 17:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-30 15:15:28.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-02-05 08:31:43.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-12 15:37:35.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"5bdf1b21dbf63e199a07fa7beb1f138c4faf7333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-04 15:35:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38d8be928c4a0915dea4aad5e79fc3e85f79e28b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-04 15:37:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"44161fb12a6c440ce2eaf8fcb726f815ffd924bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 17:30:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1eabf95f7aa4142a572f1eed59979c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9378fc36cef4492382d149fdfe77252a : SUCCESS in 9m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94a9efdaefa648239acf4ab7834657e2 : FAILURE in 26m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b71904b676499c88af949ef16d08a6 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4964dca643c94627821c0b0ab8947d9b : SUCCESS in 6m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3cf25a4e36c425f8ee32931b264d6d8 : SUCCESS in 10m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f3c43a99192494895dcee5bea6dd4e8 : SUCCESS in 16m 33s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c29f954f6e3a4b82a3c0fdb5008ffcb3 : SUCCESS in 1h 05m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c707b5cdfede422e850dcf722b161e72 : SUCCESS in 1h 51m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d8e49459693b4bd19ce8c5f56d6e8710 : FAILURE in 1h 04m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/817f0160e8364fad9883305cfb8bc512 : SUCCESS in 59m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2f2d1dde9138411a8420f5c81ddff1f6 : SUCCESS in 1h 19m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f9e2d3dfc105492eb3ce91731641b903 : FAILURE in 22m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa2a8414d79e4a8f96f25aa1af2fa738 : SUCCESS in 5m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33cc0c58215c436eb68a4ded3f94b9d3 : SUCCESS in 4m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2989e2fde1f3402ea43506cfd27a0465 : SUCCESS in 25m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a43ad0a831e24c99882edbdaef428e35 : SUCCESS in 23m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0de990ae79984c06bd3b3962e1babcc4 : FAILURE in 12m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/d41f563a87a045b4be6c519e177b7b1c : SUCCESS in 57m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3f66a70cbbb6461bb145892f952d312d : SUCCESS in 46m 15s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ace37cca7ca445e87249780d812b8c6 : FAILURE in 21m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c7f76ed4b6a64fe9905202948bd120f1 : FAILURE in 6m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ff52db84f2254cb2a5d259aaecffeb8a : FAILURE in 6m 03s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/5fed3538660449ba97c366f87647b20b : FAILURE in 4m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3f5066b25e7e446596679c8113b2b1bb : SUCCESS in 8m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6332445be21b62ce2c467fe9b8ebb213203f002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 16:26:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"64e4b49901f391836d0237b6a502654c23d0930f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 19:31:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a763667d3ff74b1f8a11cc923269f911\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21433e008cab49ed9dc96ae98ba2ddf8 : SUCCESS in 13m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab6b7d63f1134938bbcf8660433428bd : FAILURE in 36m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a3c8bfa6f2949df8e237d1e1557f26a : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1f55e3456df4281b630152df6d063d4 : SUCCESS in 10m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4c176e04a3645e783463a508066427d : SUCCESS in 7m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5b69810462b544b199a3b1bc2ecaae39 : SUCCESS in 1h 15m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b9a3a562e23a4f989a05a3db9e63eda2 : TIMED_OUT in 3h 02m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/87935a6733474e40911549039d2c5e28 : FAILURE in 35m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a98706dbb3724ddf84dc0b7a1d63bbc7 : SUCCESS in 1h 07m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0c967f552ef5494c81ad74e2dca1975d : SUCCESS in 1h 33m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a09503babc3c4f45b029aa5617e0314b : FAILURE in 17m 06s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/55ffa91b8b644e45a999a600f1c22858 : SUCCESS in 6m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06835dfc9e2d44b090e495308e41a4ce : SUCCESS in 5m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e31cccf72c2a48e38bcc538521e7f68d : SUCCESS in 21m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/896521f1840f4bd7be62355b03098b15 : SUCCESS in 40m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1f96a286be7249f183542974b5ca95df : FAILURE in 19m 06s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/ecefe8dcbaa04e64ab98c75ba4c3244e : SUCCESS in 1h 38m 59s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d0e1a248655f42a99535fae6c6999397 : FAILURE in 43m 25s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79226bf3a51946bcb09de3fbf889c032 : FAILURE in 18m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e711289eb14c42319a1649327ece2ee2 : FAILURE in 6m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d76eff2c86ef44bcbc927dca2dd6ff3f : FAILURE in 14m 31s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/69522e03d4044648bb7c61cd5f033640 : FAILURE in 4m 16s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a7ba5ade38f64d81a4e3261e8bf77287 : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57b6f5546f7f65a77a7d67cf28c615f759087cfe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 12:10:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bf04d40878446d3cf4e2a21b7d87a16d4b59849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-20 19:22:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"033f38be2386d71ae9b0826beb1d01b1c51a996c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:58:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5f1ec73e61842c391cc18d3f342f46b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7da98972cc5433b831b6cf6d37fa2ee : SUCCESS in 11m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ab1f9ff2d264404b35d5077a41e122e : FAILURE in 33m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/901aea2e63894bab8fa23efe1646ccc0 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e45a9de8edb548c89d0ef275d010bcfc : SUCCESS in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb3bbf4f60434da79bb390073dbe6c43 : SUCCESS in 7m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4cc7a00d7d4941a6a1a9de13462c5275 : SUCCESS in 1h 16m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a07c78076a594ff8bb1381873f5739eb : SUCCESS in 56m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cb76d5e309d149e3b927d9de0199e66a : FAILURE in 53m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/215a52da716d48c3b72fe0ea5821336f : SUCCESS in 1h 00m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b7ed7df145f4402da19fb0b44919ace8 : SUCCESS in 1h 30m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/92a7318d12dc42689b6a8ed9aac99b3f : FAILURE in 1h 18m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b462d8387f9144e79420b3d4e7b8f575 : SUCCESS in 2m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0126da11aa2440149d0da9726dd22d18 : SUCCESS in 2m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8ef7790c52a44bddafa3d28148f2dc1c : SUCCESS in 34m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4c4480c9f6a3413688bd6845f22266e9 : SUCCESS in 36m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/12f268b816a4457993dc547724362d47 : SUCCESS in 39m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2bc2de54ff0e4cb5843fbae6a8c51fa8 : FAILURE in 32m 30s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/c1158f74fc7e448485b025f0c054f8b8 : SUCCESS in 1h 17m 37s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2e825fff537b48008fa205e60f5125dc : SUCCESS in 59m 40s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1bb5e2aaec94c09800cdf198331f1fc : SUCCESS in 42m 10s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/399cb01d319a400a90b6dfbc60f2e6e5 : SUCCESS in 34m 04s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/67ce308998604de68e39d3fd04d8a51e : SUCCESS in 16m 32s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/253a3526c0ed4eab93dada3391c72f00 : FAILURE in 4m 53s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3edf35a188eb40edad27072590fa6e30 : SUCCESS in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0426bb0c4106351f3a0b12d6b94cfb62b469e2d2","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:58:01.000000000","message":"Hashtag added: remove-manila-api-v1","accounts_in_message":[],"_revision_number":3},{"id":"35c9f67cd069281dc492690b1455beeab684f7c5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-01-30 15:15:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d7ef6d773b23428e27c241dc2fd8b2ebdbc86f49","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-05 08:31:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/d7c8f27637b541a4913f63556a03eef8 : FAILURE in 1h 00m 28s","accounts_in_message":[],"_revision_number":3},{"id":"2e19e684b84015464703b877c0f7f00bbd8890fb","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-12 15:37:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5093f4f156a6d79fe5b27ddd1597bd6e5b3d2c89","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-12 19:56:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f955c6126536789d3b3e6b3ff2100ff4d583358c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 19:57:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b46d0a6729246c4d3b3751e6ccecb5259675680c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:15:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2aa3c403d8d40c78dc4441df9e0e631\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474c52f070d14effa1bb07a7bb21cc09 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58aa3799b9764cbba0082b70634139a8 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1305a79dc044323b14268b933b7ad14 : SUCCESS in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/030b0f69f517446bb0c829581c98b189 : SUCCESS in 6m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/42f978282867414abce0b0885179d42f : SUCCESS in 1h 14m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74a3c7d8dc98474ab3121e3fde4ff56b : SUCCESS in 24m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b7fb2c42f6904c52b6ff929f35c4e299 : SUCCESS in 33m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3e5c773c436c42b0bc3c8c082e17ee69 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":3},{"id":"e1174ef12329cff262d71937eb776aaebae78ad0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:15:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8aea8eaefd9527a7761e8f4820c660df1be8f41e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:16:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cd42c8295144042b7b5ac1df27435f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c545f8d52c204855843af335cb754ea8 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"811be781db28ffe7297c30832c665b1b7271929c","revisions":{"7183d7f0205890e281387837ee5d2cebea949852":{"kind":"REWORK","_number":1,"created":"2025-09-04 15:35:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/959549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/959549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/959549/1"}}},"commit":{"parents":[{"commit":"756d5f1200f03fbf80de8735ccdf13cc13628955","subject":"Fix duplicate registration of barbican options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/756d5f1200f03fbf80de8735ccdf13cc13628955"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 10:58:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 14:05:11.000000000","tz":60},"subject":"api: Move v1 controllers to v2 (1/2)","message":"api: Move v1 controllers to v2 (1/2)\n\nMove the controllers that are identical between v1 and v2 to the v2\nmodule. This was already done for quite a few modules, as seen by the v2\nimports in the v1 router module.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ie7978bff35597d6ce93cd07fec58788a4e81235b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7183d7f0205890e281387837ee5d2cebea949852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7183d7f0205890e281387837ee5d2cebea949852"}]},"branch":"refs/heads/master"},"a46e4d9106ba53410be7a26adf5fbad6e10b0c95":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-30 16:26:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/959549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/959549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/959549/2"}}},"commit":{"parents":[{"commit":"58e8f893e342c5f318ce0d7b21dd5fee27919123","subject":"Edit deferred_deletion doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58e8f893e342c5f318ce0d7b21dd5fee27919123"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 10:58:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-30 16:25:38.000000000","tz":0},"subject":"api: Move v1 controllers to v2 (1/2)","message":"api: Move v1 controllers to v2 (1/2)\n\nMove the controllers that are identical between v1 and v2 to the v2\nmodule. This was already done for quite a few modules, as seen by the v2\nimports in the v1 router module.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ie7978bff35597d6ce93cd07fec58788a4e81235b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a46e4d9106ba53410be7a26adf5fbad6e10b0c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a46e4d9106ba53410be7a26adf5fbad6e10b0c95"}]},"branch":"refs/heads/master"},"811be781db28ffe7297c30832c665b1b7271929c":{"kind":"REWORK","_number":3,"created":"2026-01-20 19:22:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/959549/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/959549/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/959549/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/959549/3"}}},"commit":{"parents":[{"commit":"7b2392a9a244f11a3e313cc3784c7d16a305ee61","subject":"Merge \"NetApp - Implemented volume autosize reset support for REST\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7b2392a9a244f11a3e313cc3784c7d16a305ee61"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 10:58:19.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-20 19:21:32.000000000","tz":-480},"subject":"api: Move v1 controllers to v2 (1/2)","message":"api: Move v1 controllers to v2 (1/2)\n\nMove the controllers that are identical between v1 and v2 to the v2\nmodule. This was already done for quite a few modules, as seen by the v2\nimports in the v1 router module.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ie7978bff35597d6ce93cd07fec58788a4e81235b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/811be781db28ffe7297c30832c665b1b7271929c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/811be781db28ffe7297c30832c665b1b7271929c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
