)]}'
{"id":"openstack%2Fmanila~959552","triplet_id":"openstack%2Fmanila~master~I8c981fae53afe8fac40b7f1222b1ebf577fec031","project":"openstack/manila","branch":"master","topic":"remove-v1","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-16 12:19:49.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-19 18:03:02.000000000","reason":"Change was submitted"}},"hashtags":["remove-manila-api-v1"],"change_id":"I8c981fae53afe8fac40b7f1222b1ebf577fec031","subject":"tests: Remove tests for v1 API","status":"MERGED","created":"2025-09-04 15:35:05.000000000","updated":"2026-02-19 18:04:00.000000000","submitted":"2026-02-19 18:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"959552-remove-v1","meta_rev_id":"5171a4d2cd042e5977b4cb6f72f636b6001526fe","_number":959552,"virtual_id_number":959552,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-19 18:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-02-19 14:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2026-02-19 06:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2026-02-19 17:55:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 15:42:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-04 16:15:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 16:23:58.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-12 22:20:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-02-19 06:10:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-19 14:59:40.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"fe6ad12daf77152231d179125626276e35005054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-04 15:35:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e118fca548dba41802f3b1ad447eef502750138f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-04 15:42:09.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"19d205915e96a22869e61db4df1c71316e562ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 16:15:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2145c76525be40d98784e13deac693a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e209dfa9f742348ba8837f3f945d80 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e40facdd882e4ed698279b11fc81d525 : FAILURE in 35m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbee9a272cf846ef952c45e46de4e757 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7ce73893f1a4adfbdbe3ca92c680848 : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6a0116f31c54902b728e58d1bfb8fd6 : SUCCESS in 10m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c123268d4554438bdaef76d8a818c0d : SUCCESS in 12m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5573fb7edd0946b595f620235152e1fe : SUCCESS in 3m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c96a30617a2432781042ce5b7b55ca4 : SUCCESS in 5m 05s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/477efc20e41045cd97ab56756a06e7f0 : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41a35e01bb3c662f6d981a0cafbaacc0db94ef36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 16:26:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"420cc1365769ad2f0d79c9f85bf53ffa252b363a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 17:10:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0312a076bbbc495f8cfb52849cb5ca7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0891b649674528a388453a023dd8e1 : SUCCESS in 13m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816bc662f5b74c6ab49b425be7e3464b : FAILURE in 30m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e5e7300d364cdeb79df758c3bd86fd : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/803566169df14e3b9b4ce9e0598e6ae2 : SUCCESS in 10m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/723d5728908f47faa551dac1ddc067bb : SUCCESS in 13m 31s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0091937dfe834b26bf644b146cb87838 : SUCCESS in 4m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/572436dae861400f9bc6ff57701f2122 : SUCCESS in 4m 38s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7e1ee58221ac49538584ab28f9a144ce : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b4b1f5383e19916fd69fca10adc8ab3ca22e717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-21 21:43:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6482c50853197cdd0664e6698e6bda213b0f0cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 22:18:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123d41daa2f24ee48d012e91ec62b5b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9b3ddb8f714da89ddc106d830441a7 : SUCCESS in 12m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62ac566457f54dc8a1361fc5377eb71d : FAILURE in 30m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c830a297ba86403a91bf399812c047a9 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c465eca8b1b64ab197b3b26fd83f6902 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c9514c8d416480d8b64882302e65a0c : SUCCESS in 7m 03s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/28e3f3cbb4854767b8a7a348f9d31257 : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b007d6f3495e43f399ac8b2e8774a056 : SUCCESS in 4m 10s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5ae691d6005e477b92129f21d1e4c68f : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f30d1fe3df88ae63d299ab931e138c136269776","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-29 19:57:47.000000000","message":"Hashtag added: remove-manila-api-v1","accounts_in_message":[],"_revision_number":3},{"id":"bc3c29d19272ed8f62af9a5141b19aac658cb4d5","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-05 16:23:58.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/da8f7df8d2524a6cac107d85fdd9375c : FAILURE in 1h 16m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6639f35847d347d61cc650906e5462fe8b9376c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-12 21:23:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"055f403e9dd20e2ccd4b9bbf409f0be53e62566b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 22:04:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9ea07b1f894228a364bce84c456767\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce8bfdcef674930b0caeed269bbeb7a : SUCCESS in 12m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51cdcfd476404fe1b882d6cafe96d214 : FAILURE in 34m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed406d3d42b461dac17ff43e3d3ec0f : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05130d9975ad458b92b49d678a65655a : SUCCESS in 8m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e87e3db19a14490ea79a5e27eb37d8fd : SUCCESS in 6m 42s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/271252ed9ec54a3899d60a965d0c5a08 : SUCCESS in 3m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd13434f9d5441068eccbbd9e5384b4d : SUCCESS in 5m 27s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/30c58a0a3a334710aeddf47063b5b1f1 : SUCCESS in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f854efcd6e4f86b983d22e33f9d6ef2ed1ab8bac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-12 22:20:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/959552/4/upstream-check-manila/manila-tempest-plugin-ontap-dhss/6c0509b/ : RETRY_LIMIT in 8m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a0ee844ed6ee7a25b1d654db1670a45428a11003","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-12 22:45:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/382b01ee8bff499aa2fccf181802c2e4 : FAILURE in 36m 26s","accounts_in_message":[],"_revision_number":4},{"id":"e2403e6e1e68bc915ef9f6645a15050aa01b7052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-13 22:23:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"55627071f3ed926c0f4657b71c68f6f0d839a93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 22:36:45.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/028c13a5fc6a4b23981e8246507b1bf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d77bd35e64b4136adbdca1eb1500fa2 : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3136782c4d44494947c3002699e4d64 : FAILURE in 2m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4eecfabe504b3ebf601f21ec4977c1 : FAILURE in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4802c3b4f0ab4ec49a20d385814b3810 : FAILURE in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7fe817c720fb4e85bfd0828ab1d6d6d7 : FAILURE in 4m 22s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e71f1a1668104214be9c0de1faca8ab3 : SUCCESS in 6m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/418f0e11df364e5c803c8569350721b6 : SUCCESS in 5m 34s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/be6ef46ebf6a4e6a99f941574688af1f : SUCCESS in 6m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file manila/api/v2/shares.py","accounts_in_message":[],"_revision_number":5},{"id":"88916027e12667da875a43488c9270c16862fa7e","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-13 23:35:22.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/1eacf6968fd2493e8ec2607686bbd79f : FAILURE in 36m 58s","accounts_in_message":[],"_revision_number":5},{"id":"d314478f26f5c07a9f2aa55772e90c0d24ef6e63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-16 11:42:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a4a2d7af6a20a29a55b5a233d2fd258d4685ff21","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-16 12:19:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/44a2d5449cb84e9d94a5770e90b81794 : FAILURE in 35m 44s","accounts_in_message":[],"_revision_number":6},{"id":"d10f0a658d3bc8229d13ddb9703335bd0a072931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 12:21:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/712d28f1e76c427c94bd96e475e20ab4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521064e6dfd046568e0c9554bb60f3e6 : SUCCESS in 12m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/527ef2cc51344f2c90a81c1ca5a71190 : FAILURE in 31m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046f61139ba2496db4b53f2d1223ccdb : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33a341b8572946f5b7f004f4c5a45d49 : SUCCESS in 8m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db9d560a735d4d32b40c2cb0f0901de4 : SUCCESS in 7m 12s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/dc4856ec677447b9a86a4814ce7ab3c2 : SUCCESS in 4m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d1035e24eb0d4031a6c54b9bab21dbe1 : SUCCESS in 4m 12s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/36c06486eed5491487a474a7e2d4f947 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ea1b5379fb248b627a72f64bea84fa742de4ca40","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-16 13:36:46.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/959552/6/upstream-check-manila/manila-tempest-plugin-ontap-dhss/cfa0b4e/ : SUCCESS in 1h 14m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"0ad1a73c4bd2b34effc5720cc10ee75def8b860a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 06:10:03.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c6a06929a7d35b90d03e90bbba49f1e051e64c7c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-02-19 14:59:40.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"472115bcf1950b1fce70250b70ced34a3d8ec733","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 17:55:44.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"96541ec97d4ec5d4db6a4b039baa03b7b709ca28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 17:55:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f4ea1093fcc850bc3be96b7b9ff8bedb15d29ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:03:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e005603a5f064aa4849aa1ce2af6da18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d8d8f1d352447baa194b5f89415d4e : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f37d22dea443fba1e4d36fd1cdc0af : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad40473417d5419787f98c580ff63534 : SUCCESS in 6m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4780c04358d94b7b99fe3f0c3ca52349 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":6},{"id":"36fbb2825c86031211a8977ed51fff019e94d1d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:03:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"5171a4d2cd042e5977b4cb6f72f636b6001526fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:04:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9eb236582054aa5a26bf553c1a3e4f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f51a67d9534b42949d89be55dddf4b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e1fb16f880853cc0e87d24a0feea3f256b90e3ed","revisions":{"220ba067b30fbcb17bff9dcabda96b27411e2864":{"kind":"REWORK","_number":1,"created":"2025-09-04 15:35:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/959552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/959552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/959552/1"}}},"commit":{"parents":[{"commit":"42c515369ac947a9c27396c32574e46f927abbbf","subject":"api: Remove references to v1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/42c515369ac947a9c27396c32574e46f927abbbf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 11:24:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 14:31:38.000000000","tz":60},"subject":"tests: Remove tests for v1 API","message":"tests: Remove tests for v1 API\n\nThis us allows to later remove the controllers in a more piecemeal\nfashion. Many tests are squashed into their v2 counterpart where they\ndon\u0027t directly duplicate a test already there. The remaainder are simply\ndeleted.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I8c981fae53afe8fac40b7f1222b1ebf577fec031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/220ba067b30fbcb17bff9dcabda96b27411e2864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/220ba067b30fbcb17bff9dcabda96b27411e2864"}]},"branch":"refs/heads/master"},"3c1634328825ace351b47d2a93fdf892fed2b461":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-30 16:26:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/959552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/959552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/959552/2"}}},"commit":{"parents":[{"commit":"e976e41caa3c295a9d68ab36cb4066e43388b167","subject":"api: Remove references to v1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e976e41caa3c295a9d68ab36cb4066e43388b167"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 11:24:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-30 16:25:39.000000000","tz":0},"subject":"tests: Remove tests for v1 API","message":"tests: Remove tests for v1 API\n\nThis us allows to later remove the controllers in a more piecemeal\nfashion. Many tests are squashed into their v2 counterpart where they\ndon\u0027t directly duplicate a test already there. The remaainder are simply\ndeleted.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I8c981fae53afe8fac40b7f1222b1ebf577fec031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3c1634328825ace351b47d2a93fdf892fed2b461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3c1634328825ace351b47d2a93fdf892fed2b461"}]},"branch":"refs/heads/master"},"441ee74262516b281fe8e5915c5538bd444674d3":{"kind":"REWORK","_number":3,"created":"2026-01-21 21:43:06.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/52/959552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/959552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/959552/3"}}},"commit":{"parents":[{"commit":"c1624150a3cafca489498710767efc0c23a24f5c","subject":"api: Remove references to v1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c1624150a3cafca489498710767efc0c23a24f5c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 11:24:59.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-01-21 21:42:30.000000000","tz":-480},"subject":"tests: Remove tests for v1 API","message":"tests: Remove tests for v1 API\n\nThis us allows to later remove the controllers in a more piecemeal\nfashion. Many tests are squashed into their v2 counterpart where they\ndon\u0027t directly duplicate a test already there. The remaainder are simply\ndeleted.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I8c981fae53afe8fac40b7f1222b1ebf577fec031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/441ee74262516b281fe8e5915c5538bd444674d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/441ee74262516b281fe8e5915c5538bd444674d3"}]},"branch":"refs/heads/master"},"58da0ec9ca6013f780d8e0289b5ffeb2516d871f":{"kind":"NO_CHANGE","_number":4,"created":"2026-02-12 21:23:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/52/959552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/959552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/959552/4"}}},"commit":{"parents":[{"commit":"576d9b11b2d0ee6f90a121f19dd079e799886d0b","subject":"api: Remove references to v1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/576d9b11b2d0ee6f90a121f19dd079e799886d0b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 11:24:59.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-12 21:23:39.000000000","tz":0},"subject":"tests: Remove tests for v1 API","message":"tests: Remove tests for v1 API\n\nThis us allows to later remove the controllers in a more piecemeal\nfashion. Many tests are squashed into their v2 counterpart where they\ndon\u0027t directly duplicate a test already there. The remaainder are simply\ndeleted.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I8c981fae53afe8fac40b7f1222b1ebf577fec031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58da0ec9ca6013f780d8e0289b5ffeb2516d871f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58da0ec9ca6013f780d8e0289b5ffeb2516d871f"}]},"branch":"refs/heads/master","description":"Rebase"},"b064f342619959d11be24e3ef4ce239cc5829a9f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2026-02-13 22:23:49.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/52/959552/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/959552/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/959552/5"}}},"commit":{"parents":[{"commit":"38885601e012b71ffd71b2cdc44cff9258f9a5bc","subject":"api: Remove use of mixin classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/38885601e012b71ffd71b2cdc44cff9258f9a5bc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 11:24:59.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-13 22:23:30.000000000","tz":-480},"subject":"tests: Remove tests for v1 API","message":"tests: Remove tests for v1 API\n\nThis us allows to later remove the controllers in a more piecemeal\nfashion. Many tests are squashed into their v2 counterpart where they\ndon\u0027t directly duplicate a test already there. The remaainder are simply\ndeleted.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I8c981fae53afe8fac40b7f1222b1ebf577fec031\n(cherry picked from commit 58da0ec9ca6013f780d8e0289b5ffeb2516d871f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b064f342619959d11be24e3ef4ce239cc5829a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b064f342619959d11be24e3ef4ce239cc5829a9f"}]},"branch":"refs/heads/master"},"e1fb16f880853cc0e87d24a0feea3f256b90e3ed":{"kind":"REWORK","_number":6,"created":"2026-02-16 11:42:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/959552/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/959552/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/959552/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/959552/6"}}},"commit":{"parents":[{"commit":"9e631a614aae0a1d77a65e63d28ab33dc0c02343","subject":"Merge \"api: Remove references to v1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9e631a614aae0a1d77a65e63d28ab33dc0c02343"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 11:24:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-16 11:40:40.000000000","tz":0},"subject":"tests: Remove tests for v1 API","message":"tests: Remove tests for v1 API\n\nThis us allows to later remove the controllers in a more piecemeal\nfashion. Many tests are squashed into their v2 counterpart where they\ndon\u0027t directly duplicate a test already there. The remaainder are simply\ndeleted.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I8c981fae53afe8fac40b7f1222b1ebf577fec031\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e1fb16f880853cc0e87d24a0feea3f256b90e3ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e1fb16f880853cc0e87d24a0feea3f256b90e3ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
