)]}'
{"id":"openstack%2Fmanila~960642","triplet_id":"openstack%2Fmanila~master~Ib89ac298b18fcb2109e6ec8349046f21d20c9d3f","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-09-11 17:57:45.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-11 19:43:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib89ac298b18fcb2109e6ec8349046f21d20c9d3f","subject":"Return X-OpenStack-Request-Id header in response","status":"MERGED","created":"2025-09-11 16:53:59.000000000","updated":"2025-10-20 20:05:15.000000000","submitted":"2025-09-11 19:43:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960642","meta_rev_id":"0e2672818039e11385ed3465b030b6da91c344f8","_number":960642,"virtual_id_number":960642,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-11 19:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-09-11 17:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-09-11 17:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-09-11 18:00:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 16:55:48.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-11 17:22:16.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-09-11 17:49:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2025-09-11 19:02:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"faa2acb7f4254706f95969abbd3216ac158f7c28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-11 16:53:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bda8d431ac404db721e30b2f229b673cd907cf14","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 16:55:48.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"9e1adb19771421fb9f7ec53092d3fd869032cf50","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-11 17:22:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f345f0a63c67dd611614616d7258c28e9b522ee","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-09-11 17:57:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"688d8fd7410f6a6c5af110086fe2afde7596ac42","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-11 18:00:26.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4e3dc3a6d4aeb7df3e9a71da2b3050060bb9610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:02:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e4a8da1f2d43b883a9a97a59bef41f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04f2b71f08d4961bb5410ed8af06b64 : SUCCESS in 9m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25fb7d399d4842aeb6b4f978d9fcede5 : SUCCESS in 32m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e77af0cfa5d4d3489b409c4056dff46 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a263b073d8c434c896f56379ca5cef7 : SUCCESS in 9m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75f179adde7c472bbfa4dcdd85fb7da1 : SUCCESS in 11m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/117cd9512dc4402b91cf7dda6d7dd75e : SUCCESS in 11m 55s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/191508da12a44acf9b30f86f3811df1f : SUCCESS in 34m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/88fc444b91fc4d37be4cb3511e4422aa : FAILURE in 2h 05m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1dff8b5adff34878b9cc8b18bd955b6f : FAILURE in 1h 04m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3a59f7d2913849d88c65bfb7b1af156a : FAILURE in 27m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/28d30764dbc84ab39b8f334e854980d0 : SUCCESS in 1h 30m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/94ec653d6369427087816e0f09fb680d : FAILURE in 24m 38s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c1a152d0df124aa7884675af86f20783 : SUCCESS in 6m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8811c9f7c84e4339bc78a153d3e367f5 : SUCCESS in 3m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5c882b3339964ac9ab3c056798eb142c : SUCCESS in 31m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3b485367e6cc4ff5aca1e8a728efcc7b : SUCCESS in 26m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ea4999864d6246ecae85c7a1014df452 : FAILURE in 12m 22s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/d33bde690da141739bac0b119bdd060b : SUCCESS in 1h 08m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bcd9fe3f7c9f430c93c5df68a286099c : SUCCESS in 36m 08s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd8253d7155141709039387722572c1b : FAILURE in 10m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/da4240ed787846c4b65acbf92b559ccd : FAILURE in 12m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/938c3ff833a54cc18a81d1c785c5512f : FAILURE in 5m 15s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/732cab62f340405ba379ebc9a219492e : FAILURE in 2m 53s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/6f9bd0be0b9c4fc9a2a4ac07734d114e : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"417b214fc9e87130e9d2f3f48b842d651c1480bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:02:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8734a1feaa6b646a7e06e7d3a5eae263b94c4afd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:43:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eecf8c8e25ac42768c3158d4122d8667\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53cdbab18df4a51a20a26014baef2dd : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55502c21f870496ab6b43eb2fc66d533 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6028866df1a46658550e71405163712 : SUCCESS in 12m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52b6a920502e4cdd99130d8712afc0d0 : SUCCESS in 7m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e35cd6f4da9e414985bf30a26836a4fd : SUCCESS in 32m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/982edabff21d4c67b692142340c6cdd6 : SUCCESS in 36m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/26e6da630da44302b5a648a0df2bd4b2 : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7378dcfe61f7b1486e93065f7337056873c369da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:43:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6b471871c6eed3f0940b6857afe11a15aa5c795","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:44:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f846bdbd02a40d2820df80dca5f34ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47282fe9a356401db413bdef5eecf41a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"898449082f86e29a52b4921add0f1fb05cb999d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-12 01:18:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e2672818039e11385ed3465b030b6da91c344f8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-20 20:05:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9effba1fad65583794a05e19672ed7f9b1c3735a","revisions":{"9effba1fad65583794a05e19672ed7f9b1c3735a":{"kind":"REWORK","_number":1,"created":"2025-09-11 16:53:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/960642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/42/960642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/960642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/960642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/960642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/42/960642/1"}}},"commit":{"parents":[{"commit":"564a55bd08a27083e1582d617179b817b6e80084","subject":"Merge \"Dell Unity: Fix driver startup issue with LACP configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/564a55bd08a27083e1582d617179b817b6e80084"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-11 16:49:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-11 16:53:50.000000000","tz":540},"subject":"Return X-OpenStack-Request-Id header in response","message":"Return X-OpenStack-Request-Id header in response\n\nIt is very useful for debugging purpose, because it allows operators\nto identify logs which is related to a specific client request.\n\nChange-Id: Ib89ac298b18fcb2109e6ec8349046f21d20c9d3f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9effba1fad65583794a05e19672ed7f9b1c3735a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9effba1fad65583794a05e19672ed7f9b1c3735a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
