)]}'
{"id":"openstack%2Fmanila~961339","triplet_id":"openstack%2Fmanila~stable%2F2025.1~I085dbf2ad8eaaa7f1d43d5f0cb42fad5e18eca3d","project":"openstack/manila","branch":"stable/2025.1","topic":"bug/2020745","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-09-29 15:08:20.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-09-18 06:32:28.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2025-09-29 19:07:59.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-09-29 13:16:30.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2025-09-29 16:06:12.000000000","reason":"\u003cGERRIT_ACCOUNT_35759\u003e replied on the change","reason_account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}}},"hashtags":[],"change_id":"I085dbf2ad8eaaa7f1d43d5f0cb42fad5e18eca3d","subject":"Dell Unity: Fix driver startup issue with LACP configured","status":"MERGED","created":"2025-09-16 06:32:56.000000000","updated":"2025-09-29 19:08:57.000000000","submitted":"2025-09-29 19:07:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961339-bug/2020745","cherry_pick_of_change":954209,"cherry_pick_of_patch_set":1,"meta_rev_id":"23faf46bbe2901a31fd084cb49c92025c33bbfd0","_number":961339,"virtual_id_number":961339,"owner":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-29 19:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-09-29 15:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-09-18 06:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-09-29 13:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2025-09-29 15:08:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 06:36:16.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-09-16 07:17:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-16 08:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-17 06:42:04.000000000","updated_by":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"reviewer":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"state":"CC"},{"updated":"2025-09-29 08:31:06.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2025-09-29 08:31:06.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2025-09-29 08:31:06.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"5801d9ac5f00e60f998aefa672c3a6016787debe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-16 06:32:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7433e51ab708bf673da18579f20290a62b82e086","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-16 07:17:25.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9393a409710249d5a1bccba029f4f7f6\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80389474f3ea46e28e72f551644b78a0 : FAILURE in 36m 13s\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc6e1817ea1e46d9b632f69bb5d6be9d : FAILURE in 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e6ffb3290a5956257c017653cd7dd2786ca386e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 08:15:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf6009642e734a848de057c00faf961d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b500bb36fb4915b9d4eebff5833d94 : SUCCESS in 12m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bba3623aae844b46aa05e9b474f0c10b : SUCCESS in 28m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad093ae6f844d46a243fc2993ba3d6e : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cca17db0de5425e8caa4b88f089cf68 : SUCCESS in 8m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bd14e8416ed43ec994f322f397277bc : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ca899bc99b348088284a6929183739d : SUCCESS in 9m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5de0640b477c42fe8b605319522cab20 : SUCCESS in 1h 09m 16s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/4eaa307438b048c4afa56411c97f4b68 : SUCCESS in 55m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2a5f30b3a61841d886256f73c501f00f : FAILURE in 1h 41m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1c5e02fc604544549e564798a23eec21 : FAILURE in 59m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/269f2fb7d0fc471a8f5ca78088a908d1 : SUCCESS in 1h 01m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/23b850cade3b43f79648a5232cc491f6 : FAILURE in 1h 28m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/93ac62ae67a44989b8f0357478a118cf : SUCCESS in 1h 14m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ea59dbea33a049ab8594925b7e03953d : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f22919fa3254d10afe2a0c45c8da312 : SUCCESS in 4m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/010e0273f4d54394b4613f833dcf1faa : SUCCESS in 29m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c26e65d601304c23a1f7efd875199af6 : SUCCESS in 41m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c4f06038eff34d37bae5100a727e3d86 : FAILURE in 51m 51s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/171ff89d7b464c349ceb262a0f30efe7 : SUCCESS in 1h 00m 00s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d1278e574823495c8882e69fbd0f3e11 : SUCCESS in 49m 47s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bcd9c8755b2643c6b0a43f8e74da1675 : SUCCESS in 43m 23s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0cdb7564a81a4f7480a73d9c0734eb1a : FAILURE in 5m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9277ecbc3b0f47089ad7e964b9921a74 : FAILURE in 5m 07s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/07a349b7afcf4299b371eb613a99e663 : FAILURE in 4m 33s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ab61cff129e44d95b313097eb3230b49 : SUCCESS in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adcf010a704a8b458f1e73d47dedf9da3a5cdef6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-18 06:32:05.000000000","message":"Topic bug/2020745-stable/2025.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"51db3fd4ec2f7a0d42899eb11ad2e00460e8a95a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-18 06:32:09.000000000","message":"Topic set to bug/2020745","accounts_in_message":[],"_revision_number":1},{"id":"4ad3160f752af91b3756b6bfaf4c708570a24472","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-18 06:32:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a86797b942ffe853000e3850213f6e6e34efe8c1","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-18 09:22:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ca4a9883a766e275a4cc833ba7a835ec038e084","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-09-29 13:16:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bddde1002c24639b793dc13b2edd1f515141ccf1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-09-29 15:08:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"288feb45bfc603cbe5b0cce0033f9ddbf56c2c98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:08:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6c418ce5514df1f0fd4ba0cb5e835f1a1caacd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:51:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df206040678a45f0bc1460223e4f3a43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b934d8b2d2a143119c5d9591cfc06dc5 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4a6e2cdb7341bd8339b59cb2bd6cb6 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3881d7107a38469598e80b8995713ed3 : SUCCESS in 10m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6879fbf9c4d4da8a08911ae5d46ada3 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09409d6a36d24af1a4c43c9e3c73ddb4 : SUCCESS in 7m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/74e68eef38744ea3915e79ec3a56bb20 : SUCCESS in 41m 53s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/71924460b2b645f495174e8fd40d3ef9 : SUCCESS in 37m 31s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8acd12aa9f254069af4a21d817a4b599 : FAILURE in 6m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/32bb00dcbf314794a6857f36dffc9569 : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7f4fd0555316dbfe61444858d93c74b8147f3c9b","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-29 16:06:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c45011302fd80f16b76c3783c0a63ef642c9b8ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:46:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd8b1262a23947e1b7ca4dec6dc15d4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0b24cc51f74a82bb4d9c177d25a5fb : SUCCESS in 12m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4da6d5bd04b4139850bffc2737a8604 : FAILURE in 19m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45243b4f32cd48488d4366a43ea1a6d1 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ab11281f6cc47938c76efba6f22e6f6 : SUCCESS in 7m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2042f02867434cc1b82850ea9ef3fc58 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4045724d5616442391bf6cbd13a5f79b : SUCCESS in 8m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8b0414dcaecb42caa2359d80b4ea8275 : SUCCESS in 1h 19m 04s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/744f83bb385d493192c5db1bfc3b889b : SUCCESS in 1h 13m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1e656e1825b441018c66448f2b3cfc44 : FAILURE in 10m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/61a33e569c5340d4af87bf6392ab9183 : FAILURE in 1h 35m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5dec0472efb545708e7e0a24d902af76 : SUCCESS in 31m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1305a3ad651e48ecbac0ceb3f9e54d3e : FAILURE in 1h 12m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/21c6740d2d004acea3dc7d96cb3c850f : SUCCESS in 1h 00m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8eebe18345dc4923837050a82d6cb76a : SUCCESS in 4m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/18980c09904346369ab157bb4b2efde8 : SUCCESS in 2m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1304b0e5ccbd482a9eed89820f356502 : SUCCESS in 23m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/951321c279ba40dcb1389baab27b87e8 : SUCCESS in 25m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0c2616a2ea374ef09a28aa954791d2b3 : FAILURE in 52m 50s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/909d8f56b4f741b6b4bb80091e5fef18 : SUCCESS in 32m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0db5e7f0911b4b93b356d122d499d7ae : SUCCESS in 46m 16s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53226ba61bc249d6ae7f512a51a8bfe7 : SUCCESS in 54m 23s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3b822e1e2614467dbc07d57a6a3b1059 : FAILURE in 4m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f7483ca03c9b4984b2f8cdcea43d7b08 : FAILURE in 4m 57s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/496f18e66cc9481da4f7682fc82baeab : FAILURE in 4m 41s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e84848f754f5415ba11c885b8285bdaa : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdd23244b5ca412a79582cacc6b3290c294038a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:47:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eab86e9a0567cd6404a8807f641b0d080e522d89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 19:07:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b24017bfe2b94ea8817f8b0e5a88018a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63a53161b954f05ba34ff63211451b5 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c92091a29174c0386a023d2e3f42491 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fbb19d3c4a241759ef71b9602d5e37c : SUCCESS in 7m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2e1ad9e53b04687a1ed69d9cda457e5 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/004c8aa245b042c989818dfccd31e76c : SUCCESS in 5m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2b7716f6d7204ac1b27ba08f874caaea : SUCCESS in 41m 08s\n- manila-tempest-plugin-lvm-jammy https://zuul.opendev.org/t/openstack/build/d6dfc513572b4144a3e9a81d6725c27d : SUCCESS in 1h 20m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3732e4cf1bdf4b86b248793c0edceae3 : SUCCESS in 22m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c642c6b09f9d4901b70a9e4fcc87b1d4 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1722408aba46da3c34d9dcad1f8ea573550d87d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 19:07:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23faf46bbe2901a31fd084cb49c92025c33bbfd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 19:08:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/406330007199466a99b3c9c0c5cf0ac4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abc7056af364209b594d7f224a95d51 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/606afce3c7b642589118b396a6cfc8d6 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c7fb884b181a54afc70263c2e6c1e8819eb0537","revisions":{"7c7fb884b181a54afc70263c2e6c1e8819eb0537":{"kind":"REWORK","_number":1,"created":"2025-09-16 06:32:56.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/39/961339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/961339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/961339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/961339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/961339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/961339/1"}}},"commit":{"parents":[{"commit":"cfc8ccef6605e4d79e9f7bd44ef0bb32897c5606","subject":"[NetApp-REST] Fixing basic share creation failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cfc8ccef6605e4d79e9f7bd44ef0bb32897c5606"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-07-04 08:30:31.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-09-16 06:32:56.000000000","tz":0},"subject":"Dell Unity: Fix driver startup issue with LACP configured","message":"Dell Unity: Fix driver startup issue with LACP configured\n\nFor Unity 5.x, port property `parent_storage_processor` is\ndeprecated. Read port property `storage_processor` instead.\n\nCloses-Bug: #2020745\nChange-Id: I085dbf2ad8eaaa7f1d43d5f0cb42fad5e18eca3d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n(cherry picked from commit be441c98796685e69187735a261a024281bb63cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7c7fb884b181a54afc70263c2e6c1e8819eb0537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7c7fb884b181a54afc70263c2e6c1e8819eb0537"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
