)]}'
{"id":"openstack%2Fmanila~962182","triplet_id":"openstack%2Fmanila~master~I8d964c575ee0a4638c5f43cfa89745c783e036ae","project":"openstack/manila","branch":"master","topic":"bug/2125572","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-10-17 10:20:02.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-10-17 14:20:35.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2025-10-17 10:42:10.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"36761":{"account":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"last_update":"2025-10-17 14:20:35.000000000","reason":"Change was submitted"}},"hashtags":["bug/2126844"],"change_id":"I8d964c575ee0a4638c5f43cfa89745c783e036ae","subject":"Fixes for targeted restores in the share backup API","status":"MERGED","created":"2025-09-24 12:32:28.000000000","updated":"2025-10-17 15:33:45.000000000","submitted":"2025-10-17 14:20:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962182-bug/2125572","meta_rev_id":"fedb1fddcb8b7a265c97c7a403b815b6942acc6c","_number":962182,"virtual_id_number":962182,"owner":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 14:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-10-17 10:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-10-17 10:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2025-10-17 12:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 14:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 22:11:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-10-16 20:52:17.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-10-16 20:52:25.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"400047f0ae6ba6634cdddde8aaa8d91282e8356c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-09-24 12:32:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db07805f9ea39bbb947de82e0951100d9bd95996","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-09-24 12:32:43.000000000","message":"Topic share_backup_fix_continue_progress removed","accounts_in_message":[],"_revision_number":1},{"id":"32815ebe9b65a02ff8e79f537e97edfc777283c2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-09-24 12:33:45.000000000","message":"Topic set to bp/share-backup-out-of-place-restore","accounts_in_message":[],"_revision_number":1},{"id":"8056e72584101676315893d09f5f80bb44bffabc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 14:45:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71abb07822744c0f8e55dd880402760c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb7796da67d45f4bbe470449d0f6881 : SUCCESS in 12m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44b67a86de2d47a0b598b62c2826df88 : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f1b11f57c44054b99ba23f0d4cefd8 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba1226988ca5497489974bba01bb85ec : SUCCESS in 10m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6563eb215c848978926bb9594b3a4b9 : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcd6e4dc27d341308ca01425c3b93ac0 : SUCCESS in 9m 49s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f129dfb01b9444aa8024325466d71922 : SUCCESS in 1h 32m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/106e81bef82249418cb2c5a794353d96 : FAILURE in 2h 06m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3fe2e001999f4b99af2246902eeb60a9 : SUCCESS in 1h 23m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e8b15617fd8744eabf9b4aba8729fe96 : SUCCESS in 30m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/62e50bba4afd43cda4aeba92dec0b478 : SUCCESS in 49m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/37fd8f099e2f4c389f76a05ff87956e6 : FAILURE in 11m 14s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/303856b757ac429d9cf6cfd885e0de53 : SUCCESS in 4m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e761a188192c46219487c8190057daf6 : SUCCESS in 2m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f1065faa927741b6a44c7d960d727bec : SUCCESS in 28m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/55176dbf315540498f36b1aef5df0370 : SUCCESS in 24m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a6de91c678e9469cab4744b18833cc93 : FAILURE in 14m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/32f6eb7149244b9c8afe62765d5d1a1c : SUCCESS in 47m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0a9abd3d507e4fb4a5c6ffd8deba7bc0 : SUCCESS in 58m 08s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43f91f87abfc4904a2e25250136cda15 : FAILURE in 10m 43s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/75c0845bb9744688a7db9c13ce45f122 : FAILURE in 12m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9d0f5375340c4e99be887a91aa2b96d0 : FAILURE in 4m 10s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/993d45dd9ed24c4ea2f80e2479d6b67a : FAILURE in 4m 53s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/681509ce1f514eeebd4d4a37bd08eb51 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f38d496b76903a6734ecc6891ec5d920092b8e13","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-25 22:11:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f94d1b5c2a4d18226b931ced8ba533a2dc18e888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-09-26 14:45:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"45fad64ef31bfd5133ed2bb60148498bdffe0e04","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-09-26 14:47:32.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6d1eb0dcde0ccc4d70204ce0e91e72a455f10dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-09-26 15:01:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a64b45b210decc5a58de8774326af41735a3562c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 17:48:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cb79930699643209acec716c6c59e7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ddafab5cea4c42a7b74e75c38089a2 : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24c23683f7b54055bceb0f4e6cf9ee00 : FAILURE in 23m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0310a3fd04e4b27925976a62e462f07 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a337527caf4f4af7a2707d2fdbac21eb : SUCCESS in 12m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/194e8f6556d74f0595548ffa7dac1e9d : SUCCESS in 7m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30d23fe2caca45029444c6975ccf2cc5 : SUCCESS in 14m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e772cf21ef64563b474bd546df0b864 : SUCCESS in 7m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/83acab2584a643d1a6a97f3de01560f7 : SUCCESS in 44m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/436c3e3a482e4ee8b0b7bf9f60847168 : FAILURE in 2h 40m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c2a153c77af04d9296264d8fb083504b : FAILURE in 40m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/337adf2ab64b49fd9a800a72a5e9e260 : SUCCESS in 1h 12m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/269b844a4df54aaea80da7b6bd48ba27 : SUCCESS in 39m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6e55b49394f643f680c96adb55777f82 : FAILURE in 21m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/64757f0a13d64a8596eee46a117fbb21 : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa9db5005d7c4189b3d2ec08abe0f456 : SUCCESS in 3m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e3b5060346e54bbfa6dbab84359fc795 : SUCCESS in 35m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/50e4026cb533420680bf63131a348c2d : SUCCESS in 41m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7da9f3c712a94b7ab6c927c24059e5f4 : FAILURE in 17m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/95c2d445b3eb40129946b8a559dda447 : SUCCESS in 1h 05m 38s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/44670066534c4d40bf53caba3c2e64a4 : FAILURE in 7m 19s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c17cc2ae83b4df8977b32d0ac4b42fa : FAILURE in 23m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8898835804db4b47bde6415b370b55d1 : FAILURE in 9m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1910dff11e1747dca8ec5b8cb26b5720 : FAILURE in 11m 18s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f5de1bf5def348b5b4fc60a7b1884b0c : FAILURE in 5m 07s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d3658e19095d4f9c99b1a3f538e3d408 : SUCCESS in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b92f55de14effd723f217b453c5c15addd9e1b30","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-04 07:02:35.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65027113d4395ea4ba06fc4857b2bff40d5a994a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 08:24:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f31b7537ec1447ffb6704721d5fe11c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc1d18936914c4381278be16e290e8f : NODE_FAILURE in 0s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/376d33da0f954c23941361303fb1f4db : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aedbe21cf9954bd9a7ad9ddf4d67c39f : NODE_FAILURE in 0s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5377c4ff539f4059b71b16877be08947 : NODE_FAILURE in 0s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bfeebec7d37454c806f24baca86c802 : NODE_FAILURE in 0s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2010c976eb1d419dbe00fb674dcd3563 : NODE_FAILURE in 0s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74d531b71c545e5b7307e2ee804153e : NODE_FAILURE in 0s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8d33c871061646eab3b159917aaa25e2 : NODE_FAILURE in 0s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/621251d35d8b4f7c8705027818e3a814 : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0c6ce546ef544deebfb13f0e141b99c0 : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/52fafd76d6764182bf895ce99e4c9c71 : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/794e0397c04b4004bb48d54a9fce01df : SUCCESS in 1h 21m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8bb07d8a3f9f4f0f9eacc94e32da40c1 : FAILURE in 21m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2b4febbc5a144b979518888e2f696cac : NODE_FAILURE in 0s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2574ec5714aa451cb8a820c6438f3fa7 : NODE_FAILURE in 0s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1c17f750961b463d8eda8c4f2310f883 : NODE_FAILURE in 0s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd2d3a8dcd43416898b4c4f43058e027 : NODE_FAILURE in 0s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7d1d5591631f4dcf8dd8668f6591b0b1 : FAILURE in 13m 50s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/c9cec40d244547418373e2d095b0cf75 : NODE_FAILURE in 0s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6ed4c288e0304e92a35b41fa20616ed8 : NODE_FAILURE in 0s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa7037f1fc0a4f7887c792aa52bae04d : NODE_FAILURE in 0s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/aeb3f3c0b03640eb9b337ddbce6d5c22 : NODE_FAILURE in 0s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/995bbc4c074f46e0956dbeadd0c934dd : NODE_FAILURE in 0s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/57f61d3e120145faa29d1b0c9c3d85a1 : NODE_FAILURE in 0s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3d987270fdd44248b7d9157abfcd0807 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c27484d7c690e808953547db509a17fd0beba3c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-04 15:54:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"831cd0b6a47de20b1384c755e2e5a40a8bb25df0","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-04 15:58:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6d4482db7feec7b21da0e4970b60617998fd4445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 17:57:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e7be452363d4bc69c21d74b2e5742a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a790a7083a6f457fbf6fd450726936a3 : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b10d9b618a574feeb2d5b16a9ed9ac83 : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e31d22a2f1440ead94d3dc0ab3ab5b : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e405024067f8424a9e948bbeafb270d6 : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5ab6e8abd3f46c3947e7a49b60a4ea8 : SUCCESS in 10m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8627acd79c3545759f854e807927bab1 : SUCCESS in 9m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5ea6a9450d94c448868859a3706a9ae : SUCCESS in 8m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/da55f0c36e7648ac8011d151fb4069dc : SUCCESS in 1h 16m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fccab9f07864429780f185429b984b66 : SUCCESS in 2h 00m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4ad86081206408a85a0ff9d837898ac : FAILURE in 1h 19m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/118d0ab703b4472cb53df7f9b3ef45f5 : SUCCESS in 26m 32s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/62c08ec5ff134d17b2968827301adcbd : SUCCESS in 1h 00m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6355c4e923f04458b6e5d62a6dae526f : FAILURE in 10m 42s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ab21e91837db403bb0b70528a1c8d289 : SUCCESS in 4m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2032743eb1fc49fa82f9be64c8394b21 : SUCCESS in 4m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22dad1275f134e88a941d7dc16083260 : SUCCESS in 26m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4028f2c416f64b989305eb22dadd67b7 : SUCCESS in 43m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e81581d70b694f1e92a9e1a9d2482316 : FAILURE in 15m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/0e8b0dbb7e9247e58040cdc79d7d6daf : SUCCESS in 51m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f0255eb623494127b71dd37a4b63d388 : FAILURE in 7m 02s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdb7227fb79b40c985080e45980d6c23 : FAILURE in 15m 11s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1b5045dacc3a4194b6857727ae78622f : FAILURE in 11m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f4da791272294e66863dc98ea42e5349 : FAILURE in 7m 55s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/4b67bc06cb014962830e3e55952e6683 : FAILURE in 4m 51s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0d714baaa9b04ac5b977f6a8b52ebf63 : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"66ba2e69d0e776861a3f9773b3ed05245efc99e0","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-16 15:40:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef4bbbf3bbc824e989579785ac0e7c36fa64da94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 17:43:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/750fab37e0924138be3a6fbacdda99a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a96792c7e140959a82b7222ba76756 : SUCCESS in 12m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/440387a1a6264eadb6533065a18d4bbe : SUCCESS in 37m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a55a6bfc9434f24a282093c9e8ac395 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5216db9372a45b6b65c301639caa41b : SUCCESS in 10m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a107e45006ed41c1a7916ec39698e977 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e552a50bb5c47e99f526b61a6bc8cc9 : SUCCESS in 8m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/29adfd67d3474414af846a8cb5d1c02b : SUCCESS in 51m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/524feb9554414d98a39dfa9e239f404f : FAILURE in 2h 00m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/887802b097224a318ffca19a94f68f42 : FAILURE in 35m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f54061d3c6bc44268dbf309c33b2088e : SUCCESS in 1h 07m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1ca0eeba922f41448f39836c8b065068 : SUCCESS in 1h 29m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/168339669ccd4f0582e0e21690f95d9b : FAILURE in 22m 34s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0d672d7f2e1f41ddab012b5e077aad11 : SUCCESS in 3m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e7e1205cf584a5e8178ea8d05093f1a : SUCCESS in 5m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5cf632dd5984434eacfcda25aacc5029 : SUCCESS in 21m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5152e8561f544d81a66328f1a954a738 : SUCCESS in 24m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a229abfed18844f093664e5af8e5137a : FAILURE in 11m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/3d7fdc1320764a3e912ea6f4c3f970eb : SUCCESS in 1h 14m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b05284f0d47e4ba381530d23571866c2 : SUCCESS in 52m 06s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3f9bd7d6d064a96913bd97da6b41499 : FAILURE in 24m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bd4cff3966fb4d359ce3ad0149d794b0 : FAILURE in 4m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/544efd8bbc7946b9968a7a347a1b66a8 : FAILURE in 5m 11s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c5afb27f6bb74597af09fd8fc8f004dc : FAILURE in 3m 21s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4b1b9b79b19e49aab4efc06d37019f75 : SUCCESS in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6efeab85298fa99d97d456a6de100d319795a7a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 20:52:11.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cb36b4d0ad8e01d37bb19ba62d9f11d309b684f4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 20:52:27.000000000","message":"Topic share_backup_fix_continue_progress removed","accounts_in_message":[],"_revision_number":4},{"id":"42692e3de2b95c24d43ece27e60a3e07c075901b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 20:52:35.000000000","message":"Topic set to bug/2125572","accounts_in_message":[],"_revision_number":4},{"id":"f9f007eb4c3278042ebd659975e57411ae5ec6a6","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 20:52:42.000000000","message":"Hashtag added: bug/2126844","accounts_in_message":[],"_revision_number":4},{"id":"f71620ad541d1b12ff8894f9a8e3d866ccda8bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-17 10:10:28.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fd6f646fd4ba010433c7f232a8184471a5ca5fa7","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-17 10:11:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4e4b713ae7d07c0ed1c18de850a3c3aa4f00099b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-17 10:20:02.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71bf59d436b4529d7fbf825af750b2011d615d20","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-17 10:42:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bfecc869e17ba8b30e380b82a067d06216313df9","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-17 11:22:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"05c5efc6d62f96c4ed51040b10145f6b0ae0e48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 12:38:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399480a7a982446ca4ff1e94112c9942\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0640cf86374c05943574e7360fb1df : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c34ef3ed70904c24989aa633463cf306 : SUCCESS in 31m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41019678f36940e2a35c5967aa92f88b : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4400d5e41d7145db8505a1216fffe310 : SUCCESS in 10m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/234ce57d32c049d6a8e9fafab864ad18 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cdd6a30afbb4db596c35a9cc74be290 : SUCCESS in 8m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/07c0c86d0ef949b3a90320212ae66a75 : SUCCESS in 58m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1dfdf88ccc5440179fc7d92b41943595 : SUCCESS in 2h 23m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3b95a18c1d4243c6b83ba02d6b2a2314 : FAILURE in 40m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9878ee6aef7c45269047364ba17ec1ab : SUCCESS in 1h 13m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e180bfd456904d38abddb53ca7bfb6e3 : SUCCESS in 1h 26m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/79f255f03b26483ea31400ff3a1278a2 : FAILURE in 22m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/642de5a95cc04e5690bf7f1576b2c48f : SUCCESS in 3m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a253ee64f4894bba845707bda6eb4cba : SUCCESS in 4m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/99f4a08f2c1f4c528f94be77302030e4 : SUCCESS in 33m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/606b384fbf2242f89442fd81c6e82749 : SUCCESS in 43m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9266d716e0064da28ca6d103d493e8e3 : FAILURE in 18m 21s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/85fbb9d23d5f47638021ac553de98efa : SUCCESS in 1h 17m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2f4657c299924e7bb96e2a01eba64bbd : SUCCESS in 54m 01s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5da04750438c4034b6b1c80820cdaf3e : FAILURE in 23m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7f7001d014a14e8ab88697efe43b4173 : FAILURE in 11m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3c8a2cb200864297ba5d60e3f689095c : FAILURE in 11m 07s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/ac61e3d4582a4ef9a4f98922800bc0e2 : FAILURE in 4m 50s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/11faca68af1c4103ada1dfb87810e637 : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a31a7dc6d518cbcf4433cc7e7a2bfd2997328d4b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-17 12:45:40.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e2ff7453ef8f35f63a30ee7479902cafb9eab841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 12:46:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2f3df5e49889a6f2b303964d6c0e4f6ab1fa4711","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 14:20:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55057e65b9a942c0b15fb4a7d15750d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd3bffe01ff4855ad87b7c4456e8369 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c10e6b947cb412aabeca40752b51114 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12dd8307e22a47ae967931b088debc11 : SUCCESS in 10m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4374a9a65d9e461e8196923a6908e8f8 : SUCCESS in 14m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d39bd0612c4a7492c0b88b33a818ef : SUCCESS in 9m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/974aba3442ee4f39b5285503d69fc4f7 : SUCCESS in 1h 30m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e756014dceee49bc9c4cc0178376b165 : SUCCESS in 32m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1050699256424da099083633f25a5a42 : SUCCESS in 33m 56s","accounts_in_message":[],"_revision_number":5},{"id":"24bbf13af48adbe284514bdcab2dc46509249b09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 14:20:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"70674d002e1538a4d179120904c4499dcbcbf411","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 14:21:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8df85c22385b429d99cd063a2fc55d7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15940225184246fe8704a891b03bff67 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5212ac410fd94595a9fbe236dfc28f58 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5},{"id":"fedb1fddcb8b7a265c97c7a403b815b6942acc6c","author":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"date":"2025-10-17 15:33:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dab20c3068039622c371ceb678036d2d162461d2","revisions":{"713052f46313e17cc7cee5c36c9826a9bcd483b7":{"kind":"REWORK","_number":1,"created":"2025-09-24 12:32:28.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/82/962182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/82/962182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/82/962182/1"}}},"commit":{"parents":[{"commit":"59ed29572d0e659aa720028cdde1f0d8bcb36a4d","subject":"Merge \"db purge: raise on missing tables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59ed29572d0e659aa720028cdde1f0d8bcb36a4d"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-24 12:28:37.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-24 12:28:37.000000000","tz":0},"subject":"Fix fetching of backup and restore progress for share backup","message":"Fix fetching of backup and restore progress for share backup\n\nEnsures both a external backup driver and a data manager driven\noperation can both properly pass on the progress of a restore\nand backup operation back into manila.\n\nFixes bug 2125572\nChange-Id: I8d964c575ee0a4638c5f43cfa89745c783e036ae\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/713052f46313e17cc7cee5c36c9826a9bcd483b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/713052f46313e17cc7cee5c36c9826a9bcd483b7"}]},"branch":"refs/heads/master"},"ca27b3b67f5a3b1d7952214335d03a079a33dc00":{"kind":"REWORK","_number":2,"created":"2025-09-26 14:45:44.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/82/962182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/82/962182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/82/962182/2"}}},"commit":{"parents":[{"commit":"59ed29572d0e659aa720028cdde1f0d8bcb36a4d","subject":"Merge \"db purge: raise on missing tables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59ed29572d0e659aa720028cdde1f0d8bcb36a4d"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-24 12:28:37.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-26 14:43:19.000000000","tz":0},"subject":"Fix fetching of backup and restore progress for share backup","message":"Fix fetching of backup and restore progress for share backup\n\nEnsures both a external backup driver and a data manager driven\noperation can both properly pass on the progress of a restore\nand backup operation back into manila.\n\nCloses: bug 2125572\nChange-Id: I8d964c575ee0a4638c5f43cfa89745c783e036ae\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ca27b3b67f5a3b1d7952214335d03a079a33dc00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ca27b3b67f5a3b1d7952214335d03a079a33dc00"}]},"branch":"refs/heads/master"},"319e04463615bd89bcd1dcc03a6638377b1e32e4":{"kind":"REWORK","_number":3,"created":"2025-09-26 15:01:12.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/82/962182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/82/962182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/82/962182/3"}}},"commit":{"parents":[{"commit":"59ed29572d0e659aa720028cdde1f0d8bcb36a4d","subject":"Merge \"db purge: raise on missing tables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59ed29572d0e659aa720028cdde1f0d8bcb36a4d"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-24 12:28:37.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-26 15:00:57.000000000","tz":0},"subject":"Fix fetching of backup and restore progress for share backup","message":"Fix fetching of backup and restore progress for share backup\n\nEnsures both a external backup driver and a data manager driven\noperation can both properly pass on the progress of a restore\nand backup operation back into manila.\n\nCloses: bug 2125572\nChange-Id: I8d964c575ee0a4638c5f43cfa89745c783e036ae\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/319e04463615bd89bcd1dcc03a6638377b1e32e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/319e04463615bd89bcd1dcc03a6638377b1e32e4"}]},"branch":"refs/heads/master"},"0aace4c07b7aa51485b6a965a419e763b2c6e40f":{"kind":"REWORK","_number":4,"created":"2025-10-04 15:54:50.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/82/962182/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/82/962182/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/82/962182/4"}}},"commit":{"parents":[{"commit":"59ed29572d0e659aa720028cdde1f0d8bcb36a4d","subject":"Merge \"db purge: raise on missing tables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59ed29572d0e659aa720028cdde1f0d8bcb36a4d"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-24 12:28:37.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-10-04 15:54:20.000000000","tz":0},"subject":"Fixes for targeted restores in the share backup API","message":"Fixes for targeted restores in the share backup API\n\nEnsures both a external backup driver and a data manager driven\noperation can both properly pass on the progress of a restore\nand backup operation back into manila.\n\nAlso allow the NFS driver to perform targeted restores by setting\nthe corrosponding flag to true.\n\nCloses: bug 2125572\nCloses: bug 2126844\nChange-Id: I8d964c575ee0a4638c5f43cfa89745c783e036ae\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0aace4c07b7aa51485b6a965a419e763b2c6e40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0aace4c07b7aa51485b6a965a419e763b2c6e40f"}]},"branch":"refs/heads/master"},"dab20c3068039622c371ceb678036d2d162461d2":{"kind":"REWORK","_number":5,"created":"2025-10-17 10:10:28.000000000","uploader":{"_account_id":36761,"name":"za","display_name":"zgoggin","email":"zachary.goggin@ecmwf.int","username":"za","status":"Manila share backup enthusiast at ECMWF :)"},"ref":"refs/changes/82/962182/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/82/962182/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/962182/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/82/962182/5"}}},"commit":{"parents":[{"commit":"59ed29572d0e659aa720028cdde1f0d8bcb36a4d","subject":"Merge \"db purge: raise on missing tables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59ed29572d0e659aa720028cdde1f0d8bcb36a4d"}]}],"author":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-09-24 12:28:37.000000000","tz":0},"committer":{"name":"Zachary Goggin","email":"zachary.goggin@cern.ch","date":"2025-10-17 10:10:07.000000000","tz":0},"subject":"Fixes for targeted restores in the share backup API","message":"Fixes for targeted restores in the share backup API\n\nEnsures both a external backup driver and a data manager driven\noperation can both properly pass on the progress of a restore\nand backup operation back into manila.\n\nAlso allow the NFS driver to perform targeted restores by setting\nthe corrosponding flag to true.\n\nCloses: bug 2125572\nCloses: bug 2126844\nChange-Id: I8d964c575ee0a4638c5f43cfa89745c783e036ae\nSigned-off-by: Zachary Goggin \u003czachary.goggin@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dab20c3068039622c371ceb678036d2d162461d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dab20c3068039622c371ceb678036d2d162461d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
