)]}'
{"id":"openstack%2Fmanila~963465","triplet_id":"openstack%2Fmanila~master~I223b035220788111502f4187dcb75f1d456c72f1","project":"openstack/manila","branch":"master","topic":"bug/2127023","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-10-17 22:18:46.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-10-16 19:57:35.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-10-17 22:18:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I223b035220788111502f4187dcb75f1d456c72f1","subject":"Fix shares getting stuck in ensuring status","status":"MERGED","created":"2025-10-08 21:36:32.000000000","updated":"2025-10-17 22:19:55.000000000","submitted":"2025-10-17 22:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963465-bug/2127023","meta_rev_id":"c93db453ddfe5a2cd788e0cfe9524f691742086d","_number":963465,"virtual_id_number":963465,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 22:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-10-16 19:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-10-17 13:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-10-17 20:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-08 22:12:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-10-08 23:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 17:48:42.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-10-17 13:40:55.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"5f2329f2a5f2e19dceebf3d938845bf233c722d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-08 21:36:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29dfff4c1c01db5051bd1e82b6937564552599f4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-10-08 22:12:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef9297a9706cae6793df43f624df2bd70040c3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 23:50:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b04f6cc6dd054f8bbaf7d74eaa16dcbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed752462d74400ebdcb2333685d9852 : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f3ef9f5b6c94888bd24639bce3fd899 : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c65f1f87f243e7834c9117634cf0be : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f773e5b3428d461aa6debe2a816dbbae : SUCCESS in 12m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/880c62e357754bc690cb506c9bbb65b2 : SUCCESS in 11m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/daf5124080744ce9aadfc9c7dff4cf22 : SUCCESS in 14m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3f68002de334c0d897482fc42002ce6 : SUCCESS in 10m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/45060f08a4a14de7846e0bcaf9f5bb53 : SUCCESS in 1h 14m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/635e39b88e414e439942347b694e288a : SUCCESS in 2h 11m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/49a876fae4fb400caff72c95c68f4afc : FAILURE in 48m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ba48dc7f0178461d8c2fd652b3de5742 : SUCCESS in 1h 12m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b87ae971caff4f54945d2c3fc071c250 : SUCCESS in 1h 07m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aa798594decb44c1be7ec88db56e0c9d : FAILURE in 29m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a9a46059e17e4a6a8362ac01c0ba5cb5 : SUCCESS in 4m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01098e60a8d04da69176cd3cf8118a0f : SUCCESS in 5m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e839575608de4f49aa89194bff1193e6 : SUCCESS in 39m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cbdce6719a314b87abfed4b272bb3ac2 : SUCCESS in 32m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/55d021cd036c429f95880bbf68ef68ab : FAILURE in 17m 24s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/38ed4d535e7e433895552a7e72a866b2 : SUCCESS in 55m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/06326ce8b6504d79848c0a2d107517df : SUCCESS in 54m 55s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ddc49d3fd0fc4d68b2950bb6104b0c4f : FAILURE in 21m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0ff0cf30abbf4b56871aa80d5a013a8f : FAILURE in 5m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0015e5fb14fd4ea79690dce819f90fae : FAILURE in 14m 20s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/333135c52d6448e1b6ce83ceee131403 : FAILURE in 4m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5a0d948377974e6b9ff889736eef5b80 : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c16f78d88109088d00d3f63252d121c0ba194412","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-09 17:47:50.000000000","message":"Topic set to bug/2127023","accounts_in_message":[],"_revision_number":1},{"id":"6e1fe801936605413cbc1e22f61c2563f27187a6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-09 17:48:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef3ef662c882184faa2045ee604e42ec81f47250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-13 18:27:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e86b30b729f56ac2b3b280a441d6021e70564cc1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-13 18:27:33.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3e1aca17cc4cae734fb29bda507a895aba6b59ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 21:13:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e95d65067fec4182b3562d1d28502c2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ca52fbbba04794a4540e06a6a5373a : SUCCESS in 9m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e78a202fabf4dadb14bb31b4bcdfc93 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc94a76909834cfa80c5f78d91d0aef7 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e8e221c51fb4c02874d78a366bffca5 : SUCCESS in 9m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/643f476467804a2b9df9d6f4f92f2014 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91a6d65cfc1042eda6893bb4f8d7fd7e : SUCCESS in 9m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61d3532d6ef44c8d9b8c2119a8a79922 : SUCCESS in 8m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d56d8698c4d6421e8fad265799b0ba22 : SUCCESS in 1h 13m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cb648f6622d940948bf7dca4ddf3d760 : FAILURE in 2h 40m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/011ba7e61d4942bcb666b396ecd0e091 : FAILURE in 31m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/20122bec641b485aa4d4745641ffb47f : SUCCESS in 26m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b3091d79f2e348caa6abe0ec650c3d22 : SUCCESS in 1h 42m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5f8ad708020b4f4184597f398ab44777 : FAILURE in 23m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9467e4c8edb04acba12db33d45e76a04 : SUCCESS in 5m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71fad438c90d4af0929be8e2c96e0d26 : SUCCESS in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/07f7606f886f429e9fed34b1d16ca1d9 : SUCCESS in 32m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3572e21586484185bfabd69994c1822d : SUCCESS in 44m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1881c11bb2cb43cb8f497cdc18bf2cca : FAILURE in 27m 29s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/f5eaa1d67214482684170256d0efde92 : SUCCESS in 1h 16m 09s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/68fcaa48e635418e870c933ff4034739 : FAILURE in 9m 18s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f773de677614a12accce183525d05a7 : FAILURE in 11m 19s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/840570faa4ae479b8201b96c7492e24a : FAILURE in 12m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a610bf54040243c7a03f3d36ffa8889a : FAILURE in 5m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/1930ea5316244b62b2b953b2f9b885ac : FAILURE in 4m 37s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/88edda4e35844f6f8f4d72a0769cce12 : SUCCESS in 3m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04bd4d15c83a650231cbedae0aae3c07ad7a5134","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-10-14 12:52:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54b30cd3439a8530cdea142e5ca2e9dc3290f86b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 14:00:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbc4cf86cc714b11866c4f2eece00693\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0891920f6ede4fd1928362b5ef578e74 : SUCCESS in 11m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aad7b3146244ede8fb2c3a1d71b1a06 : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2599716e98f34642be83c98ccec07fe7 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a789facbe14e4ac98f35fd0a47248769 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bc917df101442bdaf1ebfd6106d7e39 : SUCCESS in 7m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdde0b766d5a4c3c977c35105af147c0 : SUCCESS in 12m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd579bd1397443258de50dd9a01890ef : SUCCESS in 8m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/02ad1bf864d4428dbdf2113d9abf67e1 : SUCCESS in 36m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c1c8e43f7349404eb1513fa0439fd142 : FAILURE in 17m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6a052609cb5141e5a94028df9540012f : FAILURE in 50m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4bb2d147d9cd481ba41e94320981ed39 : SUCCESS in 28m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9ab58a938f9e4711b952e29c3805fe6a : SUCCESS in 56m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/77095a56fc524e80b6bcbe41f338258e : FAILURE in 10m 35s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f484617b3f8848c9b4337077a498f1ae : SUCCESS in 6m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b46fb8c7895247fcab9b65969fcb2f06 : SUCCESS in 6m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bbc7a906eee642868976b89b51ddc3e9 : SUCCESS in 32m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e7dcb2f65dcc4d18abe8f7afc4b42450 : SUCCESS in 45m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1773a514b2774211bb7e9c1549522ee7 : FAILURE in 24m 16s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2f33dc68cca14f1aba81a94dd191b363 : SUCCESS in 56m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3a771e95a48741819b0c4c226009af67 : SUCCESS in 55m 52s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6774e8793b24242b64267aa2faae1c6 : FAILURE in 24m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9712ea7225e24a0c939e318aaaea7196 : FAILURE in 8m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e77d27bcfa434a89a301a8ee8f4c0d1f : FAILURE in 14m 14s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b21cc9924d4b40adacf5fa675b4e7400 : FAILURE in 2m 47s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3454b9b287e942b9af962c2eeb2a0833 : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0bd31770880dcdecc27ae328f84bee4c4b6089c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 19:56:40.000000000","message":"Topic shares_stuck_ensuring_wrong_data_type removed","accounts_in_message":[],"_revision_number":2},{"id":"aa930c14ac53dcb2dacf810a17ca73d06d530b98","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 19:56:45.000000000","message":"Topic set to bug/2127023","accounts_in_message":[],"_revision_number":2},{"id":"e779e89b6b7f739e8162702042e630d93ae8c38a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-16 19:57:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a86837a039a302d27b17e48f986e44ccdd7711c4","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2025-10-17 13:40:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57ed6c2a3bf0a33eb811772a74d46c4cb4224d65","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-17 20:47:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c1e82be49a6251f16cb2f97e0bf08da4036cc53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 20:48:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8169b66f8f37681d9a10a7f5b4070b1a0222711f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 22:18:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/502bef90626f4f83b8426b475609b099\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cf4bf8936c4a0fbbcf70bc5ed07120 : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244bc393181f441c8ccec839302dbf59 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2018e5acc66d4038a4b08515ab4228ef : SUCCESS in 10m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e536262d1f2c4f26abcd31ec76ce16bb : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a13f63f309f4c2e93794b3b2a860396 : SUCCESS in 9m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/000a3141e7ff4b13b8c3feed0a56647b : SUCCESS in 1h 26m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a495c8e80c1a4b36896b4ca7cb37f03c : SUCCESS in 41m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2610cc5adc4242b696811dc0921402a5 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2961e923ef6cbf20c6a7f3f08273d75b7cef17ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 22:18:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c93db453ddfe5a2cd788e0cfe9524f691742086d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 22:19:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce6f3e1b67994fb1bc58ac4ad7d5ca99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971a85c3ad294c16be37f6f701768601 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6c6e9097ec1476899b7bec1a8d9362a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"795ff69c55ea686d2918c58f9e99dff5114c2e88","revisions":{"7528fbcf5333556c2136d3297ec9f6dc188bf59d":{"kind":"REWORK","_number":1,"created":"2025-10-08 21:36:32.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/963465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/65/963465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/963465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/963465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/963465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/65/963465/1"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-10-08 21:32:07.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-10-08 21:36:20.000000000","tz":-180},"subject":"Fix shares getting stuck in ensuring status","message":"Fix shares getting stuck in ensuring status\n\nShares are getting stuck in ensuring status when the share backend\ndoes not support the bulk ensure shares operation.\n\nAt the beginning of the operation we update the shares status but\nwe never revert the shares\u0027 status in case the updated_instances\nlist is not updated when the backend ensure the shares\nindivudually.\n\nThis change fixes that behavior by rolling back the status of the\nshares individually and also updating the service status when the\noperation is done.\n\nCloses-Bug: #2127023\nChange-Id: I223b035220788111502f4187dcb75f1d456c72f1\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7528fbcf5333556c2136d3297ec9f6dc188bf59d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7528fbcf5333556c2136d3297ec9f6dc188bf59d"}]},"branch":"refs/heads/master"},"795ff69c55ea686d2918c58f9e99dff5114c2e88":{"kind":"REWORK","_number":2,"created":"2025-10-13 18:27:31.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/963465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/65/963465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/963465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/963465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/963465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/65/963465/2"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-10-08 21:32:07.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-10-13 18:27:20.000000000","tz":-180},"subject":"Fix shares getting stuck in ensuring status","message":"Fix shares getting stuck in ensuring status\n\nShares are getting stuck in ensuring status when the share backend\ndoes not support the bulk ensure shares operation.\n\nAt the beginning of the operation we update the shares status but\nwe never revert the shares\u0027 status in case the updated_instances\nlist is not updated when the backend ensure the shares\nindivudually.\n\nThis change fixes that behavior by rolling back the status of the\nshares individually and also updating the service status when the\noperation is done.\n\nCloses-Bug: #2127023\nChange-Id: I223b035220788111502f4187dcb75f1d456c72f1\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/795ff69c55ea686d2918c58f9e99dff5114c2e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/795ff69c55ea686d2918c58f9e99dff5114c2e88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
