)]}'
{"id":"openstack%2Fmanila~964355","triplet_id":"openstack%2Fmanila~master~I308d942ae460e881cf37f95f393c799c75d5bd62","project":"openstack/manila","branch":"master","topic":"bp/json-schema-validation","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-11-22 00:27:47.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-11-21 21:18:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"38358":{"account":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"last_update":"2025-11-22 00:27:47.000000000","reason":"Change was submitted"}},"hashtags":["json-schema-validation","openapi"],"change_id":"I308d942ae460e881cf37f95f393c799c75d5bd62","subject":"Add schema validation for Share Snapshot list API","status":"MERGED","created":"2025-10-18 22:00:55.000000000","updated":"2025-12-05 18:38:01.000000000","submitted":"2025-11-22 00:27:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964355-bp/json-schema-validation","meta_rev_id":"db07d7be9962dae1a3137f31e5315d58491143e4","_number":964355,"virtual_id_number":964355,"owner":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-22 00:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-11-21 21:18:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-11-21 21:18:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-18 22:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-26 15:32:31.000000000","updated_by":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-11-07 20:20:08.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"3f1d658d6046fe99fcbe8274080552ba07e28e62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-10-18 22:00:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a017508dfd2a92601c2a3c38d6703f7f8aa0b240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 22:47:51.000000000","message":"Patch Set 1: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21e24fd89481405ea6eee3c5bdd1e2fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8135ee2373ba433eb4907f924552caf7 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec9577c091c843a789deec71650b107d : FAILURE in 23m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3d95d3b70b4e37afba71a448e2bb80 : FAILURE in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/187e7f8419cc4e5ea1fe9ea291bd2175 : FAILURE in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c50900d6c0242f9b55e899b1a4f6a47 : FAILURE in 6m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/111c52d9ca8b42a8b86dfbf0f95f07c9 : FAILURE in 44m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/82d1aac3b1de4056be5425b696df7ea7 : FAILURE in 45m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ffc51ac5e7c64470830f236714d5f9b6 : FAILURE in 26m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/761b0cf8b8f344dcb9446f66bb53eb79 : FAILURE in 37m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/25eb04d30f3f4456822c64b37e5692ae : FAILURE in 30m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c32ad85b883243ee80f2b63e81f4cc23 : FAILURE in 14m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9e0f27a4212c4d82a26ce343f485a2d4 : SUCCESS in 6m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f6dd22f8b9e749b694b897675a2bee55 : SUCCESS in 4m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/faeeae09e48342798b7af4b5c3ce56cd : FAILURE in 15m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/817967cb3d4146f080e5de32abb497ab : FAILURE in 11m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bebaf29fc54c476395d36dd0c79ddf6d : FAILURE in 15m 47s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/82825055f5ea4d0fa223f2b06ec1ef84 : FAILURE in 22m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e75bc01340a94660af8e980a8c265f1d : FAILURE in 41m 47s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48c3aa5117bf420181a826228a5657ab : FAILURE in 23m 11s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cb2e65f2a40543b28883575eb79474b3 : FAILURE in 11m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b5dc7cfa44c44bb09ca030632c2c6c04 : FAILURE in 10m 43s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/beafcb14c6c5415abdb071c30fd5e608 : FAILURE in 4m 24s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/331f6e5bbb14408299492f4db2181164 : POST_FAILURE in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83298195a53053ebbe6373429424fa10fcd7d96d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-10-19 17:19:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"66f0e373f21027fd85a9ba49c39a195df0fe4861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-19 18:57:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4a0a48662d24e2e93ef07ed65c052b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad8deafd29c4caeabd5ac3217479de5 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3109f24c08e64c24bf0c69d14285778e : FAILURE in 20m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f0b391d9764940a587c3adb6eb2437 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/619bdc3f220c45fbbde3334ce9a2ffab : FAILURE in 8m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26f0705f927d4fb88971ddbffae364d7 : FAILURE in 13m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c8bfa1fb1f864eea9eecada86809e937 : SUCCESS in 52m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9dd4a37f9e3042c5a1499a1540460789 : SUCCESS in 1h 36m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c2837d211ff443a94c58d52e0df1905 : FAILURE in 30m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e106f0c995334d61ab52e33181a36c79 : SUCCESS in 1h 00m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ffb162bbe4bb44b4b0bc37ffa2808338 : SUCCESS in 46m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/64b7a600a3284c7da2d861fb9dceb820 : FAILURE in 22m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b7a83c1ab8464b6b8257d874d8bf2fb4 : SUCCESS in 7m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/294078d80bda46b58f69b1f309ddf3b8 : SUCCESS in 4m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a6218cd86b94c57b56d29d2c2ee09c1 : SUCCESS in 19m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5845d5414e13498799dbe3945df2c675 : SUCCESS in 40m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5733b27b9e4040efaff831a24cfc6247 : FAILURE in 16m 46s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/48736d8d318840d58055d4fd30d48681 : SUCCESS in 1h 14m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8840bcba6e304ed4a1df4a9d5e282566 : SUCCESS in 35m 17s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6815d762cf1f4988bd467df57386cbcf : FAILURE in 22m 38s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4ad7c18e1a434284a4f672d960107393 : FAILURE in 5m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/348bdec16d7546ddaf43b66ab33bb067 : FAILURE in 6m 10s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/cfff50a2f177464bb941757d0dc2df2c : FAILURE in 4m 58s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/18b3413808484a06bf756e265aa65583 : FAILURE in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eff05a1b52caa1e9198423684e5d77847ba1e3f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-11-02 00:05:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"225732d373d97adc5a4c41d832b66121b5e1b9e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-02 02:16:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d67b91219b945c184fef2adc1d3e1d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27df0f9cc8304b03a01502185d6cd663 : SUCCESS in 12m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0359cd0ea1034940bdc4c4ebfc5bbed6 : FAILURE in 31m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61411ed438943189fb04a5912bcedee : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b05211a1922b4cedb5f65bfb85b21013 : FAILURE in 9m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2400ccda8d2b48309f44904f9795fe55 : FAILURE in 10m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a76436ece9d145c090cdd85bc884d710 : SUCCESS in 46m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b1c0b1e3535e438da2420fcb9c8cb1e8 : SUCCESS in 2h 05m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8e9281eaf15d41ffbbbd93d0f588c243 : FAILURE in 52m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0d6ee42b17374722994535014ceeb1ed : SUCCESS in 1h 05m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8d89edb88be74e60a732c27a58a00065 : SUCCESS in 1h 24m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d047be1810fe4efb9c8933dfdc5b716d : FAILURE in 23m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b58ee4c623c84c7d9f011cad5cfc98bb : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0b4f64846274c8db4de72e2ee1bcad3 : SUCCESS in 4m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3beb81b3ef364364a73c11c9cca7f709 : SUCCESS in 27m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/03e1447566864e63aa22ddec28c620bf : SUCCESS in 44m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a6b4abad407742f28e6bd23b119070b7 : FAILURE in 26m 12s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/b535c46a2b6f40efbfd38b9fc89e9394 : SUCCESS in 51m 28s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b4ce48cac1ef40be9df0990c183a4d4c : SUCCESS in 57m 12s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ff897566b04490581f689ba8256f380 : FAILURE in 22m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2c97e59aafac4f0b83c2e927a54e7107 : FAILURE in 10m 16s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3bed29831ccb47dca1b8e367ba1c004d : FAILURE in 5m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/d73af0ac70c24b17ace89efd7d4cf4a9 : FAILURE in 4m 26s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/51c24806497b4ad686476b04f9891dbd : FAILURE in 6m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dd7e25513f4048d01b1b267e28f9e0d86497c5a6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-07 20:20:08.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"42d46ab7e84fdfcc11b914306719266b04a00db7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-07 20:20:15.000000000","message":"Topic set to bp/json-schema-validation","accounts_in_message":[],"_revision_number":3},{"id":"5714a206af0c6d3a3b9e5e6fef66334fd38af503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-11-10 17:52:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8f5bb34279b2c908da0e887170e1691fad957e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 20:42:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75534e852a3f446e9c70d93512f5c7bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcac0df5f598423d985f30f8c67f3aa0 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c2845bc151941aa9b85b62d89222b6b : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15bd44ee5a45478ebc8f56070bd762de : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/877d1eb613b442939d788025466af35d : SUCCESS in 10m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a13d0663b1254f68952fc667ea963b5b : SUCCESS in 13m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6030b8a728df4957a6e5ba189fc85886 : SUCCESS in 1h 12m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/18d298e5ee714cf3a3a99286f62d4c65 : FAILURE in 2h 44m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b09d1bfc324140d6906fdd8eac79b44e : FAILURE in 1h 08m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/04c504e468c446c192a3b27f47893729 : SUCCESS in 1h 00m 38s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2533db3d1dc74bd59acc0ff90130a051 : SUCCESS in 1h 52m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2ed2f6ef879c4851a507a62bf1edd365 : FAILURE in 24m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/17928c4f74ea4016af33d787d3a06a0b : SUCCESS in 3m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ff5d8d979904779a2dbd79c9c8fe50d : SUCCESS in 3m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b5ff482a57b3473f864355007427aafc : SUCCESS in 26m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/378f7d7b2ba8475a8caee6d8a03f80a7 : SUCCESS in 40m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fb9b79dba23444b3881dd771aececa6d : FAILURE in 25m 58s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/f53b3b0ea16a415b92ad09fdaf1632ef : SUCCESS in 52m 47s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3dda8662d4744680918ae7b3b9d8320a : SUCCESS in 50m 00s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0132fd0ecaa1428e8857c5a90a7c0c81 : SUCCESS in 51m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9f25278680dc42659b2c5ff3b145965b : FAILURE in 5m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2b21d850c14149c8b1abd01ecfe5e0fb : FAILURE in 5m 13s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/75bba1ca57214142958a7a3b6cd2dc2a : FAILURE in 4m 22s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0a74358d9c2240bc961cb3e5e6607e42 : FAILURE in 7m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"57aaed577a73fcc46f69becfcb4b84ec15af89c2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-12 13:50:37.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"265ce1fcce726670b95883d335afd728ff817076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-11-12 20:31:51.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d5da5198fd83861a4cc6d334c7796f4183244c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 22:23:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97bb013ed4324caf81194aefaf17954c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a465b14c004e3da81f5c593e82aa9b : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0dde2d4846b487db69bbcef3f8678e2 : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42da824a1c994c91ac22deaf0f21f9ae : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e0bc8ebdeee4576b619b612f715ac87 : SUCCESS in 8m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57e60b100a234f5b91e8f4e01e2a1a3b : SUCCESS in 12m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/76cc62e8c9d946d7b45b5d8f24b0da27 : SUCCESS in 1h 10m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bc97457bc6df43f3bb3b446d6f56f716 : FAILURE in 1h 50m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/04ac75252a2b4aebaba45f3a80dfb610 : FAILURE in 49m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f511806ec88e4709a187d27a3fd1b5a0 : SUCCESS in 1h 04m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/13fa4e19b1084b36936b2134890d0618 : SUCCESS in 1h 31m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4a00ca0894ed47b9af447334e30953d2 : FAILURE in 20m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7ea2432a72d84436b05812efdc2a5818 : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec3e977a7c0046bbac719fef4a7b0573 : SUCCESS in 3m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a0df67ee1f004417a4fb50a0063057bb : SUCCESS in 27m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7637f3c55dae4bf48a6f81118aa12e0f : SUCCESS in 45m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/573e528ebb4b4ebca54d6588c7fde3e2 : FAILURE in 24m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/4a24044d748e401c9b34f6e4b202cb1a : SUCCESS in 1h 19m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d618076d20c74967a509b4961e4428da : SUCCESS in 50m 22s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ad3b53d53e041529ecb4985eab945db : SUCCESS in 46m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/095b9ca004da4816a017d7d7ca2ec394 : FAILURE in 5m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f89093a65ccf4d77a6125a6a1baba8e2 : FAILURE in 7m 01s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/e2b17ff4a6bb42f9b015b10c0a0db6e3 : FAILURE in 6m 18s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e0cab40ba0d04e5189a941d61fe3b7ec : FAILURE in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"16371c4bbec26793e6267812490c86161cedb4d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-14 17:32:08.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c6df168bcbf1becb2ef6bb2aabbc8dad93dbb658","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-14 17:32:24.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a30150e2ebbb39443f36fc8012a8d80c9165eab9","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-11-14 17:37:02.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4bb9d94a098c8a8f7cb72ccbfc68b25f77fdd2dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 19:25:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58469d31f9444019b1077175e4cb5967\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4c4c6325a540fda648f64718c675e1 : SUCCESS in 12m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffafdb345d3a47cd8789358188766f0d : SUCCESS in 32m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b468e37a117489e9c44eaa955493331 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a3170f8c9e14ef88aee6e11f34e71dd : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37b1db8cc4ac443488ff76057f58adfb : SUCCESS in 8m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/acdae50d3ae346dd800bc6106d317774 : SUCCESS in 55m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b78c8e0772604d25b86fb7894315d1b8 : SUCCESS in 1h 43m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9ba5e8925e534425b2c4b500ba8ec3c7 : FAILURE in 41m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5b9cc0589a2d4fae831a239757661016 : SUCCESS in 34m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8ef87d72e32b42afad0da4192df3c6a7 : SUCCESS in 1h 40m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/543e6eeff8514e39b54fd401817c5870 : FAILURE in 25m 19s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1f331d5eb9ff4a2d8ad57161c7ae86b2 : SUCCESS in 5m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9e5543058824426999a8d0f8b5b181d : SUCCESS in 3m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c6345e31a81b43b58c3bbabdd54ca0cc : SUCCESS in 35m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/67f6b5fab7184ee18e5ec3604d75f6a8 : SUCCESS in 45m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9e45be9926db49479fb6bafa4fdd1b7c : FAILURE in 14m 11s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/6373ac48180e48c9a97d9f685bf7bc90 : SUCCESS in 40m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2a9eae9365414a8b9d2e2c3290e1c0d4 : SUCCESS in 47m 03s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20d09652017943d3917881de0796bf2f : SUCCESS in 42m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9e498f6ba29742d2949a6050be42d6c3 : FAILURE in 12m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1f98b01ec2ff4ae29fe48510df5f505e : FAILURE in 13m 36s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/373f2d10c45349aca628a19bc620ebc4 : FAILURE in 6m 30s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e83a6f0e841f4c169f878742afd1b43e : FAILURE in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b5b5a55540451e679b04c262a89a77eb2423c287","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-14 19:45:45.000000000","message":"Hashtag added: openapi","accounts_in_message":[],"_revision_number":6},{"id":"55e49501f5efe94870a471611460f0668055b654","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-21 19:41:44.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f07084665029268a03de81cea44f7d10be8e6b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-11-21 20:40:37.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"8d3f5ac036a8623b7a1fa6c7978fdaf38009ed6b","author":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"date":"2025-11-21 20:41:56.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"dd4431fd4dae70800cccfb341f35095513d58ee9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-21 21:18:37.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c79640cd65e6c2c817a278b582e0f6ca90066863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 22:50:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e02fb848b2134ab18164f94be36930e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16478829ca9041cdbb6adf8e0093c848 : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fc850a494b7480582092c87c112f2fe : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99267b1cf314b63a75ada0cdd9953c9 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c50f8df9abd3490aaaf98ea6662dad68 : SUCCESS in 8m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c4b96f871344dbdbabf77e149890abe : SUCCESS in 7m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c5a218aae4674d359469a564d983483d : SUCCESS in 1h 10m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8d09fc02797a473ca20145ce499eb623 : SUCCESS in 2h 07m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ac2abea215b84267976a58892c1197d6 : FAILURE in 36m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/961611a315854752b2116709bb2d9044 : SUCCESS in 48m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c8f974bd00d047fbb961abca8b8510ac : SUCCESS in 48m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1d6069a19ba4477a91df5fca7b86ce5d : FAILURE in 23m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a2485530da54470a8f53467ab65e7e0d : SUCCESS in 2m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c580dc10aaae48eaa6a26d7f546c196f : SUCCESS in 4m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ff060ebe4e9849bbb5b5a7a78d81b4de : SUCCESS in 33m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6e5a5257df6a44eaa162c3158339b4ea : SUCCESS in 28m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/59a481472aae4ce6ba04603632efda39 : FAILURE in 12m 47s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/41f34ada8aa34e678ea4a256e0c97460 : SUCCESS in 36m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/39228aa190d34699aa534da0871ec19d : SUCCESS in 59m 24s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d11bfae62e614febb48315b6d28e246d : SUCCESS in 56m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f98d1dc98bde421399c824eb81f4a158 : FAILURE in 11m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f0765af2a5754618b4cd798dc2ecb1b9 : FAILURE in 6m 25s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a5e559cc380b46d1b6884e667a2a19a1 : FAILURE in 5m 08s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/38ca1c094e0c458c84b648c3852c8954 : FAILURE in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2bc7e4010b9052c9942505a9496177333a1718da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 22:50:28.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2f8427de2cf28125ff4a6ac9fa0499aa4d3e3f20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 00:27:46.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fd2b1d5a2224e83a778e7c004eae9ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de2e3ede04649d997aaa5f87f8b33ba : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d857fb8039fb4187baad403ba48b0a3e : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a28d6f094dcf48eeb203d73d9b84ce09 : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10750404963d42cea896c609a8c39d3e : SUCCESS in 10m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/729cc2d0d4d7471396ca469e129bf7b5 : SUCCESS in 1h 17m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8f0c4cf69eb14ba7bc3291f92cbb443a : SUCCESS in 32m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/99cd104eb6374838bd0d539f92e704ae : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":7},{"id":"a942c379199c4fbc6ecb4faa02a878fbebe1bf91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 00:27:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"4d228e4c218aa853c54cec2280551643b8a579fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 00:28:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d812d616b7a4e298ef27913748a5ce2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf2848170a64f4fa3077fcc82368218 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":7},{"id":"db07d7be9962dae1a3137f31e5315d58491143e4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-05 18:38:01.000000000","message":"Hashtag added: json-schema-validation","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e07ea29033a0e42acd31f8d482b7a0c06105a934","revisions":{"0ecbacdb21fe2343df6970f81216d29cd2087ab8":{"kind":"REWORK","_number":1,"created":"2025-10-18 22:00:55.000000000","uploader":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"ref":"refs/changes/55/964355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/55/964355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/55/964355/1"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 21:59:17.000000000","tz":-240},"committer":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 22:00:54.000000000","tz":-240},"subject":"WIP added list share snapshots schema and validation","message":"WIP added list share snapshots schema and validation\n\nChange-Id: I308d942ae460e881cf37f95f393c799c75d5bd62\nSigned-off-by: Marco Wehrhahn \u003cmarco.wehrhahn@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ecbacdb21fe2343df6970f81216d29cd2087ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ecbacdb21fe2343df6970f81216d29cd2087ab8"}]},"branch":"refs/heads/master"},"90ec5ba0f6c3797e7b484d818c2a7671302b3423":{"kind":"REWORK","_number":2,"created":"2025-10-19 17:19:56.000000000","uploader":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"ref":"refs/changes/55/964355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/55/964355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/55/964355/2"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 21:59:17.000000000","tz":-240},"committer":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-19 17:16:35.000000000","tz":-240},"subject":"WIP added list share snapshots schema and validation","message":"WIP added list share snapshots schema and validation\n\nMade the index_request_query a single base dict\nFixed some styling to pass the pep8\nRemoved the body response validation\nImported cfg from oslo_config\n\nChange-Id: I308d942ae460e881cf37f95f393c799c75d5bd62\nSigned-off-by: Marco Wehrhahn \u003cmarco.wehrhahn@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/90ec5ba0f6c3797e7b484d818c2a7671302b3423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/90ec5ba0f6c3797e7b484d818c2a7671302b3423"}]},"branch":"refs/heads/master"},"0d9577ac56bb9c93a3e4f924c4a7f634cdbd4828":{"kind":"REWORK","_number":3,"created":"2025-11-02 00:05:42.000000000","uploader":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"ref":"refs/changes/55/964355/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/55/964355/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/55/964355/3"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 21:59:17.000000000","tz":-240},"committer":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-11-02 00:02:49.000000000","tz":-240},"subject":"WIP add microversioned schema for Share Snapshot API responses","message":"WIP add microversioned schema for Share Snapshot API responses\n\nThis patch introduces version-specific response schemas for the\nShare Snapshot API to align with behavior changes across\nmicroversions\nChange-Id: I308d942ae460e881cf37f95f393c799c75d5bd62\nSigned-off-by: Marco Wehrhahn \u003cmarco.wehrhahn@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d9577ac56bb9c93a3e4f924c4a7f634cdbd4828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d9577ac56bb9c93a3e4f924c4a7f634cdbd4828"}]},"branch":"refs/heads/master"},"3699517da285acc444dc575e0e10892e72135f53":{"kind":"REWORK","_number":4,"created":"2025-11-10 17:52:35.000000000","uploader":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"ref":"refs/changes/55/964355/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/55/964355/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/55/964355/4"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 21:59:17.000000000","tz":-240},"committer":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-11-10 17:47:08.000000000","tz":-300},"subject":"WIP add microversioned schema for Share Snapshot API responses","message":"WIP add microversioned schema for Share Snapshot API responses\n\nThis patch addresses the comments from Goutham.\nIt adds 4 different microversions.\nRemoved a previously added validation wrapper in patchset 1 that was leading to build errors.\n\nNext steps:\n    - add request body schemas for create and update operations\n\nChange-Id: I308d942ae460e881cf37f95f393c799c75d5bd62\nSigned-off-by: Marco Wehrhahn \u003cmarco.wehrhahn@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3699517da285acc444dc575e0e10892e72135f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3699517da285acc444dc575e0e10892e72135f53"}]},"branch":"refs/heads/master"},"00dfded977e5238249f1e05ecb060f95d754551b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-11-12 20:31:51.000000000","uploader":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"ref":"refs/changes/55/964355/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/55/964355/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/55/964355/5"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 21:59:17.000000000","tz":-240},"committer":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-11-12 20:31:51.000000000","tz":0},"subject":"api: Add schema validation for Share Snapshot list API","message":"api: Add schema validation for Share Snapshot list API\n\nIntroduces JSON schema validation for the share snapshot list (index) API across supported microversions. It defines request query and response body schemas for versions 2.0 through 2.79, covering pagination, sorting, filtering, and count parameters.\n\nPrepares the groundwork for full controller validation, which will include request body schemas for create and update operations in a follow-up.\n\nChange-Id: I308d942ae460e881cf37f95f393c799c75d5bd62\nSigned-off-by: Marco Wehrhahn \u003cmarco.wehrhahn@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00dfded977e5238249f1e05ecb060f95d754551b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00dfded977e5238249f1e05ecb060f95d754551b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fabbcb776c2f4ae7796f346016819d78435d963c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-11-14 17:32:08.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/55/964355/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/55/964355/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/55/964355/6"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 21:59:17.000000000","tz":-240},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2025-11-14 17:32:08.000000000","tz":0},"subject":"Add schema validation for Share Snapshot list API","message":"Add schema validation for Share Snapshot list API\n\nIntroduces JSON schema validation for the share snapshot list\n(index) API across supported microversions. It defines request\nquery and response body schemas for versions 2.0 through 2.79,\ncovering pagination, sorting, filtering, and count parameters.\n\nPrepares the groundwork for full controller validation, which\nwill include request body schemas for create and update\noperations in a follow-up.\n\nChange-Id: I308d942ae460e881cf37f95f393c799c75d5bd62\nSigned-off-by: Marco Wehrhahn \u003cmarco.wehrhahn@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fabbcb776c2f4ae7796f346016819d78435d963c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fabbcb776c2f4ae7796f346016819d78435d963c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e07ea29033a0e42acd31f8d482b7a0c06105a934":{"kind":"REWORK","_number":7,"created":"2025-11-21 20:40:37.000000000","uploader":{"_account_id":38358,"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","username":"marco-wehr"},"ref":"refs/changes/55/964355/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/55/964355/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/55/964355/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/55/964355/7"}}},"commit":{"parents":[{"commit":"40107f961671fd55710f70eef30ea526298b5e22","subject":"Add jsonschema and bump oslo.db requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40107f961671fd55710f70eef30ea526298b5e22"}]}],"author":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-10-18 21:59:17.000000000","tz":-240},"committer":{"name":"Marco Wehrhahn","email":"marco.wehrhahn@gmail.com","date":"2025-11-21 20:39:53.000000000","tz":-300},"subject":"Add schema validation for Share Snapshot list API","message":"Add schema validation for Share Snapshot list API\n\nIntroduces JSON schema validation for the share snapshot list\n(index) API across supported microversions. It defines request\nquery and response body schemas for versions 2.0 through 2.79,\ncovering pagination, sorting, filtering, and count parameters.\n\nPrepares the groundwork for full controller validation, which\nwill include request body schemas for create and update\noperations in a follow-up\n\nChange-Id: I308d942ae460e881cf37f95f393c799c75d5bd62\nSigned-off-by: Marco Wehrhahn \u003cmarco.wehrhahn@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e07ea29033a0e42acd31f8d482b7a0c06105a934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e07ea29033a0e42acd31f8d482b7a0c06105a934"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
