)]}'
{"id":"openstack%2Fmanila~964749","triplet_id":"openstack%2Fmanila~master~Ic223080867e1e12b0e2f1c4ed02f1c92ef5b50e0","project":"openstack/manila","branch":"master","topic":"bug/2031194","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-11-13 18:44:54.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-11-03 22:48:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-11-13 18:44:54.000000000","reason":"Change was submitted"},"38469":{"account":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"last_update":"2025-11-13 18:44:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic223080867e1e12b0e2f1c4ed02f1c92ef5b50e0","subject":"Reduce NetApp unit test runtime: time.sleep","status":"MERGED","created":"2025-10-24 06:45:34.000000000","updated":"2025-11-13 18:46:51.000000000","submitted":"2025-11-13 18:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964749-bug/2031194","meta_rev_id":"c8b439a3f77bf7507810aec4ea6db135c6791d1f","_number":964749,"virtual_id_number":964749,"owner":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-13 18:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2025-11-13 17:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2025-11-13 17:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2025-11-13 17:26:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-24 07:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-24 16:24:31.000000000","updated_by":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-10-24 16:24:31.000000000","updated_by":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-10-27 20:55:42.000000000","updated_by":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REMOVED"},{"updated":"2025-11-03 22:48:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-11-07 20:48:27.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"2760808be872b100c967ba142862505c04fc92b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-24 06:45:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"348195c5804b7f4e06f606a8462f9c84eb42500c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 07:05:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff6a6118b9f4bc99cb7c59ccb02d03c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8af49497a34ac9874b352ed459992e : SUCCESS in 11m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12c17076b67b495fbeb04cdc62da0b09 : SUCCESS in 19m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085d14f72cb2448f81a248c47e5b4127 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a69a87c134449a0a792cd79cb6ea312 : SUCCESS in 9m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b38798a88bdf44ab85e846fe043ed64f : SUCCESS in 7m 37s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4d0af9071f4c42c59982377b4e5bc111 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/07c26900462d4abdbdff2be51a901e74 : SUCCESS in 4m 39s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/309f51c443634282bffc076004044da7 : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"567ce78b96455cdf4cec80d55e7bd862c8c8a754","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-24 22:22:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad00a37f3ffa77ab6461fc96c154b813815ebe6c","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 12:11:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0d905960990627c182b309d5284b4fa1888f153","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-27 14:45:11.000000000","message":"Topic Speed_up_NetApp_Slow_Unit_Tests removed","accounts_in_message":[],"_revision_number":1},{"id":"1bbd796d20cba77197a506d5f5853e013a3c57d3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-27 14:45:19.000000000","message":"Topic set to bug/2031194","accounts_in_message":[],"_revision_number":1},{"id":"ba176ede756e085906410d9bc8534e52b011a879","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-27 14:48:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3798a04201bb138c16e9af03c93c9c02994f870","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 18:37:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b299bff5cf89e0e785d46b19e893e188c699c4da","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 18:38:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a4ba7385fc9e0e25518084233b2d9bd1b7e17ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 20:51:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"328863dae699cd00d30751f328fc88b1c39cda61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 20:53:35.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"81fc2ab2cc748402a0e90009e33b1a3eeb3f411f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 20:55:17.000000000","message":"Topic Speed_up_NetApp_Slow_Unit_Tests removed","accounts_in_message":[],"_revision_number":3},{"id":"2b9fc36dbf0ae94cb52d9a2ad8613058bd036226","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 20:55:26.000000000","message":"Topic set to bug/2031194","accounts_in_message":[],"_revision_number":3},{"id":"fdc70aa918ca468b8303c995ac5705a391c1c8bb","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 20:55:42.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_29632\u003e.","accounts_in_message":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"_revision_number":3},{"id":"0241ee537579d2871c93ac2988b65ebca266422e","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-10-27 20:58:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"80f721d5c03b16102a53389587db6442e83fedbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 21:16:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5634c952400f485a994c1fb540459ed5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6c748fc685444ea02e2b8cef099949 : SUCCESS in 11m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb9f283d9ea841caae12f4046baacfc6 : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f45bad0b644c9fb326e4af65018288 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e039e831a3d54d0dbcef0a2e43e8e62b : SUCCESS in 8m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ded57f03ac694d46bca8b32145ffc736 : SUCCESS in 6m 48s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/22aa1d23760a485685e0bfa48f7d67f6 : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b58b4d88afa4063807bfc09bc9a8cd9 : SUCCESS in 4m 30s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/356c4f57b71f440db3508c42562642b8 : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02c8ce24d6daf6a9b97c58babaa1e9fcc3384fee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-03 22:48:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1533b1360fa9abab0025a327b1ad2d182423fa06","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-11-07 20:48:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8992791180e23435e642245b288ba895d5da83d7","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"date":"2025-11-11 13:30:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"59c25db8db25677b2956ef709d88a4fdb6769563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"real_author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-13 17:26:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_38469\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"}],"_revision_number":4},{"id":"28f732757efb10b271d5534a56b8df07ad397ff5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-13 17:26:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"960dc48732858272421708b2fe9601f1b8332090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:04:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb711ce1c2014a6a945e3869298d8849\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc40d36f8394f8b9d491e7f5a46bf93 : SUCCESS in 11m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c2a90bbcc1f437da1c56ca3a61a6e04 : SUCCESS in 33m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62f6ac5a5dd64bd0b9e0adf73d251d36 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1ae1b13c28a416bbbc7198f0e5214ee : SUCCESS in 10m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a95297330274a4bb85db99e6bc9a85d : SUCCESS in 13m 59s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e854f1aa43a24e4085ea885bb3b1d604 : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e6ccd4ecdde46e5bed9f28b5acc2a3f : SUCCESS in 4m 38s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3990efa7348b4d7198c7074bf7ef1ac1 : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edd0437961145c7a25f52468fa903f4c2574474e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:04:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9bb594a9a5df1f09a2e9eeb5c62e738fd1546d06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:44:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67753d94a6754320b10df5feba959ac1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1640b4473a2455c8d54a34188597875 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2655fce1484d491486ad3f78c74356ad : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc145039128542639c7275c3f6a5f0b1 : SUCCESS in 6m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b54a644f2cc24e4e9f451287bf2206cc : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":4},{"id":"7c4fd9fb2db557501f7abc95270ff53dbd088ce0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:44:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c8b439a3f77bf7507810aec4ea6db135c6791d1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:46:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e8e5577eaa44221814d55688c9121f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f585ec02fea44651bc886298a22f4586 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8c5874ac77130e303b4e7e3622727d17042fe33f","revisions":{"bcbbb6d4e94fdd80b216b8ac86cb1a5cda0e1d90":{"kind":"REWORK","_number":1,"created":"2025-10-24 06:45:34.000000000","uploader":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"ref":"refs/changes/49/964749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/964749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/964749/1"}}},"commit":{"parents":[{"commit":"1529ced1a7ce90f45fc6143db78619e4704ece78","subject":"fix netapp delete_share private_storage cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1529ced1a7ce90f45fc6143db78619e4704ece78"}]}],"author":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-24 05:52:33.000000000","tz":180},"committer":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-24 06:43:40.000000000","tz":180},"subject":"Reduce NetApp unit test runtime by mocking time.sleep","message":"Reduce NetApp unit test runtime by mocking time.sleep\n\nSome NetApp driver unit tests take excessively long to run\n(up to 58 seconds)\n\nThis change introduces a mock for time.sleep in the affected\ntest class (NetAppRestCModeClientTestCase) to eliminate\nreal waiting time without altering test logic.\n\nThis reduces overall test runtime.\n\nTest Plan:\n- Verified all NetApp unit tests pass with no timing delays.\n- Confirmed strestr slowest no longer lists test exceeding 1s.\nOn testing, the slowest test took  0.359seconds\n\nCloses-Bug: #2031194\nChange-Id: Ic223080867e1e12b0e2f1c4ed02f1c92ef5b50e0\nSigned-off-by: Rose Kimondo \u003crosewairimu1@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bcbbb6d4e94fdd80b216b8ac86cb1a5cda0e1d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bcbbb6d4e94fdd80b216b8ac86cb1a5cda0e1d90"}]},"branch":"refs/heads/master"},"636d514cdba8644b1849f071107996c30dc280d7":{"kind":"REWORK","_number":2,"created":"2025-10-27 20:51:03.000000000","uploader":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"ref":"refs/changes/49/964749/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/964749/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/964749/2"}}},"commit":{"parents":[{"commit":"1529ced1a7ce90f45fc6143db78619e4704ece78","subject":"fix netapp delete_share private_storage cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1529ced1a7ce90f45fc6143db78619e4704ece78"}]}],"author":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-24 05:52:33.000000000","tz":180},"committer":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-27 20:42:07.000000000","tz":180},"subject":"Reduce NetApp unit test runtime by mocking time.sleep","message":"Reduce NetApp unit test runtime by mocking time.sleep\n\nSome NetApp driver unit tests take excessively long to run\n(nearly one minute)\n\nThis change introduces a mock for time.sleep in the affected\ntest classes to eliminate real waiting time without altering\ntest logic.\n\nThese are:\n- NetAppRestCModeClientTestCase(test_client_cmode_rest)\n- NetAppClientCmodeTestCase(test_client_cmode)\n- NetAppCDOTDataMotionSessionTestCase(test_data_motion)\n- NetAppFileStorageLibraryTestCase(test_lib_base)\n- NetAppFileStorageLibraryTestCase(test_ multi_svm)\n\nThis reduces overall test runtime.\n\nTest Plan:\n- Verified all NetApp unit tests pass with no timing delays.\n- Confirmed strestr slowest no longer lists test exceeding 1s.\nOn testing, the slowest test took  0.359seconds\n\nCloses-Bug: #2031194\nChange-Id: Ic223080867e1e12b0e2f1c4ed02f1c92ef5b50e0\nSigned-off-by: Rose Kimondo \u003crosewairimu1@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/636d514cdba8644b1849f071107996c30dc280d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/636d514cdba8644b1849f071107996c30dc280d7"}]},"branch":"refs/heads/master"},"0040fa562eabb876d5068656b42fa4ff021a31f5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-10-27 20:53:35.000000000","uploader":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"ref":"refs/changes/49/964749/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/964749/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/964749/3"}}},"commit":{"parents":[{"commit":"1529ced1a7ce90f45fc6143db78619e4704ece78","subject":"fix netapp delete_share private_storage cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1529ced1a7ce90f45fc6143db78619e4704ece78"}]}],"author":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-24 05:52:33.000000000","tz":180},"committer":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-27 20:53:35.000000000","tz":0},"subject":"Reduce NetApp unit test runtime: time.sleep","message":"Reduce NetApp unit test runtime: time.sleep\n\nSome NetApp driver unit tests take excessively long to run\n(nearly one minute)\n\nThis change introduces a mock for time.sleep in the affected\ntest classes to eliminate real waiting time without altering\ntest logic.\n\nThese are:\n- NetAppRestCModeClientTestCase(test_client_cmode_rest)\n- NetAppClientCmodeTestCase(test_client_cmode)\n- NetAppCDOTDataMotionSessionTestCase(test_data_motion)\n- NetAppFileStorageLibraryTestCase(test_lib_base)\n- NetAppFileStorageLibraryTestCase(test_ multi_svm)\n\nThis reduces overall test runtime.\n\nTest Plan:\n- Verified all NetApp unit tests pass with no timing delays.\n- Confirmed strestr slowest no longer lists test exceeding 1s.\nOn testing, the slowest test took  0.359seconds\n\nCloses-Bug: #2031194\nChange-Id: Ic223080867e1e12b0e2f1c4ed02f1c92ef5b50e0\nSigned-off-by: Rose Kimondo \u003crosewairimu1@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0040fa562eabb876d5068656b42fa4ff021a31f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0040fa562eabb876d5068656b42fa4ff021a31f5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8c5874ac77130e303b4e7e3622727d17042fe33f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-11-13 17:26:21.000000000","uploader":{"_account_id":38469,"name":"Rose Kimondo","display_name":"Rose Kimondo","email":"rosewairimu1@gmail.com","username":"rosekimondo"},"real_uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/964749/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/964749/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/964749/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/964749/4"}}},"commit":{"parents":[{"commit":"3946294f2ba884e80b734e40dc20f235862b3f4b","subject":"Merge \"api: Combine versioned API resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3946294f2ba884e80b734e40dc20f235862b3f4b"}]}],"author":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-10-24 05:52:33.000000000","tz":180},"committer":{"name":"Rose Kimondo","email":"rosewairimu1@gmail.com","date":"2025-11-13 17:26:21.000000000","tz":0},"subject":"Reduce NetApp unit test runtime: time.sleep","message":"Reduce NetApp unit test runtime: time.sleep\n\nSome NetApp driver unit tests take excessively long to run\n(nearly one minute)\n\nThis change introduces a mock for time.sleep in the affected\ntest classes to eliminate real waiting time without altering\ntest logic.\n\nThese are:\n- NetAppRestCModeClientTestCase(test_client_cmode_rest)\n- NetAppClientCmodeTestCase(test_client_cmode)\n- NetAppCDOTDataMotionSessionTestCase(test_data_motion)\n- NetAppFileStorageLibraryTestCase(test_lib_base)\n- NetAppFileStorageLibraryTestCase(test_ multi_svm)\n\nThis reduces overall test runtime.\n\nTest Plan:\n- Verified all NetApp unit tests pass with no timing delays.\n- Confirmed strestr slowest no longer lists test exceeding 1s.\nOn testing, the slowest test took  0.359seconds\n\nCloses-Bug: #2031194\nChange-Id: Ic223080867e1e12b0e2f1c4ed02f1c92ef5b50e0\nSigned-off-by: Rose Kimondo \u003crosewairimu1@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8c5874ac77130e303b4e7e3622727d17042fe33f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8c5874ac77130e303b4e7e3622727d17042fe33f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
