)]}'
{"id":"openstack%2Fmanila~964751","triplet_id":"openstack%2Fmanila~master~I71fe54ffd8294b75b952a41c2bdc6defde6dbd65","project":"openstack/manila","branch":"master","topic":"bug/2129736","attention_set":{},"removed_from_attention_set":{"36741":{"account":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"last_update":"2025-10-27 15:11:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71fe54ffd8294b75b952a41c2bdc6defde6dbd65","subject":"[doc] Add \"mount_point_name\" request parameter in the POST share API","status":"MERGED","created":"2025-10-24 08:22:11.000000000","updated":"2025-10-27 15:12:56.000000000","submitted":"2025-10-27 15:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964751-bug/2129736","meta_rev_id":"04adf78aa1e680d3866e760c86c4bd752756f661","_number":964751,"virtual_id_number":964751,"owner":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 15:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-10-25 23:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-10-27 14:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-24 08:39:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-25 23:14:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"7773309b838ec5132ed4f2a1f832fd88b937cdbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2025-10-24 08:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddc1998154acdfc5d4ad7eeabeecd83977ed1bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 08:39:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a835080b62f4e4c8205f515a9d24e31\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8fac0329bd6c4f61907a740f915e99af : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c2db673f88448fbbb1a1258754efe8 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b96d276c1a403cb9d7f7556067fba8 : SUCCESS in 2m 55s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/73273b6f351c431a94590bdc9f5639c1 : SUCCESS in 4m 45s\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e1d44948a8bd40b0a9accfa52387bf7c : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da31f007ad04e21687661d23b99cd5628f258b19","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-25 23:14:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbaedf239a857fa7af19b295d94639a48ba6b60c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-27 14:54:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca4c6e1de0db46f1040ef8fc06a03f8815ff005b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 14:54:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fac7d7a8c9d9939fd3f6696bb58276ef69da9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:11:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f808cce45e4e2d8584867a9083d84e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f2a4f1dc08b64db3b69be69811cd0c8b : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c122f05d82444dda9e15d27446b51609 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e53bccd33b34bc09ffb99cf4670adfb : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"007dd65947738cac39f801c58349ffdc3c185edb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:11:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"04adf78aa1e680d3866e760c86c4bd752756f661","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:12:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c492b593d254a04bb8da4c76497675d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/07445e383bbd47bf88737e5a0676d2ad : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91592a4b95614267a0b800df0344eb59 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea32b47ff40a6c927cfb08ee299d53a353c5935f","revisions":{"ea32b47ff40a6c927cfb08ee299d53a353c5935f":{"kind":"REWORK","_number":1,"created":"2025-10-24 08:22:11.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/51/964751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/51/964751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/964751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/964751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/964751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/51/964751/1"}}},"commit":{"parents":[{"commit":"8ae12adc99ffd6977036a476f032d4957653bbc2","subject":"Merge \"Doc for deferred share/snapshot deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8ae12adc99ffd6977036a476f032d4957653bbc2"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-10-24 08:20:35.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-10-24 08:20:35.000000000","tz":540},"subject":"[doc] Add \"mount_point_name\" request parameter in the POST share API","message":"[doc] Add \"mount_point_name\" request parameter in the POST share API\n\nAdd \"mount_point_name\" parameter to the POST share API\u0027s request body\n\nCloses-Bug: 2129736\nChange-Id: I71fe54ffd8294b75b952a41c2bdc6defde6dbd65\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ea32b47ff40a6c927cfb08ee299d53a353c5935f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ea32b47ff40a6c927cfb08ee299d53a353c5935f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
