)]}'
{"id":"openstack%2Fmanila~968640","triplet_id":"openstack%2Fmanila~master~I7a2e5ab7ebe44f451e774b58d26c1bc92fce43e3","project":"openstack/manila","branch":"master","topic":"hpealletrampb10000driver","hashtags":[],"change_id":"I7a2e5ab7ebe44f451e774b58d26c1bc92fce43e3","subject":"[WIP] Add share driver for HPE Alletra MP B10000","status":"NEW","created":"2025-11-27 12:29:31.000000000","updated":"2025-11-30 15:52:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"e53ea70b7c58dd6fc1966d16b735f657a6ad8728","_number":968640,"virtual_id_number":968640,"owner":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-30 15:52:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-11-27 14:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c112b6eaf331972cb8edaa8e30ef56238cfbfed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2025-11-27 12:29:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"310882d6b146507ac8e0480b854d88ed74987a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2025-11-27 12:33:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c050334ad467f25e41590724f3cf4d3a54dfaf89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2025-11-27 12:34:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a415838c70e13ee455199fff07113baad0ea9c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2025-11-27 12:34:37.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f1bd26433513f3d1cda33961502a40cab59a24be","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2025-11-27 12:39:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"48f83fcbd1bc6c02bdaafeb204c911e2300e290a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 14:21:45.000000000","message":"Patch Set 4: Verified-1\n\n(772 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b3286459b4c400992b46af1ef9ac1e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489e089123134ac5b06dde30b8b3583c : FAILURE in 7m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ce094f905414a7a972e182758cad22b : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5caae9080d14ef9aca5788f53a40b6b : FAILURE in 7m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c91035ab7a0440859f51b53e2ab849cf : FAILURE in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fa0dfd0f7e44f2386bbcb7219226d84 : FAILURE in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/476155c02d554667bd79916840e68a02 : SUCCESS in 6m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9b33aa7d87f74885bd167257f373a354 : SUCCESS in 1h 16m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3a4278f8e48e4b9e884508b47c7cb467 : SUCCESS in 1h 39m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f66374cd2c484fb4b28aae7e13ea40e0 : SUCCESS in 1h 37m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/390cc56301a546abbd53f316cf42b3be : SUCCESS in 1h 09m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f0ae110a83054c6ebf27c97332ebd8a3 : SUCCESS in 45m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fa966d52e16c4a7fa0e44253e97a9ba5 : FAILURE in 21m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8328bfeb91e04e4b83da3687c94807dd : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40708585e0884da0817701982b8123cf : FAILURE in 5m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/067e0e8106f5459d907e9e759eec13d0 : SUCCESS in 29m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f9d9335247ce4612bb940b8c7406676d : SUCCESS in 28m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2141e4d4a0a74c6eabd20ff6bc972642 : FAILURE in 14m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/1959561ad0ac451f861d21111b109e7b : SUCCESS in 1h 20m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/07438e1dfa084638b9b2c797ff6a9875 : SUCCESS in 41m 24s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1657ad94ca640018a579c399a85936d : SUCCESS in 42m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a2edd6fd275c49228395a10ea6bab564 : FAILURE in 11m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e69a14ac491746c28ebfbab1205285f0 : FAILURE in 5m 40s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/1ac9214fb1fb43389c93e8cf019d7f2d : FAILURE in 4m 38s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c5214716c4774aa4865837e4c7680b13 : FAILURE in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9911066b0c20abf9d1ae86c14a1876884ca1669e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2025-11-30 13:12:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e53ea70b7c58dd6fc1966d16b735f657a6ad8728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-30 15:52:27.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c6a40035b3449f6ae123c2adaa747e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1199e0e475405aa4cde5933d448f6b : FAILURE in 9m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aac1e72dc8e9466f835f55cd935ea314 : FAILURE in 17m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b82760dc542474dba88337ece555ac3 : FAILURE in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70f68c88c84b4be5b26de59f84506b9c : SUCCESS in 8m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84afd99141694a6380b335db9267fb89 : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a88d1a1614054a4eba6c265e427544f9 : SUCCESS in 9m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7712639e09434835963c4554c59a861b : SUCCESS in 33m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4082296a8a4e4c1eb7d2471f753410af : FAILURE in 2h 34m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0ce85e77007645048fd853b70c597ff8 : FAILURE in 53m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f9b7c79a65ff41bcbddddff230090961 : SUCCESS in 1h 07m 20s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a5a0de3f29a14700b330e6ba6bb05f5f : SUCCESS in 49m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/045796cf0370492680a14d7391c27f7e : FAILURE in 10m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d8b40fd3e82e4a4798158562b3989da9 : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c5257da3f5047a8af8fda3afa5a6217 : FAILURE in 2m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a3da8fd16c554b2d90c4af4184a4aea9 : SUCCESS in 32m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c3bd886ec5964a5082644287802bde90 : SUCCESS in 44m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/66c532dcc6044d1cb27d8121a4b2845f : FAILURE in 23m 51s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/157b9a458d024c91af64b2c613868c24 : SUCCESS in 58m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fc72ffbfb67e4173aa23ef4f27a58ac9 : SUCCESS in 40m 05s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e527fcca1e744e3db63fe765ccda51ee : SUCCESS in 52m 35s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c8f71c8725c9481f83b62904cfb3f395 : FAILURE in 8m 09s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/07d235a408d54c02851dd3ee739b1471 : FAILURE in 11m 09s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/2e8d8fba11e846c3b290bfe72d680426 : FAILURE in 4m 47s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0fc6454612b44fce9539d929f93f099f : FAILURE in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"71161b2576c86a8210219656918a16a489a0e5cd","revisions":{"bd9274812217e2a844ae1e861f2d39a026c885a6":{"kind":"REWORK","_number":1,"created":"2025-11-27 12:29:31.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/40/968640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/40/968640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/40/968640/1"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:27:10.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:29:02.000000000","tz":330},"subject":"Add share driver for HPE Alletra MP B10000","message":"Add share driver for HPE Alletra MP B10000\n\nChange-Id: I7a2e5ab7ebe44f451e774b58d26c1bc92fce43e3\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bd9274812217e2a844ae1e861f2d39a026c885a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bd9274812217e2a844ae1e861f2d39a026c885a6"}]},"branch":"refs/heads/master"},"00716dfe588cc3b8d93d8ca74e9b48473dce73c0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-27 12:33:14.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/40/968640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/40/968640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/40/968640/2"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:27:10.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:32:56.000000000","tz":0},"subject":"WIP: Add share driver for HPE Alletra MP B10000","message":"WIP: Add share driver for HPE Alletra MP B10000\n\nChange-Id: I7a2e5ab7ebe44f451e774b58d26c1bc92fce43e3\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00716dfe588cc3b8d93d8ca74e9b48473dce73c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00716dfe588cc3b8d93d8ca74e9b48473dce73c0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"418ec34aa701ff29d7553ee58be48dd45165272b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-27 12:34:30.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/40/968640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/40/968640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/40/968640/3"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:27:10.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:34:30.000000000","tz":0},"subject":"[WIP]Add share driver for HPE Alletra MP B10000","message":"[WIP]Add share driver for HPE Alletra MP B10000\n\nChange-Id: I7a2e5ab7ebe44f451e774b58d26c1bc92fce43e3\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/418ec34aa701ff29d7553ee58be48dd45165272b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/418ec34aa701ff29d7553ee58be48dd45165272b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"16c94e1f3c22e2e16b91a28867deaf6eec3ace9a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-11-27 12:34:37.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/40/968640/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/40/968640/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/40/968640/4"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:27:10.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:34:37.000000000","tz":0},"subject":"[WIP] Add share driver for HPE Alletra MP B10000","message":"[WIP] Add share driver for HPE Alletra MP B10000\n\nChange-Id: I7a2e5ab7ebe44f451e774b58d26c1bc92fce43e3\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/16c94e1f3c22e2e16b91a28867deaf6eec3ace9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/16c94e1f3c22e2e16b91a28867deaf6eec3ace9a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"71161b2576c86a8210219656918a16a489a0e5cd":{"kind":"REWORK","_number":5,"created":"2025-11-30 13:12:34.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/40/968640/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/40/968640/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/968640/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/40/968640/5"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-27 12:27:10.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2025-11-30 13:05:35.000000000","tz":330},"subject":"[WIP] Add share driver for HPE Alletra MP B10000","message":"[WIP] Add share driver for HPE Alletra MP B10000\n\nChange-Id: I7a2e5ab7ebe44f451e774b58d26c1bc92fce43e3\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/71161b2576c86a8210219656918a16a489a0e5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/71161b2576c86a8210219656918a16a489a0e5cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
