)]}'
{"id":"openstack%2Fmanila~970097","triplet_id":"openstack%2Fmanila~master~Id6414e6b4c32129a8154558bf72c75ccc20c5315","project":"openstack/manila","branch":"master","topic":"bug/2133203","attention_set":{"38373":{"account":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"last_update":"2026-03-05 14:55:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2025-12-09 11:57:51.000000000","reason":"\u003cGERRIT_ACCOUNT_20733\u003e replied on the change","reason_account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2025-12-24 08:05:19.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-05 14:55:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-02 18:12:09.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"Id6414e6b4c32129a8154558bf72c75ccc20c5315","subject":"Fix NFS export IP prefix matching in access rule check","status":"NEW","created":"2025-12-08 06:30:50.000000000","updated":"2026-03-05 14:55:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b34b42c31c5b3ebfa2612e204219dee7498280e1","_number":970097,"virtual_id_number":970097,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-24 10:28:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":-1,"date":"2026-03-02 18:12:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"value":-1,"date":"2026-03-03 06:13:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-08 08:39:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 10:18:54.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-12-09 11:57:51.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2025-12-09 11:57:51.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-12-09 11:57:51.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-12-24 08:05:19.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-03-02 15:52:01.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2026-03-05 14:55:08.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"state":"REVIEWER"}],"messages":[{"id":"f12f19c28ab72ec4842057bc4d53571f03f16a59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2025-12-08 06:30:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57542dd46098f73e1a17db8f829d4bc3803afa75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 08:39:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e22b504108d84077b53d2b9a81ea3eac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b46f4a204742799f0d6954257194f9 : SUCCESS in 12m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8436b579be7441e48df318cffdbdce77 : SUCCESS in 18m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1094aa5f1b9944d598679c22b773bcb3 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9838c9324c5b4a07b8104195accb1873 : SUCCESS in 8m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/472671a8a6c54b898b6f8a9877a72d68 : SUCCESS in 6m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b5bf9c3c369a464dbcd61fc23e39055c : SUCCESS in 1h 13m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4b88c4b5640a4a96903809983028895f : SUCCESS in 2h 05m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/846cf0dc58cd416cab798e3b44ab6f1d : FAILURE in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/48097c375ecf4a17b41084a0a21a9849 : SUCCESS in 49m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/21dc980705384f5291bb052ad202ee36 : SUCCESS in 1h 29m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a4e220c2d91e492187493d287b08b6f2 : FAILURE in 25m 58s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/42a13167f9b14796a0002bbd34fdf16e : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b4d3b054bcbb4e3fa38ee2252e8491b5 : SUCCESS in 2m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ffe1c5eb20024683a6bf2273d00274a0 : SUCCESS in 25m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b8f2a5f795994e5f8b7617e92b5a3ee5 : SUCCESS in 32m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6940728c5b5049a1939acea56a7913b5 : FAILURE in 28m 21s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/5b47e505a3374e30bf15e7c7064256dc : SUCCESS in 32m 47s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/731e154e82ca415e9ad6808710f046c5 : SUCCESS in 41m 08s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f16fff6e67e4bc9a2f43fd6b6213b45 : SUCCESS in 1h 08m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/76a7fb4e8ac34d14a096f0fa6fbeaae9 : SUCCESS in 36m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c44b5bbc3ef04bde816bb4d610ed3566 : SUCCESS in 32m 19s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/9c13d33e261441a0b23f5568a8b5d43d : FAILURE in 2m 24s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/baeac405824a4138b8a12ba7a0bc95de : FAILURE in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40420d7ac684326bb89bd886048bd8471b6dd4a9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-12-08 10:18:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/97/970097/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/1159de0/ : SUCCESS in 1h 37m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"518d40415910d5c176e516bc581b90d619676db0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-19 18:29:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50977f73b781dbbc8992f180df07597e54719624","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-19 18:29:59.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"41fc706f2a4993c8e946662dd856ed8ae23d2370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2025-12-24 08:03:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3da3c790bee266a4398d269362f308fe36535a5b","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2025-12-24 08:05:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-alletramp-manila-file-driver build/5f41eb203d7e4670a6ce755522cf8e61 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2},{"id":"eb03f4701517f2e33e8e2307aa1ff18eb9fc40df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 10:28:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d348827417dc4e06b9594840f5278d17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034e67a2dfb14071a785cee79500dfd8 : SUCCESS in 10m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8405dab662a04a8fb67cd22f8426db6b : SUCCESS in 34m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715282b53e154541b0a6f215d29cdd62 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8ff69e7264945e7a5aae8c7a61cd0b8 : SUCCESS in 9m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f62ee7ac3edb4a57940a1ec4f6a437de : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78a50c90cc414efe862f04ecab0fa41c : SUCCESS in 8m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d5e4c9056ca44ff59b2f90d3d6ecf5d6 : SUCCESS in 1h 17m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7f909dedcad040d1935d82ec290c2014 : SUCCESS in 2h 10m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5f8319222b5147489e38c41924e07bcf : FAILURE in 1h 07m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/66885c9ee8da4cbb98ade74ba4635450 : SUCCESS in 1h 00m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1881ba23761c44d392fa3e143e779eec : SUCCESS in 1h 07m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/989b0306b6da41f4af5b363a950b39a8 : FAILURE in 23m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/484084db6be84f2ea004d7303dcd9cfa : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04d47eb981494d508bce47c5bbebdcd4 : SUCCESS in 3m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6821eb043dab42a4b4667562cef389ae : SUCCESS in 29m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/23c1424625f144c689c9242dc315c40a : SUCCESS in 24m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b9c88e828b094e14a1f1ca5804e22eec : FAILURE in 11m 13s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/91d895667f6c4e02ab3a6ee3ca7b1545 : SUCCESS in 1h 20m 31s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/56df6c058b3549ff926950fbec9f0dd9 : SUCCESS in 57m 22s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c4e9e84e782435b9b3b597241cc8070 : SUCCESS in 56m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2f02ff5fec2140c2b5fe5752ecd28c90 : SUCCESS in 32m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/297f6554f97e4ad98346be7131c8a276 : SUCCESS in 35m 20s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c7e94727b8c34c6b8e8aa327ddc5ef05 : FAILURE in 4m 23s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ea284ba1dcfd4b0daed0f171707365e9 : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0415b3d5ea326b1c0a79fb077302228b07343ada","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-12-24 11:54:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/97/970097/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/8cbe81e/ : SUCCESS in 1h 25m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e11466f08cb8204a6297ae3dbee4c4e7ce8aaed2","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-02 15:52:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dd70a07ba3f426d0c9c22cf808616c1d60c88ead","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-02 18:12:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7db799240da32f7f45cadf688f7f46f4611cd43e","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-03 06:13:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"976b4c81914190180cc4d52b38cd8e4a55716ddc","revisions":{"0d40d3720adb5438e100725075a242552fb255a5":{"kind":"REWORK","_number":1,"created":"2025-12-08 06:30:50.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/97/970097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/970097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/970097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/970097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/970097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/970097/1"}}},"commit":{"parents":[{"commit":"e8c8380777e609b0ba2dedb89ed7c8fef7afc2a0","subject":"Merge \"api: Add schemas for share soft_delete, restore actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e8c8380777e609b0ba2dedb89ed7c8fef7afc2a0"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2025-12-08 06:13:38.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2025-12-08 06:30:10.000000000","tz":330},"subject":"Fix NFS export IP prefix matching in access rule check","message":"Fix NFS export IP prefix matching in access rule check\n\nThe regex in NFSHelper.update_access() incorrectly matches IP\nprefixes when checking if an access rule already exists. For\nexample, if the export list contains 192.168.1.50, attempting\nto add 192.168.1.5 would be incorrectly reported as already\npresent because the pattern did not assert a non-digit boundary\nafter the IP address.\n\nThis patch adds a negative lookahead (?!\\d) to the regex to ensure\nthe matched IP is not followed by another digit, so only exact IP\naddresses match.\n\nCloses-Bug: #2133203\nChange-Id: Id6414e6b4c32129a8154558bf72c75ccc20c5315\nSigned-off-by: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d40d3720adb5438e100725075a242552fb255a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d40d3720adb5438e100725075a242552fb255a5"}]},"branch":"refs/heads/master"},"976b4c81914190180cc4d52b38cd8e4a55716ddc":{"kind":"REWORK","_number":2,"created":"2025-12-24 08:03:32.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/97/970097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/970097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/970097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/970097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/970097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/970097/2"}}},"commit":{"parents":[{"commit":"cef2bd8af9ffec61b3869b03ffb52a6390d19b5f","subject":"Fix schema for all_tenants query parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cef2bd8af9ffec61b3869b03ffb52a6390d19b5f"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2025-12-08 06:13:38.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2025-12-24 08:03:19.000000000","tz":330},"subject":"Fix NFS export IP prefix matching in access rule check","message":"Fix NFS export IP prefix matching in access rule check\n\nThe regex in NFSHelper.update_access() incorrectly matches IP\nprefixes when checking if an access rule already exists. For\nexample, if the export list contains 192.168.1.50, attempting\nto add 192.168.1.5 would be incorrectly reported as already\npresent because the pattern did not assert a non-digit boundary\nafter the IP address.\n\nThis patch adds a negative lookahead (?!\\d) to the regex to ensure\nthe matched IP is not followed by another digit, so only exact IP\naddresses match.\n\nCloses-Bug: #2133203\nChange-Id: Id6414e6b4c32129a8154558bf72c75ccc20c5315\nSigned-off-by: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/976b4c81914190180cc4d52b38cd8e4a55716ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/976b4c81914190180cc4d52b38cd8e4a55716ddc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
