)]}'
{"id":"openstack%2Fmanila~970296","triplet_id":"openstack%2Fmanila~stable%2F2025.2~I33275c6eb5b137d78b10a9942ac4f765d011eac1","project":"openstack/manila","branch":"stable/2025.2","topic":"fix-rally-jobs-stable/2025.2","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-25 08:36:14.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-25 07:06:53.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"I33275c6eb5b137d78b10a9942ac4f765d011eac1","subject":"Fix rally jobs","status":"MERGED","created":"2025-12-09 19:04:39.000000000","updated":"2026-02-25 08:37:48.000000000","submitted":"2026-02-25 08:36:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970296-fix-rally-jobs-stable/2025.2","cherry_pick_of_change":969594,"cherry_pick_of_patch_set":5,"meta_rev_id":"018c98df741c1756f1a4e20d672126d76aa28b42","_number":970296,"virtual_id_number":970296,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-25 08:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-02-25 07:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-24 21:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-02-25 07:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-09 21:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 21:42:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-24 21:42:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"1d63fded8fe73dfd17c18f8e2885b03fb33e1acf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-09 19:04:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"93517a0826e12c75da1a373bb1b0c9ba4904f903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 21:07:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/156719a9788b4535a46abb53ade3dc0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a99af72d234a929a9a461aa9dbe264 : SUCCESS in 9m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9573d554edb14c29907fe0912b4f99bf : SUCCESS in 20m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c65d5a0f6b45c1aaa1ec8f592275db : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55a5daa6be374db9b9ffec728b4740f7 : SUCCESS in 9m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45a61c443557440e84b9514b52864522 : SUCCESS in 7m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a6dc3e0aab5437dbaf1c92594793e8a : SUCCESS in 7m 41s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9b8090daae3e4fdeabd35684d183ed7f : SUCCESS in 1h 24m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/82b0243ad95847ce80e9fc6a9a748c79 : SUCCESS in 2h 00m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7e1afd96c8494d8a89fcb7e0d5f64dca : FAILURE in 1h 12m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/45d8fe79a0224b6f833238d19b97fbca : SUCCESS in 1h 03m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d0649469f3af473abf52924815acb823 : SUCCESS in 1h 27m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/87b61a7465844e9da374e30912eaabf4 : FAILURE in 21m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/85cc2f3000684b2785ec667623668c22 : SUCCESS in 2m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/309689eb318e4bc1b6c1e409cc3e763d : SUCCESS in 4m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3031c158e41544d7971c56e76d5fd4b1 : SUCCESS in 24m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c4c0c517508b4ba7a8fd00153dfc1dd9 : SUCCESS in 37m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c8da5f1e99874df38ccbb48c7cf30d1b : FAILURE in 11m 41s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/34e5b6dcb06e4897ae7268f028b65f15 : FAILURE in 18m 53s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/beed19633c07471aac99aea5ff0effe0 : FAILURE in 8m 02s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e29020e71f454529a44984c6b7a168a0 : SUCCESS in 58m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2dbb0f3cd5a24423a54be454da521063 : SUCCESS in 30m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d60b8f48c6be4ebb8d7ca15487553ce7 : SUCCESS in 35m 48s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/7b7560ca0fb441928013889c5d8ee6cf : FAILURE in 5m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/298b331b49484ff2b08a841dca342003 : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c42a4c083bdf80d8d7b270adec88fded8c9e760","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-03 20:45:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f29f7e5390da0e58e843108ed542118d15cb331f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 22:36:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0185123a415c41a5b4e02761adb9ca37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e37273ccdaa64876a09312db56abaf9c : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f282d5b1b0f84542b0eb128c3b3b3eb6 : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce48ec3b1da14e76b84236e90733874a : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f542d1c35c924fadb29d80345e7ca081 : SUCCESS in 7m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d98dfac12b544bcdb5434d2c12f2e3ab : SUCCESS in 10m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09b6b5f1e04241db86fca0d0c4b6dcb6 : SUCCESS in 13m 55s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4075791221d54a54aaa80fbd062b1176 : SUCCESS in 32m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a0d13c0591504fdd9d845a57bdb4ca0d : SUCCESS in 1h 27m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c04847e7881448c861cd1169872d96d : SUCCESS in 1h 44m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9b919fa83da94be8bc5460916d88bc59 : SUCCESS in 55m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/483c3f8ccc6f49b68ba24ee1454f0bf8 : SUCCESS in 1h 00m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/eade3d83562b47119c497e72e56ce106 : FAILURE in 16m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fc4a98f18f654bc391f22c2501c4ba1a : SUCCESS in 4m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f35bfcd2dfd14bf8a85edeb8d228d7f0 : SUCCESS in 4m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c75100e3f4664a938b22ade0caffa9c0 : SUCCESS in 36m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4db015199ca34f9e80aeed0122fc9fcf : SUCCESS in 44m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d114aeebef19411db7370dfa3ede34e9 : FAILURE in 57m 26s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/f77624844086408dabd33aa8e2150898 : FAILURE in 18m 27s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/48d78b1d43994bba8ad3d616ba92e95b : SUCCESS in 39m 58s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58c8b894de914090b6913af7655a3e64 : SUCCESS in 1h 04m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7677bf5bded74ad79acf20fd7714201c : SUCCESS in 31m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/19cab953648640fa92445420fc353901 : SUCCESS in 32m 52s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/d7162a85dc734368a90721d9811ccb59 : FAILURE in 4m 55s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f9110954b9df4f449fa6c47564413a29 : SUCCESS in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47505d1211f2b185a05a342ee24017062ad8a9e3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-24 21:42:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1c9b576ce108feb40a2af9d17b06fdd1b77f229","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-25 07:06:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d838a4ee4dc5adf68b620f7c7772fdae0a3f7d91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 07:07:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cee37e6e30be5f98c3dbbc7047f6d658151326f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 08:36:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2f155f244974f11ad24d6f61679171f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f862a67e62244379692890f44e02f5c : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205a0f2f5e4f4e5b9ee4514e51317ac1 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/414eaa297dbf4f4eb40279cec7eb4ea4 : SUCCESS in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee74b8b424e944a194f4815e34ba62f8 : SUCCESS in 10m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/96e979b490e6489994b390b44de4c7a8 : SUCCESS in 1h 23m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9a63744c542e4fa9bae297db1330b0ab : SUCCESS in 25m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/190f153ecd404291b87b3bee5e257c1d : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9dce10af77e15b908be36e9510aa818d97acbfa3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 08:36:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"018c98df741c1756f1a4e20d672126d76aa28b42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 08:37:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd4124b9526d4c30b6ffc2bbacd2b468\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617c6d34ec3a4e0bba34f4c08e5e67a4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2780d3a53fe94857376833a7c0781095f75df70e","revisions":{"2780d3a53fe94857376833a7c0781095f75df70e":{"kind":"REWORK","_number":1,"created":"2025-12-09 19:04:39.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/96/970296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/96/970296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/96/970296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/96/970296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/96/970296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/96/970296/1"}}},"commit":{"parents":[{"commit":"d9638f279012e6a8c6262593e5f6974695339f95","subject":"Fix shares getting stuck in ensuring status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d9638f279012e6a8c6262593e5f6974695339f95"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-12-04 10:45:20.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2025-12-09 19:04:39.000000000","tz":0},"subject":"Fix rally jobs","message":"Fix rally jobs\n\nPreviously, we forced rally jobs to do not use the global venv when\nit was introduced a couple of releases ago, as it caused a breakage.\n\nLater on, we started seeing a failure in the rally jobs, where\nos-testr installation/upgrade failed, and the reason was that we\nwere attempting to upgrade a package that was externally-managed,\nindicating that this installation/upgrade should be within the\nvenv.\n\nThis change switches the rally jobs to using the global venv,\nas the issues we saw back then were fixed.\n\nChange-Id: I33275c6eb5b137d78b10a9942ac4f765d011eac1\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n(cherry picked from commit 75e895d5722e4434d7cc0d12996f9f68902c1539)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2780d3a53fe94857376833a7c0781095f75df70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2780d3a53fe94857376833a7c0781095f75df70e"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
