)]}'
{"id":"openstack%2Fmanila~970301","triplet_id":"openstack%2Fmanila~stable%2F2025.1~I33275c6eb5b137d78b10a9942ac4f765d011eac1","project":"openstack/manila","branch":"stable/2025.1","topic":"fix-rally-jobs-stable/2025.2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-12 09:06:37.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-11 21:02:51.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"I33275c6eb5b137d78b10a9942ac4f765d011eac1","subject":"Fix rally jobs","status":"MERGED","created":"2025-12-09 20:03:42.000000000","updated":"2026-03-12 09:20:23.000000000","submitted":"2026-03-12 09:06:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970301-fix-rally-jobs-stable/2025.2-stable/2025.1","cherry_pick_of_change":970296,"cherry_pick_of_patch_set":1,"meta_rev_id":"64538ce471759415d1cc0d3976e01604d09c1ed0","_number":970301,"virtual_id_number":970301,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-12 09:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-03-11 21:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-10 14:51:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-03-11 21:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-09 21:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 21:43:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-10 14:27:32.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-03-10 14:51:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-10 14:51:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-10 16:18:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-11 21:03:07.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"c968a73c87b569848bcb97ff488697ab21c1bb0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-12-09 20:03:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"a2c9b6250a38ee0cf8cc943afae1fe86c67bd430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 21:33:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6d73d74021649959af45af97277dd2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5cf4ecff3848968f23ce3a50091542 : SUCCESS in 12m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00794ef57c834693b3ae4ace045659ed : SUCCESS in 34m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e44831e6ce2e4f1b90c3b0df585aa905 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecb30667be814a6981cf105e5f4d2dce : SUCCESS in 10m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a9f8f48f6dd4c69beeffed69e6d346e : SUCCESS in 7m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8c34fb310cf34c59806f472717b4db1c : SUCCESS in 1h 22m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/21e5a0392b1142dd9721a00ed12ddeac : SUCCESS in 1h 05m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e0b08c196ca748fb91231409930dc442 : FAILURE in 38m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b26e8a828a41405e923d1b2955282334 : SUCCESS in 1h 06m 54s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a5cd9ef5362e44528c87b0a1e3574d6d : FAILURE in 1h 14m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a053719172b04d6ab912c0c048e42359 : SUCCESS in 38m 31s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2b7b84dcb86f43acb13275c37ca987d9 : SUCCESS in 4m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/247ec9088e0f4b2f8347d4fa578c43f2 : SUCCESS in 3m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dca7b17946be49af82b5b158620a38cd : FAILURE in 6m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fbd357d221ee4e9bbe60d4806ccb2130 : SUCCESS in 37m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/474b0fbacf324e5ab6a9a7957297e558 : FAILURE in 30m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9a9ba2dc3bba4e5eac9d32fce06eb45a : FAILURE in 23m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2da7e9dddcee4477b9109eb2e7aa87ab : SUCCESS in 59m 19s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f337f2c85e340fb81114916f5c8b20a : SUCCESS in 45m 28s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/644937c1c10b427ca2d96e0264b4a3b8 : SUCCESS in 28m 22s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a984c7a2392f483f93f5fe1bb793f703 : SUCCESS in 16m 42s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/965aeed63f914d768a14da9d4fb2f938 : FAILURE in 2m 41s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/9438f428e5b94db2a572f099ed4afd35 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64ba762d9ce7fef8943377774b7f13253c03dcd9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-24 21:43:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ce5ed656ff6347aa4fffef61c66916e67279db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 23:40:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69a89a2ddc294c29b0802f08ea3ead75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec576be4c8d4ccb9c69335cc7f543ec : FAILURE in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98ca096f493c40b9b46e0bded7bae469 : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a825a279c5a049ae9540ecfc8659c42b : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed1156d3affc45d6b972076b3281f0db : SUCCESS in 12m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89f2d9af6f534c2e81b72636fdd882ca : SUCCESS in 10m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/498f4caae35c4b55bcfbd1bf984c7d38 : SUCCESS in 1h 11m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/99b784d3f7144abba050994922a1c912 : FAILURE in 58m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1a5996668a1c451a87664c47ba7a8281 : FAILURE in 1h 15m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d70eb0e18f2946f4b8e569911ad53394 : FAILURE in 1h 09m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9e909345690f43d39414f62c2adc066f : FAILURE in 1h 44m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3ee351d1d79e4f07bc5a3efb8b687c9e : FAILURE in 22m 04s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c5543ca8eb2c4fc9b8261b46b789ae1e : SUCCESS in 4m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f789b68d4aa465dbb949047a96d9d40 : SUCCESS in 5m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/edebc42a6e854fcdb7324d60136f964a : FAILURE in 8m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ce62fc8fa74d4e54bee865111230975e : SUCCESS in 44m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0592e17196c647288b9be35b9c99c878 : FAILURE in 31m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2761a5ca1f9d421d95e5aec5c33807c3 : FAILURE in 38m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3d8abc28b21c4594a08fbefe5f34e729 : SUCCESS in 59m 29s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91917ed8df1643b8aeeb342a90fc468f : SUCCESS in 59m 28s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4e2ec5c2e1214560b6dd13fdaefbeafe : SUCCESS in 32m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1d37d0d15a4a46e8a636baffdc0b8553 : SUCCESS in 31m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/fe5fd3582e5848aea1a96e46178d8af8 : FAILURE in 4m 43s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/32f463f5f30a4a51b2daa556e6a9f321 : RETRY_LIMIT in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7b659c2508ad758a3a42bf2207550b4ebd7d0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-10 12:50:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29632\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"_revision_number":2},{"id":"7977e61e12507c613797af26f538b28fe4aab4b1","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-10 14:27:32.000000000","message":"Patch Set 2:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/970301","accounts_in_message":[],"_revision_number":2},{"id":"a0191fe2b55d9d131ec3a80a7e19e8be1957f669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 14:51:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0198445db3e84bbcb2d513d4ed9ade27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45a1e8505e7409a824be3787fe95d1b : SUCCESS in 12m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33b732a6336e46bbab0971a8f34bc69e : FAILURE in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c54a09d9224b909b26cf57992a76dd : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62b5fc6ba94b4b61bc06d8aa19455271 : SUCCESS in 12m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bdd0774d1b143d3950941d05d4db63d : SUCCESS in 11m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5a9aba009c214430ab0a19194ab9184f : SUCCESS in 59m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b84f3051e8fc4ef9bc3bdbca9040bf61 : SUCCESS in 1h 29m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ca8dfac503774326816a1b16e742914a : FAILURE in 59m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7f06f455adad4a718fd4559e4cf50a56 : SUCCESS in 52m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8cc367017143480a8096587c8dd6fe2c : FAILURE in 1h 29m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0c99475f6ff445b0997acb62dd7bc461 : FAILURE in 21m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/dd2c0a1c04ea410e85810f15615490db : SUCCESS in 6m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/65702d4a15564b528be12d75ead59c9e : SUCCESS in 3m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/71c7c8017f064671bd4899b4299853ae : SUCCESS in 26m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f8cb6498fa84457793abe6646c165b80 : SUCCESS in 40m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/987dc0911dca4f0eace6fbfc191716b9 : FAILURE in 51m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/329841da35154740a93a3cb10d96254f : FAILURE in 27m 18s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/50ef8091d6ae4d589d01aa4508928b20 : SUCCESS in 58m 37s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b915a44df036421e9b64f1c0ee0320c5 : SUCCESS in 49m 21s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/135f5c1d37a64a358477ca780f857f50 : SUCCESS in 18m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5939159c431a4943a1b6f59af341ed2b : SUCCESS in 17m 44s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/91e520e60cb14f968eecec1f032056a7 : FAILURE in 3m 54s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/9ea12d4da1934d3785e522b5c92109ca : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0425e2b4188b0006aba02651db4ba77985edfb4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-10 14:51:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"beff1f4c8a52528ae5fb9a0d64fa08c0b07f2db6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-10 16:18:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/970301/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/f84abd2/ : POST_FAILURE in 1h 25m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"58ce4c39bc5a9851226a0c33697c492ff581e0a9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-11 21:02:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea90f8128940fa17290179334103b0372813f1db","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-03-11 21:03:07.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/979810 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"d5c858c59b8b9850beaa035267fe013252374419","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-03-11 21:03:37.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/979810 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"464005e83cf555062ab55986799d5476d51fb769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 21:03:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4e3f3309e75576df75fbf2f0d8f00d020fbff53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 22:54:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57a18cd9fc4c4e968d149a2e2dd965a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec9499134d1487fa8866c6ae43979b2 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f779ca6176cc4dc3a15735638fb468d5 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73a9ec1398b249d98adff33a6eb1f0cf : SUCCESS in 7m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/534dfdb494bc4e0cada6e19d856a702a : SUCCESS in 10m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5cc62ae238b94369983c9b75c9047666 : FAILURE in 1h 44m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3a61514a07b74f958e97f035838a06e4 : SUCCESS in 42m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/39a5d85f0096463e86db4c4783ca62d4 : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5c0cda73912caf20dbdf1b7719a1b9dc6d182dc1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-11 22:56:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80468499abf75e131bad84dde49834d088d47dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 00:19:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898523d63e104528b7f791888411157b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9dd1ea4f55340d5b14e528639f6ec90 : SUCCESS in 11m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d354b2f326ae4226829bcfa066418ca4 : SUCCESS in 30m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca327468c0c14bacabc4cf84b4c7ef87 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb4e055bebff4c41b95d6098ea064f39 : SUCCESS in 6m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8fedae344c84f6fa16f858727e4ae3b : SUCCESS in 9m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/79f821f81fe14b74a6904bc769f5aa29 : SUCCESS in 38m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/10596bf0364e45c7afb7026827395414 : SUCCESS in 58m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f26879e58f7547fdbaf4fa4516bf2113 : FAILURE in 1h 02m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8f993190216f443aae43a07ec566cea6 : SUCCESS in 32m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f555d579766c4155a2a4ea3401851616 : FAILURE in 1h 15m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/13e84d8b71d248aabab557dba2c280cc : FAILURE in 10m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e3dcb8110e2449b7a8f0cd0e9886aa5f : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/267f7499470849f994c27e67390d497f : SUCCESS in 2m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e14cc602af314c658c7a18c5f121b67f : SUCCESS in 35m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c58a462f3426450da65b1ad5f7e31fef : SUCCESS in 36m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/615c9d994cc04e64a3dfb3732e0e1c96 : FAILURE in 26m 34s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f3d1cd25d1ad40ffb55a416b82878228 : FAILURE in 37m 57s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d07a2bb518ae4508824f5bf5b9bdd35e : SUCCESS in 36m 31s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/72b3373350584f68b74e34ffb677d09e : SUCCESS in 43m 36s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/484a9dc96fa141b2bbe0765712547b0a : SUCCESS in 24m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dfe50d6e35e943b18b437615b0363831 : SUCCESS in 33m 00s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/d2e829f260f14057a42c06492d1a0a76 : FAILURE in 4m 16s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/27432eeb7f5443eb849c9ab241225147 : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4749d520105c7caa0228b68a8b3e846ad338733f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 00:19:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bff531c235548251f61dfe7326fcd5e1af05a4fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 01:35:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef77cf98190349cebd6de1bbc2c5874d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39433cb07114c96b07d5fdc92327947 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b92843bf33734ac3a1ad835db45f84c7 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4334678e368489c8ee945e3e826d85c : SUCCESS in 7m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/139c490a14394e10b11b4d6478e5143b : SUCCESS in 9m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e4b96c0f7201458cb40ddb8e05143755 : SUCCESS in 1h 10m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f9ead547b2b141cca2ad74f9d4e31c68 : SUCCESS in 26m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/909d919e75744073ab6020b8950c76ba : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b189bb05acf026dd0e19c4481250d9361e3e4e86","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-12 01:47:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/970301/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/27a71cb/ : SUCCESS in 1h 27m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e1615fc2217b9efddbc1189c81713c209fb5569e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-12 05:19:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"036cbbadb16707b6a81de41c803613b8929a6dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 07:49:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b453c29ee856498ea3d143c4836435c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faeaf1779c5d4df4bd39c90d2fa44600 : SUCCESS in 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8141a3a8c00c4b0197cc79b9f18aef35 : SUCCESS in 18m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b1a74f7099443fad1559252b06d0c7 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aa67a6df8794614bbda819686ee778e : SUCCESS in 10m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d6baa87371a47ee99227e5d7127ec26 : SUCCESS in 9m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c65be93263f54697aa45246686d3abea : SUCCESS in 1h 09m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f727bc63e1724fcba483c730549e64ab : FAILURE in 2h 24m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d16d72e2ffc64117a09f03e4feb52368 : FAILURE in 1h 08m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/af80c0e75bca49e18889a86f088fbcd8 : SUCCESS in 1h 09m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/27a2405ff4534f80947c02d370d6644b : FAILURE in 1h 36m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ac001e556f6748afb110a17754760617 : FAILURE in 19m 30s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7565ed4687174b2995924e3a7018b1df : SUCCESS in 4m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ca2c3617dd74f8182dd57ebd315e427 : SUCCESS in 4m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e5ec140ad39d4e538b69792c173169b5 : SUCCESS in 32m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c1bfe4446dd8412b87312d8572e38d1e : SUCCESS in 33m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c51730f5063948f9b190244013b0cc0a : FAILURE in 38m 49s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/40e454c60cbc43178fcde5a33ab940f5 : FAILURE in 22m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d9ed5962617048d2967db5cfccc2f14a : SUCCESS in 55m 30s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ace57b77e5e42f29bcdbba1e64b29ee : SUCCESS in 54m 42s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4a4ddf03a0f545398a136cd8c39ebdc2 : FAILURE in 5m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fda018e8678748f8bca1aceb9881edab : FAILURE in 6m 58s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/9f7f7aa047c04c17831f45496bf819f7 : FAILURE in 4m 21s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/6b0abd610eb246c38ce0f4f1a19c2490 : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a17d7d915adeb0a4fbe40b0fc4f947d321d84c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 07:50:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb797e243a4f721f29c90ac2932ff72023c6c7a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:06:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5f7f4752cb4d28899c6c83575821e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07162af96fe4918923abda72eea433d : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff084926bc1941ba81bfdcd434cedbb8 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8aca2307b3142a4bd07ec7a147f6870 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de40e2389c144678aaa47e234dc0819a : SUCCESS in 11m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d8397ce994c94103baddabe29ac06da5 : SUCCESS in 1h 10m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/244d98896e2d46a0827c85743698938f : SUCCESS in 38m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bec970fd5a954702bbf42d17f0ccb006 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1c21bafd60ad0c34fa821604eeadcc4610bf5dfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:06:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"097e870c496b69472df4c919591983dbcb505a44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:09:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5567c36a69d349269ef3c55df2e280dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c004a7b759ba4692aa42d29ffd3b978c : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2},{"id":"64538ce471759415d1cc0d3976e01604d09c1ed0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-12 09:20:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/970301/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/abd4178/ : SUCCESS in 1h 29m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a93e9cd2487895e07dc84f16bbbd41a17bda737","revisions":{"e003faafe1132a15a7a77309ab256d58c87612f1":{"kind":"REWORK","_number":1,"created":"2025-12-09 20:03:42.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/01/970301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/01/970301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/970301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/970301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/970301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/01/970301/1"}}},"commit":{"parents":[{"commit":"3ea115c1c31a84c7584442c0e6a5fabf962b6cf3","subject":"Merge \"Allow configuring netapp volume retention period to zero\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3ea115c1c31a84c7584442c0e6a5fabf962b6cf3"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-12-04 10:45:20.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2025-12-09 20:03:42.000000000","tz":0},"subject":"Fix rally jobs","message":"Fix rally jobs\n\nPreviously, we forced rally jobs to do not use the global venv when\nit was introduced a couple of releases ago, as it caused a breakage.\n\nLater on, we started seeing a failure in the rally jobs, where\nos-testr installation/upgrade failed, and the reason was that we\nwere attempting to upgrade a package that was externally-managed,\nindicating that this installation/upgrade should be within the\nvenv.\n\nThis change switches the rally jobs to using the global venv,\nas the issues we saw back then were fixed.\n\nChange-Id: I33275c6eb5b137d78b10a9942ac4f765d011eac1\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n(cherry picked from commit 75e895d5722e4434d7cc0d12996f9f68902c1539)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e003faafe1132a15a7a77309ab256d58c87612f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e003faafe1132a15a7a77309ab256d58c87612f1"}]},"branch":"refs/heads/stable/2025.1"},"1a93e9cd2487895e07dc84f16bbbd41a17bda737":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-10 12:50:40.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/970301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/01/970301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/970301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/970301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/970301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/01/970301/2"}}},"commit":{"parents":[{"commit":"848ebfa1cc4a6ae4d2fea9d01c4aa1469da07b84","subject":"Pin setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/848ebfa1cc4a6ae4d2fea9d01c4aa1469da07b84"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-12-04 10:45:20.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2026-03-10 12:50:40.000000000","tz":0},"subject":"Fix rally jobs","message":"Fix rally jobs\n\nPreviously, we forced rally jobs to do not use the global venv when\nit was introduced a couple of releases ago, as it caused a breakage.\n\nLater on, we started seeing a failure in the rally jobs, where\nos-testr installation/upgrade failed, and the reason was that we\nwere attempting to upgrade a package that was externally-managed,\nindicating that this installation/upgrade should be within the\nvenv.\n\nThis change switches the rally jobs to using the global venv,\nas the issues we saw back then were fixed.\n\nChange-Id: I33275c6eb5b137d78b10a9942ac4f765d011eac1\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n(cherry picked from commit 75e895d5722e4434d7cc0d12996f9f68902c1539)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a93e9cd2487895e07dc84f16bbbd41a17bda737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a93e9cd2487895e07dc84f16bbbd41a17bda737"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
