)]}'
{"id":"openstack%2Fmanila~970786","triplet_id":"openstack%2Fmanila~master~I7bc2ddde5d83e5dd81e0437fa1e745c85f308027","project":"openstack/manila","branch":"master","topic":"bp/manage-with-mount-point-name","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-20 19:16:50.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-20 18:07:04.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-02-20 19:16:50.000000000","reason":"Change was submitted"},"36741":{"account":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"last_update":"2026-02-20 19:16:50.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-02-20 19:16:50.000000000","reason":"Change was submitted"},"15595":{"account":{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"},"last_update":"2026-02-20 19:16:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bc2ddde5d83e5dd81e0437fa1e745c85f308027","subject":"Allow using mount_point_name when managing share","status":"MERGED","created":"2025-12-12 01:56:50.000000000","updated":"2026-02-20 19:18:24.000000000","submitted":"2026-02-20 19:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"970786-bp/manage-with-mount-point-name","meta_rev_id":"d22515dd2a9fb51b982dc70e6a2d525d0c83e5b9","_number":970786,"virtual_id_number":970786,"owner":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-20 19:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-02-20 18:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":1,"date":"2026-02-20 16:47:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-02-20 18:07:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"}],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}],"REVIEWER":[{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 04:28:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 08:59:06.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-01-20 15:00:11.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-01-20 15:00:11.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-01-20 15:00:11.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-01-20 15:00:11.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"},"state":"REVIEWER"},{"updated":"2026-02-05 11:48:33.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-12 11:21:20.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-20 16:47:29.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"dbc78a1b4cc6d2984bb49650438aa03278a44c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2025-12-12 01:56:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"393615c11fb706c648d6566051e3dc1a4eedcdc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 04:28:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fef651a33aea403f980a5395defb1405\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f7d5f41234fe4befa6f66a21a40bcb2e : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb24f8d3e864966a5ec987ed5f135c0 : SUCCESS in 13m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/694906d9339f45c9a96178514ce83b08 : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9789f1738990496d8c2b207db8794399 : FAILURE in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74de6b4a47d44f30a443bb40e010d751 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f5dae771faf4b5b86377486fd2a6970 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eed0070281c14241bde8fde9b7684c3c : SUCCESS in 10m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2cdea02b998b4958b99d8f42523051bf : SUCCESS in 1h 24m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/79222f2861864580ae03d2b8e1e071b7 : SUCCESS in 2h 26m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a9a853bf94924e819097ec3e93d477b8 : FAILURE in 1h 16m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3750769de8ed42d98e938dd5ba449056 : SUCCESS in 1h 00m 36s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/36aad14dd45546a5870988592cc70042 : SUCCESS in 1h 22m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f5f9a984ee2642ca8f512c5f3c2296d5 : FAILURE in 25m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e75e0c337d446c2ae909fb658b2d9e4 : SUCCESS in 4m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eef1853d03334a59a8982660132815b0 : FAILURE in 2m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dc038372fe4445c783a1daa8079549a7 : SUCCESS in 35m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/91bcd3f43cb94869bdb364c2cfc522ed : SUCCESS in 27m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/85f155cfb3ea4119977d0ddc54921fdb : FAILURE in 24m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/f4fa10ef39fb47fb80213ac7828f50e3 : SUCCESS in 38m 17s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/86607067741b4eb883b117674c1d3ede : SUCCESS in 1h 05m 28s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3960ae56b9bf4fd788e27f2bccef1fb3 : SUCCESS in 56m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/645d87147ef9411d9138b8c1cc87088d : SUCCESS in 24m 41s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/01c87d242cc843ecb159439bbd3c7544 : SUCCESS in 25m 25s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/651d831e51a648b9874628c343008a7c : FAILURE in 4m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a04dddf11ded4025a443f0169dba8bd7 : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b7e650337e8eb4bce578ced6520b1f07fa396e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2025-12-12 05:51:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"730510d855d6c6db5242c6c9e2670a04074b826d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 07:20:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a801cadb364485bec3a475b4c4ba55\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6132fbd9261645ba9d36be1e59fda664 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ef60dc1141470f8d175c9fb5426fcf : SUCCESS in 12m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/763808a012c14149be72d607428007ee : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b0424061cb4deab3827d00f44a90d8 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5618afd750604860b042b12f9b6e458b : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e9787b0d9184bdf8a27248b98ccd113 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3f441c8109948698a3e12348dcc203b : SUCCESS in 7m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/751e52a472514461aca1225c1b8ad190 : SUCCESS in 52m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3051dadea13e4f25a3e0a6a265a44191 : SUCCESS in 1h 00m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d0d9bf155b6e470198887d02777db2af : FAILURE in 1h 17m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/630bd3827d0b4e35829b10855510689d : SUCCESS in 1h 14m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/252ec18ae9d04c248a384a992bdc3fbe : SUCCESS in 1h 23m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1ef5763ca182461bac94d799bb78cfe0 : FAILURE in 13m 40s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ebb0e51f03a64762acae547abb4aca11 : SUCCESS in 2m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8743d1221a24ecfbaba04816a8cf0ab : FAILURE in 3m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/43267344373648ff9b0c28d0157fa86a : SUCCESS in 39m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/896cc8672f6147c0a195f59287863068 : SUCCESS in 25m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/04470efae066481baf0fe31523a4df0b : FAILURE in 16m 17s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/09cab9a952e54dfe9dddd6ff257c909b : SUCCESS in 32m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/863f57792084479c9e32cfa4bc761df6 : SUCCESS in 42m 40s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be7c5a5757334190988e5c78b5ca7de9 : SUCCESS in 54m 07s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/31a2fe571ac04698af723da279d97355 : SUCCESS in 15m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8bdf680d8ff0450882fa4bb5b7e2d5ca : SUCCESS in 24m 12s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0e1c28f579b34c76aa882dd953426626 : FAILURE in 4m 09s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/579cdde6f0be4679989a8f5e1dd21af3 : FAILURE in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81c9ffe53e9c4a4ad6dbc44a43a6ccc8f96fb47b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-12-12 08:59:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/86/970786/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/2edfe4b/ : FAILURE in 1h 37m 05s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/86/970786/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/6bb8c5a/ : FAILURE in 1h 30m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"1744333bc15ad6bfa4c218ed9ac78dc857028acd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-01-20 15:00:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"43ef3fe6b262ad372ead139a2d251be2d20d6527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-01-21 04:27:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"79f1add8f64c58f4aad9c52d23300d1cd6cf947a","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-01-21 04:30:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"45025bcf2c1781bf863dbe944fdc391990c31ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 06:49:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31a33a8282e842a2895803d7414e63ff\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ae95f8fe2794421caf38a4cd1c225bbc : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5a0ec3fffe4d1296bd3fb1361cbdb2 : SUCCESS in 6m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af15af6ba6f743cd860caee577d499f6 : FAILURE in 32m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3238310f6d6443dad9f3f3c8b840a2f : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f587a6d4b9cd43de9f84222d55d83249 : FAILURE in 7m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9d1544b0a55400da4921bcbdf5dc131 : FAILURE in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e664843baaa4074826ef526b0cacd97 : SUCCESS in 6m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b649d0b2efe343b8a9a725e97b9323f8 : SUCCESS in 52m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9336d9cd35fe457ebe3f21e5070205b1 : FAILURE in 2h 09m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eaaca6efe42c49fdabfce07f4c799d52 : FAILURE in 31m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3b9d0130d7a84edb89b872324f8844dc : SUCCESS in 1h 00m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dd0758df1b12422da1745186b3cb67b1 : SUCCESS in 1h 29m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0fbb73ea920b4dd0a5db9eda93cbf4f3 : FAILURE in 35m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ed9f5a9ef984460bab413289f7075a6 : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd63576031964e1d86e90f63f3541408 : FAILURE in 3m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/367cdb127bd94026b2e7816e51b1dfb8 : SUCCESS in 27m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9fa2698282af4abd9d06de653dab11e2 : SUCCESS in 19m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5f6ee15c52a64be7a2ba401a235e43f5 : SUCCESS in 34m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9bb7c562f0f8492f99b461afc85ab165 : FAILURE in 43m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/5a1610b090504fb3b719a8885ea1a63a : SUCCESS in 1h 10m 18s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8f3fbf87d5dc4e2eade5ffc881fccd11 : SUCCESS in 53m 04s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29a170d8403e4337ac2e98eb293e9d3c : SUCCESS in 56m 28s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c6a18ca9ee82490b86c9362d36cb642c : SUCCESS in 32m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/434b161eabe94d98ac281786ad8e6b1b : SUCCESS in 24m 23s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/81d1f54a16cc41c18fdc50b1a7540f66 : FAILURE in 4m 20s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/322a4f5bfee542d58076d43505ab8033 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a125490f4b265fef4b20d2ea32c72f3127a6be83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-01-21 07:20:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b6716d74f6c7376e414cc99e6d440255f64c9e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 09:04:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e350a50958624b1d9242bcc37f113309\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50881de89f2a4bc2820d99162d01d39c : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57cb42ae780493aacf9048b269ff73e : SUCCESS in 14m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb4cc02525a047ef85ca9ec91c9eb8ab : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef9e41abb854ffface327d19ebb7870 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6114000b56984d7898ee9ac1709da9ed : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6f79ff8e971465a9342cccb82ab5484 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ab8a57bf6be44d786f46641fcc72f29 : SUCCESS in 5m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e9655e8b5ac54ce2a132338b2f2213a4 : SUCCESS in 33m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c4bda4e34d614a72aaa2fcc0745b2107 : SUCCESS in 54m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d072830510f04ef8a0e6090864b22e43 : FAILURE in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2b07dcf3645c490391fc897ba2c0a985 : SUCCESS in 28m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4c97380ddc8646e988dca1b28530d7ea : SUCCESS in 1h 37m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b821c5f0f9114cae8839363f3b8a2d6b : FAILURE in 53m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f99f5afc5414d80a92722dee2f45c99 : SUCCESS in 3m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d50639d7d916485bb11bcbb7ace0c1ef : FAILURE in 4m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fdb99c20825944629718386872e6cfce : SUCCESS in 23m 45s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/28c1854400f54eef9953956a0364238e : SUCCESS in 31m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1eac02dfc1b540588b9d43929a29272d : SUCCESS in 38m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/29e699b7af1d4379971d0cb29416d830 : FAILURE in 42m 25s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/201a7be8583d46d29534b0e55f25ad1e : SUCCESS in 1h 14m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f831eb14cd244fb6a7ceb44109c29404 : SUCCESS in 43m 13s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4fcfff90e0e42188cd8ff45455b5def : SUCCESS in 36m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/600999d97c004b99a62d5012d4e32a42 : SUCCESS in 31m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2cd8789402e24deb80a84c2c1ec16e2e : SUCCESS in 34m 08s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/cb2df582209643e4a46e284ccf99874f : FAILURE in 5m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f3f50a9dce79400ca3a4eeea07029e0c : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e61fd7d1691a21ba6b94b41ddb2543b6e88a9741","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-26 18:16:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ec95ca0e8029da8e9b1b1c6ffb093eca64042431","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-01-27 08:31:11.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a11e80a871236e290ef2d977b6f3e7dc0fc252e1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-27 22:36:56.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6889e7de3dc91ac061cfcf8fbe17b5d510d21004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-02 07:42:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e204dd7c9a3523700e228cc7b8c55c36fa24352e","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-02 07:56:59.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"530037af95780c0d2f25300664dbe7b3b0e0ff23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 09:13:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe4b56e4735043c8b31527f305d98453\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d8dc1f4392d942729ad9a6a3be13fd91 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7619952fe5d4bdbb49bd53d0aed813a : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f7dc303263a4cd69b9229d381511c38 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5cfd484c754f9e95a858195ed74f69 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/293ff0a394674935a3112d32e5208ea7 : SUCCESS in 9m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b57b4fdc377e4aa082a22d7c9263eae0 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f0619bd3b95424bac71545874c0c347 : SUCCESS in 7m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d30e488a6d1e4af795fcdc4038ad95d0 : SUCCESS in 36m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7bcfca4c18e94b6dbc7be800523063f8 : SUCCESS in 56m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8cc477e586444eb8a43abf9ae604b972 : FAILURE in 35m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/05c97eeba01d4c2bbb1bc4dbc8d21b62 : SUCCESS in 33m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1e427f1c15cf41729dd35e177992ebb4 : SUCCESS in 42m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d628039dcc304e62bd3695b6117bc9d4 : FAILURE in 1h 25m 50s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f08928cbcaf494e8183e329723577ab : SUCCESS in 2m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7de2396daca4b0291ba7da07cafb08a : SUCCESS in 4m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e79873176c3c4fd2b2692ecd17206400 : SUCCESS in 34m 24s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6d5e04f3e1eb40c1b79f32eaa54bf957 : SUCCESS in 30m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/830c297a3c6e4ce7b48f91d6f57629cc : SUCCESS in 41m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b7d4572d77d24ab99c8a3fd362376e69 : FAILURE in 1h 05m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/c87d8120ec6141d5b5996b1c1b95d3a6 : SUCCESS in 1h 21m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3442dd02bd954196934bc5f6df91025d : SUCCESS in 43m 35s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb28f299849c45d3a15fe78ad4c150d4 : SUCCESS in 59m 40s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0df7d85385bf4de8a19bc0c0669db291 : SUCCESS in 21m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/81c564ee84144466ab80da3af4993107 : SUCCESS in 35m 21s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/56954944d2a84e76bf3b52817fc21d18 : FAILURE in 4m 50s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/56dc531bc6b14a7c85accb9afc6916a8 : SUCCESS in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2db9f03f940b3466e5e2bea66b21849351f383ab","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-02 17:55:18.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"63d913330846cb5b239b12f01762de03409513ac","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-05 11:48:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/b348bf31561f432b9394b6613354ede0 : FAILURE in 1h 07m 48s","accounts_in_message":[],"_revision_number":4},{"id":"382a7832df6242b72492175c21f338fc158bd248","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-02-12 11:21:20.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"edbddeac5e99e40fdf5e2e13af9542fcf3d6e6e3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-17 06:45:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1df478464869bb16fd732f4883da28fc83b9c266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-19 01:33:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"63af9067dcf3fdcfade5602db36cf37d78a7ebfa","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-19 01:35:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"eb193b50c740907be63281c5db6457662ceb11f7","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-19 01:36:35.000000000","message":"Patch Set 6:\n\nBuild Succeeded: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/970786","accounts_in_message":[],"_revision_number":6},{"id":"8bc6fcfd0f8d37ca1337e6d0f3cc6189a3531f8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 04:20:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/523e8dafec614fd0aa470a48e5955dea\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/db22fa5a72984fb1a883a8c3af664e71 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53296cfcef284f149cbd56e698f1aeac : SUCCESS in 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/828a25d3c0974ae4993dca3aa59e8091 : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2294122896084cc9aa897d89af9cfc86 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43a400b58de64c178ddbbb7c03f77d73 : POST_FAILURE in 9m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44a0542a56a947989a6a714c1b6173cf : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b39653f1044547c6b373c2ff53e75400 : SUCCESS in 9m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eccf85b52a714562b012b64ac1ec96c6 : SUCCESS in 30m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bcd97d18e7d3421584a0d87232e9d5fe : SUCCESS in 1h 36m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/14fce08f664f48048923e92d3089d733 : FAILURE in 48m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bcbedc5ae28e405998c6a818c1ba00ff : FAILURE in 1h 00m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e638b7dcf57940e48b1b86fbe61f0ab2 : SUCCESS in 2h 44m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8e22e9ef721747dca760f13eed1ad083 : FAILURE in 1h 06m 37s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d78f4a19fd7c4a2a93ac66e571d86cac : SUCCESS in 8m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4481c1dc336a4967992023a17f07246c : SUCCESS in 5m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8e5530577914f4fb9f0c468ae53724d : SUCCESS in 37m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/31407ccd008c40c6bd6bb75571fec5d5 : SUCCESS in 26m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b4f07c7b48834a2597c588feed75308b : SUCCESS in 29m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9faf5d044fc64db3bf31647befe9e004 : FAILURE in 39m 42s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/566674c9c4614130a32229576609d6c3 : SUCCESS in 38m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/055aff4887574e7ab31be6675eac71e0 : POST_FAILURE in 1h 05m 54s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f14d3d2ae46147419ea8b1afbad0038f : FAILURE in 11m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dde0792f4df447ffa5a4b2fddc51c037 : SUCCESS in 31m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f8eda7f068c44e33bb6f5974b1eac3bb : SUCCESS in 22m 30s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/8de72635d444406eb2bca1428fc5c531 : FAILURE in 4m 52s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7950e272871141ae8c2bb249ea761d14 : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"af2897218af8ea0f5c30abb7347080a826b9ca45","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-19 06:07:16.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"85824b965765ca59601ddde5f9016deedd91f1d5","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-19 06:19:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6104e962b0385fb10ce82b6905b3887cad8a2d71","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 17:51:44.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0eeff259b551dd9315344775aec379fe227546fd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 17:51:48.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7c082e6e9a49ae95d6eef0f4635e108d2e99037d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 18:14:05.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6a281c9ae12a5eeef30a6fb36c6069e029d5323b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 18:27:32.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"fb20cf909b86c51240bf6d574a31cc793287a442","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-19 19:09:35.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"61bfb54d10bc8d39177ff9a72f2928ab75fd26c4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 19:10:10.000000000","message":"Topic filter-by-ensuring removed","accounts_in_message":[],"_revision_number":7},{"id":"851d8ce37937dee6bf0ed44495fa88a24bb23949","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 19:25:02.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2d37f4a6af21076f44bbea397b69ef9a55e88acd","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-19 19:25:52.000000000","message":"Patch Set 7:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/970786","accounts_in_message":[],"_revision_number":7},{"id":"1ac8a6bd646d0bbacce0a61a4a47316b7f6439ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 20:41:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de6699588dc465eb25f7dd1842c280f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/81a1069f52d249e5b46123b277260506 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7703c88b9e74af8ab08cad73054ed0b : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58b248b011cc4b6fa1ae36e3fab340bf : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e700a5ada51c4a6c99129bb4a05cf118 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4295501e272648c7b20768bcdbb55810 : SUCCESS in 6m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d72a23e768f4acdbad2b998b4c031a7 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/907adbb7e6cb4772baad8faa9adfd466 : SUCCESS in 8m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3afb87fc72b54357bff66fa7ae627d06 : SUCCESS in 1h 18m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/64444702d1c840dc88700af6bc411e7f : SUCCESS in 2h 11m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/204ae9e1c4744060a216212adee3c0e3 : FAILURE in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/208fe01546b84a54b2a9d1fd5009dbea : FAILURE in 53m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3abb95a853f442e7a0f6795a90b9d5e3 : SUCCESS in 1h 24m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/83f3f46076ce41ec96588b6e5ec9c91f : SUCCESS in 1h 29m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b1cc3510c199425e8e0f4156f48ad517 : SUCCESS in 4m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6edc7f66caf0443d9cd34f59bae71222 : SUCCESS in 2m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd7d37fc224e4ee087b0f080275d40ac : SUCCESS in 33m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/591ce0f441214d57a3b6b4e693afde0a : SUCCESS in 30m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ddae22d392984e488cdb580105f2a566 : SUCCESS in 29m 38s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ada570f8e347412d936507790fb47c59 : FAILURE in 58m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/e720e2279e684fdfab8e414feaab3faa : SUCCESS in 1h 12m 10s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d109ff8a6c1e4db2874f81f6892c9962 : SUCCESS in 40m 33s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/530ed7f8cb0f4653982628aeb6ec99cf : FAILURE in 6m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c51f6b42acd54488bd28c6981e1f7c2b : SUCCESS in 31m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/db68fb86d95c4ed9a82969a90e57bc8f : SUCCESS in 36m 10s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/76ac8b0177874fd8a3975909329052c2 : FAILURE in 3m 08s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b28f3d111a274d5e8e03d6e9a334545f : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5e23cdf40392fd5926d9805b7e84e6ad6f74bc1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 20:41:20.000000000","message":"Patch Set 7: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila for 970786,7","accounts_in_message":[],"_revision_number":7},{"id":"df1f5cf0f6b45952d5eeeaa19cb921dac6e660c1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-19 20:41:30.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"13cef93a8a8d41485d2e2b940c4061565ae4e8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 22:09:14.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"89c7eb8233ebaf04cd822776370899dde3450d8c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 22:10:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9d75945ecc50b52ab0ad46fad76696a637b238ae","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 22:17:25.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3d42f53c9fbd96b021ab38161863c13e0837addc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 22:22:14.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"4d80e4b31283abbc6ca39d9a56f9d7aea5e184f2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 22:23:19.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"30d84e31a8c05209778325f2af2574caaf8d96bb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-19 22:24:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0b5f9a8332e28e50f03ad07a457cde5bedfc16f7","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-19 23:44:13.000000000","message":"Patch Set 9:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/970786","accounts_in_message":[],"_revision_number":9},{"id":"71af9befc3ec65e0b868ad37c4e1b59db249d1cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 01:02:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d97db7eab4d4106b85bda9b957e6869\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aa3e9ada1bad4309b10c4c580674dea6 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dab9da4937b422fb9b8bf69b4d62572 : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/273ff9713edc41909b8958429a302ccc : FAILURE in 24m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb78c07b25d4bcabec508e8ee98f44e : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c6c2e13415b4e4588a1bbe2535b0778 : FAILURE in 11m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bea424a7f9b34cd6b5ffb3f2d9e51614 : FAILURE in 6m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eeda61accf6c4e45ad7e1d40bf690e93 : FAILURE in 8m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5994f31b198b471fb8725758942a5537 : SUCCESS in 8m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/96f032fa17cb4e12ae8928cc5a6087bf : SUCCESS in 1h 10m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3423bda902054dbc98dcfb62e541a138 : FAILURE in 2h 33m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2efc978a77784b398f552ba7a514196f : FAILURE in 20m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/eac29d7b6ef941788ac2d33be91d56ea : FAILURE in 34m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/832a74c59269439ba201c89fd2f1dae1 : SUCCESS in 1h 45m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/49b0d85ea57f4b4280dfc869f2504489 : FAILURE in 35m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0e4f28505ef14c7a8e7fd8c2791a1d94 : SUCCESS in 4m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6bad4afe25ae490faf15424154e0be66 : SUCCESS in 4m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8ea3e15687a5421d87e0045ab8c0aa08 : SUCCESS in 39m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9d4449f2ca8c455e83fe722b606c87c3 : SUCCESS in 35m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b463d319faf4b8bb3ca7e7529335337 : SUCCESS in 36m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a5e879a2553c48a4930341214c423a02 : FAILURE in 46m 33s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/d07f1f4e40cb48bb8cb03839971edb1e : SUCCESS in 1h 10m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9206d51c21c64769b36ac75882ee197b : SUCCESS in 54m 50s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f56b57a42e67412d83b96b4f55e2ea39 : FAILURE in 10m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e7195c8ce2224d0aa29eabd464903ee4 : FAILURE in 28m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ff51fceeabe24f4b82fbc91109806927 : FAILURE in 13m 10s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/bb8e93f745ba4935b8238806098a7874 : FAILURE in 5m 54s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/dfc4c390a96f4e4f8dfa53ad8dfe5099 : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7803a24121a9737d0110750173bf50b741ad5971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-20 08:18:19.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":10},{"id":"c7d50b6bed35c1db398867b4bef9027313e920a1","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-20 08:41:47.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"dc17a88092cb7adf99fd2af7ad9a03ae0ce72787","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-20 08:44:11.000000000","message":"Topic set to bp/manage-with-mount-point-name","accounts_in_message":[],"_revision_number":10},{"id":"f7878a8e0fac182fb9a2371177dd53c3fc3d2a56","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-20 09:12:22.000000000","message":"Patch Set 10:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/970786","accounts_in_message":[],"_revision_number":10},{"id":"77e9ec08c5b3f960c5799af5cc37c0229a4a96cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 10:04:59.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/906f4291ac01449faa3f0003fa46f555\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/24721afc21e841ebb60d5853536e8fdd : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3d6c6b35cf40e08916ccf81cd4bfa0 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/087ccae8de1e49458bce6ed78e4de67c : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786d4cecf3444c08939e1508b8762ddc : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4725a44767cd48ac97bee2408b4b976e : SUCCESS in 8m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/761aef4db39b4b81951653b11fd49c5b : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6981c8b60e514fa4a97959054f259d84 : SUCCESS in 11m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/811b866d936b4aaba5209096bdec24fe : SUCCESS in 8m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0cf0f5a87a4e4442b208a7d92a3d047d : SUCCESS in 56m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dbb67db9fb97424bb97580fff58d28d9 : SUCCESS in 58m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/343246c080564169ad9eedb434c4e2d5 : FAILURE in 1h 12m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7aa64f4523864d8cae4f82030d39ee71 : FAILURE in 1h 22m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dbb01fc9c3514265949becffe239c334 : SUCCESS in 1h 38m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/486acd82b1344f9f9f7c31c82e9ebd42 : FAILURE in 58m 54s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b519208f77ca49578ea3a57c05dc7948 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2dede18b2bb44f339f03a410c774488c : SUCCESS in 2m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5694b0583b7c4905bc7855cafdef6174 : SUCCESS in 36m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3462ff91c2c047c88cbaa2d5a67be24b : SUCCESS in 24m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4bc2609948264719ac954ed74e937a57 : SUCCESS in 38m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8b4de811952d4f5da578af18d5f049a4 : FAILURE in 47m 22s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/f8b1a2c5c0494e7891145f4aacb66029 : SUCCESS in 1h 08m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9f525b4f32a14e9da8995b4207be5577 : SUCCESS in 1h 06m 21s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e271103a739b4577b8d77f35990fb4eb : FAILURE in 7m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/124b8a9bbc1c4711a863f88fd23e6d19 : FAILURE in 30m 41s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/82a64fa7aa804a87bf620cc5a56205e4 : FAILURE in 33m 21s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/47fcd5cfeb9d4bfa9b4ac576adaede48 : FAILURE in 4m 26s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/75b405a0d05741f9928c11177c5d50cd : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d0cf4032cc27bc04909cb8eee2f90bf8562d80a7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-20 10:17:47.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/86/970786/10/upstream-check-manila/manila-tempest-plugin-ontap-dhss/4bd8b03/ : FAILURE in 11m 09s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"df9c0774e5413078b76943f74401bd0b4ac1f975","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-02-20 16:47:29.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"57ffcb86daf0f93ed7150f8494169900dc300740","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-20 18:07:04.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"acd9a317fa441cbbf91d7dd78d20dcc28909c92c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 18:07:28.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"fc9da994403efa86b7c040429b09c7dfca4cfa3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 19:16:50.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8350e22b676c40768c7b2af779db33fa\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/32128ac1e8174438992661a45d8f7fcb : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec5d3383298c44a8bfff993c2e138ed1 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b8d07113f24c0792f1e24bb083a8b8 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d59a8821a7e4dc19fb6ef4522265de5 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24f603af7f2b44feb830de16240329e3 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f10d12e824d4e5cb23a237ec653bdf7 : SUCCESS in 8m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6632fce7a3ee46ef8cdb236094b9932e : SUCCESS in 1h 05m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0714e7a55a01439dbfeb682ba674ee6f : SUCCESS in 38m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/63a840ee71fb4e60b3c25ae19cfd9c45 : SUCCESS in 31m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5b46591aa58f47c3b9442ef0727890f5 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":10},{"id":"50b87ad6241ebc3828797713399c6e52a6f6684c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 19:16:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"d22515dd2a9fb51b982dc70e6a2d525d0c83e5b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 19:18:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeaefbf101ab4765b4317ab44e8589e4\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cec9c9370bf04c448938e86a016ae8ad : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48679f061848465f9c219138774214ec : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a22c84e6f654fd0bc2afb5202fe15ee : SUCCESS in 46s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"fa13cec403cc0b254065770fece133b7c9c2c0e3","revisions":{"46b09e09f9e98142a48e4e6a25325e697011b54a":{"kind":"REWORK","_number":1,"created":"2025-12-12 01:56:50.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/86/970786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/1"}}},"commit":{"parents":[{"commit":"13086be27244ec9318c3b3641e7055364b73265e","subject":"Merge \"api: Add schema for share unmanage action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/13086be27244ec9318c3b3641e7055364b73265e"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-12 01:43:10.000000000","tz":540},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/46b09e09f9e98142a48e4e6a25325e697011b54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/46b09e09f9e98142a48e4e6a25325e697011b54a"}]},"branch":"refs/heads/master"},"f6d3c07589db8e8116f4d7a961e3e803e661d72d":{"kind":"REWORK","_number":2,"created":"2025-12-12 05:51:23.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/86/970786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/2"}}},"commit":{"parents":[{"commit":"13086be27244ec9318c3b3641e7055364b73265e","subject":"Merge \"api: Add schema for share unmanage action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/13086be27244ec9318c3b3641e7055364b73265e"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-12 05:45:45.000000000","tz":540},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6d3c07589db8e8116f4d7a961e3e803e661d72d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6d3c07589db8e8116f4d7a961e3e803e661d72d"}]},"branch":"refs/heads/master"},"1ad036a5e6040e4a4e7762b012718093da23f9e9":{"kind":"REWORK","_number":3,"created":"2026-01-21 04:27:13.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/86/970786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/3"}}},"commit":{"parents":[{"commit":"13086be27244ec9318c3b3641e7055364b73265e","subject":"Merge \"api: Add schema for share unmanage action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/13086be27244ec9318c3b3641e7055364b73265e"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-01-21 04:26:43.000000000","tz":540},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ad036a5e6040e4a4e7762b012718093da23f9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ad036a5e6040e4a4e7762b012718093da23f9e9"}]},"branch":"refs/heads/master"},"1f68105eccaf5fa714e5a01e3aacf3f33870aa25":{"kind":"REWORK","_number":4,"created":"2026-01-21 07:20:27.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/86/970786/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/4"}}},"commit":{"parents":[{"commit":"13086be27244ec9318c3b3641e7055364b73265e","subject":"Merge \"api: Add schema for share unmanage action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/13086be27244ec9318c3b3641e7055364b73265e"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-01-21 07:18:27.000000000","tz":540},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f68105eccaf5fa714e5a01e3aacf3f33870aa25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f68105eccaf5fa714e5a01e3aacf3f33870aa25"}]},"branch":"refs/heads/master"},"84ae5b44eaaab7350c76260c366acb44843a305d":{"kind":"REWORK","_number":5,"created":"2026-02-02 07:42:31.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/86/970786/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/5"}}},"commit":{"parents":[{"commit":"13086be27244ec9318c3b3641e7055364b73265e","subject":"Merge \"api: Add schema for share unmanage action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/13086be27244ec9318c3b3641e7055364b73265e"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 06:56:13.000000000","tz":540},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\nAnd update the response of the share manage API in api-ref.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ae5b44eaaab7350c76260c366acb44843a305d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ae5b44eaaab7350c76260c366acb44843a305d"}]},"branch":"refs/heads/master"},"407d755b19c831554780f3c18a1634cc28f10d5b":{"kind":"REWORK","_number":6,"created":"2026-02-19 01:33:45.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/86/970786/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/6"}}},"commit":{"parents":[{"commit":"9e631a614aae0a1d77a65e63d28ab33dc0c02343","subject":"Merge \"api: Remove references to v1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9e631a614aae0a1d77a65e63d28ab33dc0c02343"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-19 01:31:21.000000000","tz":540},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\nAnd update the response of the share manage API in api-ref.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/407d755b19c831554780f3c18a1634cc28f10d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/407d755b19c831554780f3c18a1634cc28f10d5b"}]},"branch":"refs/heads/master"},"ca849e2fb63ecb4f7ec74d98ed2f63d31b721dd5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-02-19 18:27:32.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/86/970786/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/7"}}},"commit":{"parents":[{"commit":"e1fb16f880853cc0e87d24a0feea3f256b90e3ed","subject":"tests: Remove tests for v1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e1fb16f880853cc0e87d24a0feea3f256b90e3ed"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-19 18:20:35.000000000","tz":-180},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\nAnd update the response of the share manage API in api-ref.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ca849e2fb63ecb4f7ec74d98ed2f63d31b721dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ca849e2fb63ecb4f7ec74d98ed2f63d31b721dd5"}]},"branch":"refs/heads/master"},"6972815d9645fcb4f32bf11911a3197f206b616f":{"kind":"REWORK","_number":8,"created":"2026-02-19 22:09:14.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/86/970786/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/8"}}},"commit":{"parents":[{"commit":"8e0bbbcaef98702af7c080529cf1cd74e56a4112","subject":"api: Remove use of mixin classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8e0bbbcaef98702af7c080529cf1cd74e56a4112"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-19 22:09:02.000000000","tz":-180},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\nAnd update the response of the share manage API in api-ref.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6972815d9645fcb4f32bf11911a3197f206b616f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6972815d9645fcb4f32bf11911a3197f206b616f"}]},"branch":"refs/heads/master"},"55fb7e03e78f0c9fae7f08344d57ab95629a7383":{"kind":"REWORK","_number":9,"created":"2026-02-19 22:22:14.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/86/970786/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/9"}}},"commit":{"parents":[{"commit":"8e0bbbcaef98702af7c080529cf1cd74e56a4112","subject":"api: Remove use of mixin classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8e0bbbcaef98702af7c080529cf1cd74e56a4112"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-19 22:21:23.000000000","tz":-180},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\nAnd update the response of the share manage API in api-ref.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/55fb7e03e78f0c9fae7f08344d57ab95629a7383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/55fb7e03e78f0c9fae7f08344d57ab95629a7383"}]},"branch":"refs/heads/master"},"fa13cec403cc0b254065770fece133b7c9c2c0e3":{"kind":"REWORK","_number":10,"created":"2026-02-20 08:18:19.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/86/970786/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/970786/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/970786/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/970786/10"}}},"commit":{"parents":[{"commit":"8e0bbbcaef98702af7c080529cf1cd74e56a4112","subject":"api: Remove use of mixin classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8e0bbbcaef98702af7c080529cf1cd74e56a4112"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2025-12-01 06:22:10.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-20 08:13:19.000000000","tz":540},"subject":"Allow using mount_point_name when managing share","message":"Allow using mount_point_name when managing share\n\nAdd \"mount_point_name\" option to share manage API,\nso user can mount managed share with human-readable\nexport location as a new junction path.\nAnd update the response of the share manage API in api-ref.\n\nImplements: blueprint manage-with-mount-point-name\nChange-Id: I7bc2ddde5d83e5dd81e0437fa1e745c85f308027\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa13cec403cc0b254065770fece133b7c9c2c0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa13cec403cc0b254065770fece133b7c9c2c0e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
