)]}'
{"id":"openstack%2Fmanila~970985","triplet_id":"openstack%2Fmanila~master~I044b4096815bc6a1a46134c83a175eee45623892","project":"openstack/manila","branch":"master","topic":"compute-req-id","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-15 11:19:48.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I044b4096815bc6a1a46134c83a175eee45623892","subject":"Add X-Compute-Request-Id response header in the middleware","status":"NEW","created":"2025-12-14 14:31:47.000000000","updated":"2025-12-15 11:19:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4413ed6068b2f8f98e152907e19d3cacf337ccc2","_number":970985,"virtual_id_number":970985,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-15 08:55:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-14 16:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40854052e99171e1df8c6fc55fb818e6ee800fe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-14 14:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f6dfbbafff4b0a87c39003ae050f3e66c816972","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-14 14:56:49.000000000","message":"Topic set to compute-req-id","accounts_in_message":[],"_revision_number":1},{"id":"557a20c994acae9a597f1461716642f4fad85653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 16:55:48.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d3e4c08ef7d4127a26f7dad1e594a02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d794ad13f254f7992b03e6c5386f507 : SUCCESS in 8m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70dbc58952a41adac296b0013a2341f : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffd025d2870444eb0e699eacf5655c2 : FAILURE in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62dcb96633f54a14ab9a7b563815cf80 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7fdb5d172384e328a0226d4b42b341d : SUCCESS in 14m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8cba22720320416397e1f1fd14a0a75f : SUCCESS in 47m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d141df6769944cfdb88674662e527fe6 : SUCCESS in 2h 18m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8d12eff2cd3c4b4aadc47413173ab757 : FAILURE in 37m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6f9082a9912847968d42c9ca7126dfc5 : SUCCESS in 26m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9cb5ea1f4a0e42a89a84bf3dab66dea4 : SUCCESS in 1h 38m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d23be5fcc2674a6b834c87b899682691 : FAILURE in 27m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a295b517422d4e09a77f7b63472212e5 : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60c8893d880b4079ac54f6a447f9ff76 : SUCCESS in 4m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4ed9f8d766354ffa9db1a5b215bbcc6a : SUCCESS in 21m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6429013110ca4e81883cdfb436c27050 : SUCCESS in 40m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a2ef651ca8aa478eb609d731588639d4 : FAILURE in 17m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/b1d0d59d48f64eb1a1ea110243be784b : SUCCESS in 1h 17m 00s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5c37fd8ac3904e4ea2a350f7415bc8c4 : FAILURE in 45m 22s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a51fc73f74034776b7011bdbbf352ee0 : SUCCESS in 59m 28s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3fa4f7e4fc9048928073a6fef06d1bc7 : SUCCESS in 20m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3afd633e6a3f4e6a88f441e77610e262 : SUCCESS in 23m 29s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a5135ff8309940d091a3abaf64c59615 : FAILURE in 4m 30s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1b0a6d02ea5c4e6f8067af9a18102019 : FAILURE in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96b91abbbb4e11bc2aa81d24ee940b28c5b08a43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-15 02:56:53.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"235dfc87554e25a396ec9aea7c152915dd136884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 04:53:41.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/193fd3084113457ebb1956028d36da4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ac1de981a2481db538ede2acb313e3 : SUCCESS in 12m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9093bc5a95a74a66a8f11fbe31c600d1 : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39df8a8bb6cd4b42abc7454e271096c5 : FAILURE in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b563968b1834c75856bf1899841f7a7 : SUCCESS in 10m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b63bf89101c446a2a3047eaa09d69773 : SUCCESS in 6m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5d827d497f4b48999ee9d3aa84ee0802 : SUCCESS in 52m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/417f63ad5ed84adfa66fb9ca133ab58a : SUCCESS in 1h 55m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/303d4263b89047f2bbf84163cdda362d : FAILURE in 59m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a3d2f9f2a0ae474493a1aeea449a43ee : SUCCESS in 1h 13m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/32dee3e257de44c8be8bf1d71c823070 : SUCCESS in 1h 41m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0f5b4ab575674780a59ab8bdb6ea75d0 : FAILURE in 21m 14s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b6f0c59a5314420abfd273b5ef5f4f3 : SUCCESS in 4m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29d89a884b694fab8c83944123931af0 : SUCCESS in 4m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/858ad07bced2434893ce4012695dd165 : SUCCESS in 21m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b7fbcaa674724658bdef93222ef80c2a : SUCCESS in 40m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d59eada681040bc85629e231c8bcc22 : FAILURE in 24m 56s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/eff2f2896daa4e2a8d0f26115f806326 : SUCCESS in 1h 18m 27s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cc0a480ebff74c7088bbc964d121f34c : FAILURE in 50m 34s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f386a7345a44fe4b33d2d65b06ca2a3 : SUCCESS in 48m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ce8e68bc4ee94caa9e26f5de2f1cb319 : SUCCESS in 25m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/67eaf36bb5bb44308e1bde6943b53e77 : SUCCESS in 28m 22s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f000bceed7b041458f4c2f171ad3e2a5 : FAILURE in 4m 45s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5f6991fbc82a44f1a99cfdcb1d855b9f : FAILURE in 6m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f8d2d1ae6e12fb109ac4e20898c3873269ff1f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-15 06:40:43.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e7d203aa0a171714c342c779d541a19f40f7b91c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 08:55:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27d6c9cc10be44b190adbda70893b6bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06aab5fb0a44f9593f4531ef3a16d1d : SUCCESS in 11m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/945eac4fefcf4a78bbff4d0ce637a2a2 : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45e3e337dfdf473f8175724ef45b3d0a : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6947801d84a04f8592b933e8c6278447 : SUCCESS in 9m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23228b15e6644bf085f562676830d76b : SUCCESS in 12m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e9e4d72c0ad744c69f7da3ce5fd1b98b : SUCCESS in 37m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9726d7f44d4e43ff86b2dde56bc6c976 : FAILURE in 2h 03m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b17b392ed38e478192666c7f82f1d3d2 : FAILURE in 1h 28m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/eedf4427db0c4fe0bdbef3477c7765d0 : SUCCESS in 28m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/023e893b038a4e8aa498056816785fe1 : SUCCESS in 1h 37m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ffe1b40ddc1a47f9b01add0e1f98ff73 : FAILURE in 14m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1f279467f2ab467ca49271e0ce096062 : SUCCESS in 2m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3a10e4625fe41e5831a4bef61793887 : SUCCESS in 4m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a7c77791211f4c46a0e98516bd04a398 : SUCCESS in 25m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4ceb4b1427ac401eaf0d6cf513d0738c : SUCCESS in 44m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e6aa5850ce314c16916b21bd44815290 : FAILURE in 15m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/697afd59f43c4cd39b9c1ddbece197c1 : SUCCESS in 50m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a4c0dae6177844bba64e18c120bdc7c3 : FAILURE in 52m 19s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a95999713dc481c9bcbfdecb57856d3 : SUCCESS in 40m 19s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/949577db5f70422a8efa3ab078253813 : SUCCESS in 32m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fba4ea2f02f243d99855e2982a327246 : SUCCESS in 17m 18s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/40ab5a5d270c413c89b827aebd174871 : FAILURE in 4m 22s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/50aa84e65d2a4b64bb0f9b98b2f491fd : FAILURE in 3m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"646c49ef70c137ebe383858dd33e39ad71da29fa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-15 09:42:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4413ed6068b2f8f98e152907e19d3cacf337ccc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:19:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0deaf3d86164adc912f059e1a6cfbaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ecf6f028db4cf1b29408ad446fdbdf : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07325d781f0341829b20c337353e7a2c : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e63cb0f53f34e0ea3a27f33335e90f2 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c31081328b72457e8375e2d54ac774a9 : SUCCESS in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a305aa3d8b814a53a8fe627cd0a8fd54 : SUCCESS in 13m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/da348e887f854163ad0300bc5a9c3dd6 : SUCCESS in 1h 23m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/973f8090d5cd41398e878718591183d0 : SUCCESS in 57m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4b85a87aa2e048249a47271546318ddc : FAILURE in 31m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e47f037f0b7e499d98c4d757a7b1ef32 : SUCCESS in 1h 06m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c574ed1e4d2845c094197bba134454a8 : SUCCESS in 1h 31m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/78496db0b1834dba923f7b103b405856 : FAILURE in 24m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1c5f7a01ad034b04ae75ea470930eb12 : SUCCESS in 6m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9dd9878b08ca471f91350af9e8f055c6 : SUCCESS in 3m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f3ef9b16592d4d21b6e670606000a81b : SUCCESS in 24m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2a449cbdcb6c43c790cdace51a205b8e : SUCCESS in 44m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c1d78c85a026403faf4ed5a7a9269b06 : FAILURE in 12m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/3d53ab0263b14374a063194ae73b39a4 : SUCCESS in 35m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a2c50d575e784ee98405415a7d619859 : FAILURE in 48m 32s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/745d9854e3a8426f8f9d4087b716f5d0 : SUCCESS in 44m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/12949408bfe94d17b371516ca1593bc4 : SUCCESS in 15m 20s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8e73296b34d3443d8df89dc7dfc0832e : SUCCESS in 33m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/ce79c7b2a67c4410847e2dc8b130fbf2 : FAILURE in 6m 03s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ec66d817d4f04129a97e3baa51dc4b3b : FAILURE in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"293c5cea5d43aaed0a2d54b2ae2e3b107e3d51ba","revisions":{"2091fa9a86a3421f15deaaa73b059c505d8d4432":{"kind":"REWORK","_number":1,"created":"2025-12-14 14:31:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/970985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/970985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/970985/1"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-14 14:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-14 14:31:36.000000000","tz":540},"subject":"Add X-Compute-Request-Id response header in the middleware","message":"Add X-Compute-Request-Id response header in the middleware\n\nMigrate the logic to add the X-Compute-Request-Id response header to\nthe middleware layer so that the two request id headers are set in\nthe same point.\n\nChange-Id: I044b4096815bc6a1a46134c83a175eee45623892\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2091fa9a86a3421f15deaaa73b059c505d8d4432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2091fa9a86a3421f15deaaa73b059c505d8d4432"}]},"branch":"refs/heads/master"},"12759d0c6a47c99a8278730f20960343b6ff7f15":{"kind":"REWORK","_number":2,"created":"2025-12-15 02:56:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/970985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/970985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/970985/2"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-14 14:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-15 02:56:53.000000000","tz":0},"subject":"Add X-Compute-Request-Id response header in the middleware","message":"Add X-Compute-Request-Id response header in the middleware\n\nMigrate the logic to add the X-Compute-Request-Id response header to\nthe middleware layer so that the two request id headers are set in\nthe same point.\n\nChange-Id: I044b4096815bc6a1a46134c83a175eee45623892\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/12759d0c6a47c99a8278730f20960343b6ff7f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/12759d0c6a47c99a8278730f20960343b6ff7f15"}]},"branch":"refs/heads/master"},"293c5cea5d43aaed0a2d54b2ae2e3b107e3d51ba":{"kind":"REWORK","_number":3,"created":"2025-12-15 06:40:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/970985/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/970985/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/970985/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/970985/3"}}},"commit":{"parents":[{"commit":"6e4c315e706c1e0f95535a71b031354d3147b98e","subject":"Merge \"[DOC] Add export location metadata documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e4c315e706c1e0f95535a71b031354d3147b98e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-14 14:29:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-15 06:40:43.000000000","tz":0},"subject":"Add X-Compute-Request-Id response header in the middleware","message":"Add X-Compute-Request-Id response header in the middleware\n\nMigrate the logic to add the X-Compute-Request-Id response header to\nthe middleware layer so that the two request id headers are set in\nthe same point.\n\nChange-Id: I044b4096815bc6a1a46134c83a175eee45623892\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/293c5cea5d43aaed0a2d54b2ae2e3b107e3d51ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/293c5cea5d43aaed0a2d54b2ae2e3b107e3d51ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
