)]}'
{"id":"openstack%2Fmanila~970996","triplet_id":"openstack%2Fmanila~master~I64bf7242a488515b05a0d80585954651622dc244","project":"openstack/manila","branch":"master","attention_set":{"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2025-12-15 04:56:16.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I64bf7242a488515b05a0d80585954651622dc244","subject":"Implemented reset volume autosize api for Rest","status":"NEW","created":"2025-12-15 03:30:12.000000000","updated":"2025-12-15 04:56:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"43d3e1c72a78bfe1ac59a4d73ca2e20c6c07ab0e","_number":970996,"virtual_id_number":970996,"owner":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-15 04:56:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-15 04:56:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a26599f64559eb07da5ede8c485128c6ea8d4e9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2025-12-15 03:30:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43d3e1c72a78bfe1ac59a4d73ca2e20c6c07ab0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 04:56:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65adfcb6980045d8922e2c1b8a380680\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785d9d83680244c9aad44062e52ef75e : SUCCESS in 11m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4916d0121db542c88093397cb3182806 : FAILURE in 18m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9842def7b3d946f7aaf3af8b1b4f8cad : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5062b68371884ca6ab45f2751eb8866b : SUCCESS in 11m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a97da7ab310a4ae6b967b3c56a09796c : SUCCESS in 12m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6d6474eed3cb429c815ede1eaa6269b1 : SUCCESS in 1h 21m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0be2cc605efd4699ac783959acd80a56 : FAILURE in 5m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9189607b49bf47a0be70938d5145554f : FAILURE in 1h 13m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9f7ee23aa82a463584975a3ae3af98dd : SUCCESS in 31m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/45c450d468fd43aa91c34b9ec55c1443 : SUCCESS in 1h 15m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/48c25015701145c6b22113ca1126d542 : FAILURE in 20m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/389337b050cb44279caf2021e3933b95 : SUCCESS in 5m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05753109a3f945eba272c148652d95ba : FAILURE in 2m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/108554ea5cd242d8a2cb88ba22283136 : FAILURE in 5m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ccc2e06b00342d3a9e3c7f49637acd9 : SUCCESS in 41m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cc25ceb9c7094116a125ea755b73e6c5 : FAILURE in 13m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/c23ed7785a0e4cecb50ebf90c2563f14 : FAILURE in 42m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b7932f5294144e5cbdc54cfd8498f3b8 : SUCCESS in 43m 20s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7ab09850b7c4f9ebf0e238b701b1fb3 : SUCCESS in 55m 28s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/660799bc460143d0b9d7053eca2e5a63 : SUCCESS in 15m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f9d098f353324a34a40c718ce86a09e7 : SUCCESS in 34m 41s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/e61385417bd54f85a32de6ee9ba1b510 : FAILURE in 2m 23s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b557ae919b8e4873bf5ec514dfa75ea9 : RETRY_LIMIT in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05226fb2c4f40e808755a676e68c3b33ab50a071","revisions":{"05226fb2c4f40e808755a676e68c3b33ab50a071":{"kind":"REWORK","_number":1,"created":"2025-12-15 03:30:12.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/96/970996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/96/970996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/96/970996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/96/970996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/96/970996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/96/970996/1"}}},"commit":{"parents":[{"commit":"8fa7281cc6202c693a795635a1633aba0882fe1d","subject":"Merge \"api: Add schema for share force_delete action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8fa7281cc6202c693a795635a1633aba0882fe1d"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-12-15 03:18:43.000000000","tz":-300},"committer":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-12-15 03:29:57.000000000","tz":-300},"subject":"Implemented reset volume autosize api for Rest","message":"Implemented reset volume autosize api for Rest\n\nChange-Id: I64bf7242a488515b05a0d80585954651622dc244\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/05226fb2c4f40e808755a676e68c3b33ab50a071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/05226fb2c4f40e808755a676e68c3b33ab50a071"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
