)]}'
{"id":"openstack%2Fmanila~973789","triplet_id":"openstack%2Fmanila~stable%2F2025.2~I8ee3f84abc295c40d35c201c0360578e56239ddf","project":"openstack/manila","branch":"stable/2025.2","topic":"bug/2130717","attention_set":{},"removed_from_attention_set":{"24993":{"account":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"last_update":"2026-01-23 20:55:23.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-01-23 20:55:23.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-01-23 19:47:25.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-01-21 05:34:18.000000000","reason":"\u003cGERRIT_ACCOUNT_38371\u003e replied on the change","reason_account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-01-22 07:27:21.000000000","reason":"\u003cGERRIT_ACCOUNT_36180\u003e replied on the change","reason_account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-01-23 20:55:23.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-01-23 20:55:23.000000000","reason":"Change was submitted"},"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2026-01-23 20:55:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8ee3f84abc295c40d35c201c0360578e56239ddf","subject":"Fix KeyError in modify_volume when autosize attributes missing","status":"MERGED","created":"2026-01-19 10:59:45.000000000","updated":"2026-02-05 10:39:30.000000000","submitted":"2026-01-23 20:55:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973789-bug/2130717","cherry_pick_of_change":965205,"cherry_pick_of_patch_set":12,"meta_rev_id":"2925205bbd7984fb45248755f7f647e70d13ab7f","_number":973789,"virtual_id_number":973789,"owner":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 20:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"all":[{"value":1,"date":"2026-01-21 05:34:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":2,"date":"2026-01-23 19:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-01-22 07:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":1,"date":"2026-01-23 19:47:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}],"CC":[{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 11:03:26.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-01-19 11:03:26.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-01-19 11:03:26.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-01-19 11:03:26.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-01-19 11:03:26.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-01-19 11:03:26.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-01-19 11:03:26.000000000","updated_by":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"reviewer":{"_account_id":24993,"name":"Carlos","email":"carlos.carrero@veritas.com"},"state":"REVIEWER"},{"updated":"2026-01-19 12:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 10:39:30.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"}],"messages":[{"id":"f89b066416ba8aceaa919baa1d22c030be1de52c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-01-19 10:59:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"797333ff444aef543b3f8ef306abfac000d947e7","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-01-19 11:15:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4451b18d424f0a0bc4c0d85a3e5f0d6c9ebc219","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-01-19 11:21:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f4d8ae20d84ac99f4f8c1a23ec482aef38f4bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:54:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aa3292438d64bea832f5ef2598500a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87bfcecb7d5477c8049554f08dcd265 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1917dcbe00e4c26b236abb2fa39eca1 : SUCCESS in 23m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08060a4e5eff4f0a9fb5f8ddcbef9686 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9711e16609e48c5a6877072469fe62f : SUCCESS in 8m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70733cb8105a40dfba2704a14986137f : SUCCESS in 9m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdd25f28d89c4d08a2f8389a5eecf37c : SUCCESS in 13m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82ff2ddd928f4489a480a069b5c8c719 : SUCCESS in 9m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c06635f6d7a241ab971134edde7bdad0 : SUCCESS in 1h 02m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4159737174e647528e4474f2c1f53826 : FAILURE in 1h 51m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/95a804afc94045fabf82e8251667c86b : FAILURE in 55m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ab7336ff89c143928f5930460a583fca : SUCCESS in 54m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6a269dc5fd964b7c81ddedf4b963c498 : SUCCESS in 1h 35m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e64bfd3e74254605b806fdce11099b2b : FAILURE in 17m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d049797d9d564d7fab17abe631398dca : SUCCESS in 5m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d53ba641eac642b3b0fe97109c4936e5 : FAILURE in 3m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dc51a5fab5e342c18e6d88800f42b6c5 : SUCCESS in 42m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b8889ef75fd45d3ade4cb93f419e79e : SUCCESS in 42m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3f3c0593a374c69a5c92b2bb11affee : FAILURE in 29m 06s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/9d32156bafee41dc8dedf26005f64bba : FAILURE in 19m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0f4cc158f00b4aa985768a3835a83154 : SUCCESS in 39m 02s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/64d73f952f494a34bca268336ddd48b6 : SUCCESS in 39m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4eb97bcb13094dd78fa8fb10e1a68ca9 : FAILURE in 6m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f3c72c35761649a995d9f1378527fe0d : FAILURE in 13m 50s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/d71e8e48d6934c4c8a07670275b6c338 : FAILURE in 4m 42s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0d0adf08e17f43da80d71e1425d46593 : SUCCESS in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a63fcead95cf61d00a431e822e9300a8eeb91773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-01-21 04:52:26.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"25706dd7cf9102b2e3f6cfdaec4a21e5fae01d65","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-01-21 04:52:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a4aeb26acbfa0bf5490af44335d161fce83dee4","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-01-21 05:01:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cecc4a625f349f3d5750b1ca6ce48a6cc2eabc7b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-01-21 05:32:36.000000000","message":"Topic bug/2130717-stable/2025.2 removed","accounts_in_message":[],"_revision_number":2},{"id":"a89a606852b98f89b80114254e5be7a53be8304e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-01-21 05:32:46.000000000","message":"Topic set to bug/2130717","accounts_in_message":[],"_revision_number":2},{"id":"c20f559a5e615f2d1ea885bc185d15e7a231a651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"date":"2026-01-21 05:33:09.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2d2a318fc082b62a33ea0910a2e042cf4b5eba26","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-01-21 05:34:18.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c15ad9ccd250fe6fee024b71c85bf962a156bc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 07:15:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/673d83c13168457bbd7d4c71364e7d58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d661e3efe1a4f7cafe6f00e97cd1f4b : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/910880147670449994b8b1c4dd353667 : SUCCESS in 19m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3c910db46f44149fa82f6ab7a94352 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05e01ac560c842589540081a7429ad51 : SUCCESS in 8m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7532d85fff9b4c55a242e019bc571cc8 : SUCCESS in 10m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/962182881bfd464f9efde2fcccc69489 : SUCCESS in 8m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bcc973c6d0342d99a840809f5e90224 : SUCCESS in 8m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/830da3d3a57742a08b49b637dcadcb4b : SUCCESS in 1h 15m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4599173b832c4a6095509d76a78ad8b8 : SUCCESS in 55m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6ba49de78e984e53831fe05aa2c06f0d : FAILURE in 34m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bb5ca57f2686494593fdc6a69040d959 : SUCCESS in 48m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6877a97cafdc442587ffb8e1ddc85fd4 : SUCCESS in 1h 26m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2c80820296424c88897adb8e03fc7aa6 : FAILURE in 17m 12s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e279112ca87475cbd6e9943402fb4c6 : SUCCESS in 4m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b4976d3517c9414b8464f9d892100a9b : FAILURE in 4m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1aeeafb9857a4da2ae1e953f1d5ec520 : SUCCESS in 24m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ce3f7c1ccce749ac80666a4e7963717b : SUCCESS in 32m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/79c9bfa187ee424397a870624a29d489 : FAILURE in 27m 21s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/64eca94251fd488788ffbbcf500ce8d5 : FAILURE in 9m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2575df1e832c4929bf097db178258690 : SUCCESS in 57m 20s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d0351d8ec564d43831bccc6e1faafe7 : SUCCESS in 43m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/90bdb7eddc534240952f47fdcfa7b295 : FAILURE in 12m 22s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/25a3c4a2e4764fba802137db61bb2e35 : FAILURE in 5m 31s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/7db4295043d94802ab525d155688b0f0 : FAILURE in 5m 42s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/23a51b5f423648c2a63cbfe68cf4149c : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ff2888f5533122f6f7ee29bd560191c1e0af1ae","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-01-22 07:27:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2fbfc2f33e8d7982988b7e5119876858584f54e9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-23 19:47:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"413ac17e43eb583f6e50b297f11f5684f4c5dd6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:47:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"00a534374ecbec34c5a3f8b4827f5681b2056c5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 20:55:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d935d27465a45e3bf40c2fd218a4953\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d63c70be2746779128e62f0ff937cf : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff33c4734244a59aaf6db0bd0097beb : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cf65bac64da41619658ca8642427ba5 : SUCCESS in 7m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd91786a07f9433bb5bf85c745d21a51 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f1baa867c1a4e51804237d3cf5decd7 : SUCCESS in 6m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cbbcca88380d46e19879f6b54bdfda79 : SUCCESS in 1h 06m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74a1bd2ee69b49b4adc198f41acde5df : SUCCESS in 23m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/67b93335c7814886a64e8f9efa4f4458 : SUCCESS in 28m 18s","accounts_in_message":[],"_revision_number":3},{"id":"5ce453113aac46c3af1be153e74104f1991669f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 20:55:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e84f753a1174bffcaf2d044208ac2edce43af224","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 20:56:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d61217dc267b49a0adc23a4561056066\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f145e20e97410c8dab4a7ddb3d9ae8 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/224853e76a1f464e8599708ec1208f23 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3},{"id":"2925205bbd7984fb45248755f7f647e70d13ab7f","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-05 10:39:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/d73d4c21147e4886a31aac9ca4a15b1a : FAILURE in 1h 08m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cb033344fa0d658401880a98bc3cd920de4ca89c","revisions":{"9098f4880efc8e7f6e13f7b5b47badc664e915c2":{"kind":"REWORK","_number":1,"created":"2026-01-19 10:59:45.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/89/973789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/973789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/973789/1"}}},"commit":{"parents":[{"commit":"d9638f279012e6a8c6262593e5f6974695339f95","subject":"Fix shares getting stuck in ensuring status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d9638f279012e6a8c6262593e5f6974695339f95"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-10-29 09:42:35.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2026-01-19 10:59:45.000000000","tz":0},"subject":"Fix KeyError in modify_volume when autosize attributes missing","message":"Fix KeyError in modify_volume when autosize attributes missing\n\nThe test \u0027test_snapshot_before_share_replica\u0027 failed due to a KeyError\nfor \u0027mode\u0027 during replica promotion. The modify_volume() method in\nclient_cmode_rest.py accessed keys like \u0027mode\u0027 and \u0027grow-threshold-percent\u0027\nfrom the autosize_attributes dictionary without checking if they exist.\n\nDuring replica promotion, _update_autosize_attributes_after_promote_replica()\ncalled modify_volume() with a dictionary missing the \u0027mode\u0027 key, resulting\nin a KeyError.\n\nThis change updates modify_volume() to check for the existence of each\nautosize attribute before accessing it. The function now gracefully handles\npartial autosize attribute dictionaries, including only present attributes\nin the request body.\nCloses-bug: #2130717\n\nChange-Id: I8ee3f84abc295c40d35c201c0360578e56239ddf\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n(cherry picked from commit 4dc5390ed04caddea8e34730ab12b1452531f645)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9098f4880efc8e7f6e13f7b5b47badc664e915c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9098f4880efc8e7f6e13f7b5b47badc664e915c2"}]},"branch":"refs/heads/stable/2025.2"},"bacbba54d58c0a7bbb114876467519882f5244fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-21 04:52:26.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/89/973789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/973789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/973789/2"}}},"commit":{"parents":[{"commit":"d9638f279012e6a8c6262593e5f6974695339f95","subject":"Fix shares getting stuck in ensuring status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d9638f279012e6a8c6262593e5f6974695339f95"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-10-29 09:42:35.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2026-01-21 04:52:26.000000000","tz":0},"subject":"bug/2130717: Fix KeyError in modify_volume when autosize attributes missing","message":"bug/2130717: Fix KeyError in modify_volume when autosize attributes missing\n\nThe test \u0027test_snapshot_before_share_replica\u0027 failed due to a KeyError\nfor \u0027mode\u0027 during replica promotion. The modify_volume() method in\nclient_cmode_rest.py accessed keys like \u0027mode\u0027 and \u0027grow-threshold-percent\u0027\nfrom the autosize_attributes dictionary without checking if they exist.\n\nDuring replica promotion, _update_autosize_attributes_after_promote_replica()\ncalled modify_volume() with a dictionary missing the \u0027mode\u0027 key, resulting\nin a KeyError.\n\nThis change updates modify_volume() to check for the existence of each\nautosize attribute before accessing it. The function now gracefully handles\npartial autosize attribute dictionaries, including only present attributes\nin the request body.\nCloses-bug: #2130717\n\nChange-Id: I8ee3f84abc295c40d35c201c0360578e56239ddf\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n(cherry picked from commit 4dc5390ed04caddea8e34730ab12b1452531f645)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bacbba54d58c0a7bbb114876467519882f5244fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bacbba54d58c0a7bbb114876467519882f5244fc"}]},"branch":"refs/heads/stable/2025.2","description":"Edit commit message"},"cb033344fa0d658401880a98bc3cd920de4ca89c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-21 05:33:09.000000000","uploader":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"ref":"refs/changes/89/973789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/973789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/973789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/973789/3"}}},"commit":{"parents":[{"commit":"d9638f279012e6a8c6262593e5f6974695339f95","subject":"Fix shares getting stuck in ensuring status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d9638f279012e6a8c6262593e5f6974695339f95"}]}],"author":{"name":"medhac1403","email":"medhaj06@gmail.com","date":"2025-10-29 09:42:35.000000000","tz":-240},"committer":{"name":"Medha Choudhary","email":"medhaj06@gmail.com","date":"2026-01-21 05:33:09.000000000","tz":0},"subject":"Fix KeyError in modify_volume when autosize attributes missing","message":"Fix KeyError in modify_volume when autosize attributes missing\n\nThe test \u0027test_snapshot_before_share_replica\u0027 failed due to a KeyError\nfor \u0027mode\u0027 during replica promotion. The modify_volume() method in\nclient_cmode_rest.py accessed keys like \u0027mode\u0027 and \u0027grow-threshold-percent\u0027\nfrom the autosize_attributes dictionary without checking if they exist.\n\nDuring replica promotion, _update_autosize_attributes_after_promote_replica()\ncalled modify_volume() with a dictionary missing the \u0027mode\u0027 key, resulting\nin a KeyError.\n\nThis change updates modify_volume() to check for the existence of each\nautosize attribute before accessing it. The function now gracefully handles\npartial autosize attribute dictionaries, including only present attributes\nin the request body.\nCloses-bug: #2130717\n\nChange-Id: I8ee3f84abc295c40d35c201c0360578e56239ddf\nSigned-off-by: medhac1403 \u003cmedhaj06@gmail.com\u003e\n(cherry picked from commit 4dc5390ed04caddea8e34730ab12b1452531f645)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cb033344fa0d658401880a98bc3cd920de4ca89c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cb033344fa0d658401880a98bc3cd920de4ca89c"}]},"branch":"refs/heads/stable/2025.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
