)]}'
{"id":"openstack%2Fmanila~975372","triplet_id":"openstack%2Fmanila~master~I567a3bdbd2650e6b90c7a0f74c30f4d075056f6a","project":"openstack/manila","branch":"master","topic":"bp/manage-with-mount-point-name","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-26 10:05:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-02-26 11:37:52.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-02-26 11:37:52.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-26 06:50:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-02-26 11:37:52.000000000","reason":"Change was submitted"},"36741":{"account":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"last_update":"2026-02-26 11:37:52.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-02-26 11:37:52.000000000","reason":"Change was submitted"},"15595":{"account":{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"},"last_update":"2026-02-26 11:37:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I567a3bdbd2650e6b90c7a0f74c30f4d075056f6a","subject":"[NetApp] Fix mount format while managing share","status":"MERGED","created":"2026-02-02 07:42:31.000000000","updated":"2026-02-26 11:38:52.000000000","submitted":"2026-02-26 11:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"975372-bp/manage-with-mount-point-name","meta_rev_id":"6d5c76d8bffa44f3dd3876394457acfb0b99f976","_number":975372,"virtual_id_number":975372,"owner":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-26 11:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-26 10:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-02-26 06:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-02-26 10:05:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"}],"reviewers":{"REVIEWER":[{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 09:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-02 17:54:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-04 06:31:57.000000000","updated_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-02-04 06:31:57.000000000","updated_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-04 06:31:57.000000000","updated_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"reviewer":{"_account_id":15595,"name":"Gireesh","email":"gireesh694@gmail.com","username":"gireesh"},"state":"REVIEWER"},{"updated":"2026-02-04 06:31:57.000000000","updated_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-12 11:21:37.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-19 01:39:49.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-19 20:18:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-02-19 22:23:51.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2026-02-25 14:47:31.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2026-02-26 10:05:58.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"c778e842baebbbed18ee8cad73bc9b2cb2389d13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-02 07:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a1a5feabe8a99f1d5ec55ca2dee8b4d990a61e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 09:25:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d84bbb1c06484dc78e313be25e123c3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7053786cdfc545c39e821d7e7861a3b6 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1848a4cbffc4665960510aa25de517d : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab9c9fb2bc804ef79aee8ac118ee517c : FAILURE in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/410cd060dc67499f9218c7c79fda0958 : SUCCESS in 10m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3181981e3bd047f48bcc1a7108f800ae : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34cda5fbb48842fb8c12cfb7dbf4e53f : SUCCESS in 5m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bc552a424266492b8095cc8b8b3a421f : SUCCESS in 1h 05m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/67cc8d537585422f9906c6310cdeba50 : SUCCESS in 1h 31m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8d4d8ed861843b28ba98941bc47b4cc : FAILURE in 1h 09m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/53f330599c8d4fa5a7c585c01e1a8e8c : SUCCESS in 41m 38s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/65001d15917c440095efb11bb0cc20f6 : SUCCESS in 1h 33m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/967d3f3b47294db7ab609ec8356b2700 : FAILURE in 1h 04m 05s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4c5e99f6f5ca46c79a174026e20e110c : SUCCESS in 3m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de039e9e069843a49172ac61ae85f619 : FAILURE in 7m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04c9c33406454040b08e735c3a227fca : SUCCESS in 18m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8ee267ef96d04db4b9cd971e2f3e90bf : SUCCESS in 31m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5ca82a2463ba4b379f6aec33a1bca00a : SUCCESS in 24m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5a89d6fb4eb54d709e957b10f8a8141a : FAILURE in 40m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/6649948119804153869c247e3341d775 : SUCCESS in 49m 35s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a7ddc66dcc6647b395ab6a6178e82f7b : SUCCESS in 1h 00m 07s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e7b47065fb649318b66f3ce6f640516 : SUCCESS in 56m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e618356b96b04ba384c71e47707708f2 : SUCCESS in 31m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4a7f861d028a40b3ba461d81d9f70858 : SUCCESS in 36m 33s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/bfc6e383806c4bae9bf18c88b1e13aa2 : FAILURE in 5m 32s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/88a0374ac1c14e25866b0a6efa779dd6 : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00717d4d587e9ea479da015a1204faeb6be3152b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-02 17:54:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5f93c5af09fbff30e5743d6d17b5eb84a9ba88aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-03 04:40:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9907683a510171f090ad5e0cde3a9d1a5836af1b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 05:58:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"921db0c91019b24e203a4fbe128f8679546fe557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 06:21:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/605db9acf2c04aecb38b30ceb800aa63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02460a4398d44f21b909d5115c96d3ed : SUCCESS in 9m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66c0f52d169b481d8a546168bc662966 : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dee78fc59aa44458744ccd17c3fea70 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0193f520366b4388a9e3455b839bd746 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dd3168cc7424c14b18f9af997de08d2 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/288b54ab81274544824e320af06edba7 : SUCCESS in 5m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/546d1bd48ac847f984d516034145f82f : SUCCESS in 35m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c95752ecc555465aa9e70bf1ae8188e0 : SUCCESS in 57m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5b64579921814910ac802cbcaa304952 : FAILURE in 21m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/323e288bfaa2435c9894cf9f640ec93e : SUCCESS in 37m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/790a85fc93884261a4d7bccaf103609a : SUCCESS in 46m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3c5235aed2424a5e81febfd2b5080797 : FAILURE in 1h 38m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3873d7ef89ab4663b0f95c5bd936edbe : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d11caf4a6e1f4e4c805c0959c6242313 : FAILURE in 2m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7ca98748c795465e960167423c0c432e : SUCCESS in 20m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6f4499591b3348a19289c6118425e051 : SUCCESS in 19m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2fe71f7223cc4bbfaa2cafa1a42e249e : SUCCESS in 42m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a88fd906aa5940419accb6251b5afd12 : FAILURE in 32m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/d139fabaf99d467497bae64ddbf7e08c : SUCCESS in 37m 18s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2cdf0ac12d264e33b389981ab25163a5 : SUCCESS in 38m 48s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d620eb1580946db86f88f1ad9478fda : SUCCESS in 1h 04m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dba349e451b849b6ad9bc99be557e611 : SUCCESS in 14m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e9689579366c406e83797de87796cd6e : SUCCESS in 34m 20s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a006476c9f134fd3821057a92a07529e : FAILURE in 2m 27s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1cfb77cf94ba4c6881cf5abed638bb45 : SUCCESS in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cfb5487705cd9997e83ef578d22503f31eaf7583","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-04 12:02:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c864674627e172d9bc1baf0fcbd43dd5e7882570","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-02-05 05:24:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc873f372cdc253811c6add9d2a38c068f1a331e","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-02-12 11:21:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9af0b7562cf8e654d8d0e6a174ba19772c1120f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-19 01:33:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1fbd6c84bc3dc598a2bc6d40b85020810107f0d2","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-19 01:39:49.000000000","message":"Patch Set 3:\n\nBuild Succeeded: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/975372","accounts_in_message":[],"_revision_number":3},{"id":"38a23c54bdd28428130d2d89a7680104477857a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 03:56:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5334b4bca574c7484722c67804bb945\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a27b5b61e997434dbdb22c285c93e644 : SUCCESS in 8m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90245fecce464aeabb3c65200ed33773 : SUCCESS in 16m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d6fff5b74840b4a9231cff5a29899d : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03a14089dd4b4cbda8f5c232c712b78a : POST_FAILURE in 7m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ac49909c38448baacea51946aac7a2b : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1d24d873efa4491a9ef03f8618f4cde : SUCCESS in 6m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b511dbd40c1d4b97957189b37e13bf95 : SUCCESS in 48m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/237d76710539423faea5c76f415f1212 : FAILURE in 2h 15m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af2de9b959904f9a90c323abcde00a52 : FAILURE in 29m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/805bbb52c864415b9967cebb44057c3c : FAILURE in 40m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0024c4e0954a4416a80ab8d02ec38144 : SUCCESS in 1h 11m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5f6f6c8bcfab4366a33ae7bab907e416 : FAILURE in 35m 38s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fce48101b2ed49b6908dbbb6e65db796 : SUCCESS in 4m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60897c2e457944a9afc0acfd66862dc6 : FAILURE in 2m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9dc92d1500f44405aef4731b755e989b : SUCCESS in 27m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/146ed92a3a8e4e97867391e2606815fc : SUCCESS in 38m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3b4e594f4a32455cbb2f63fbec881023 : SUCCESS in 44m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ef5f4c4aecae4be4922d94de33f0d44e : FAILURE in 1h 07m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/65c5b3615b194fcfbdaf630685afb66d : SUCCESS in 55m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/03549a1c8eca4b07824abb144a9eaaa5 : FAILURE in 54m 49s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/072337403f234114a0eea3711320f7d2 : FAILURE in 6m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b997262f56e8404f9a0ed6df1e78fd4b : SUCCESS in 33m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7da1a825a87544009930af86b656128e : SUCCESS in 35m 14s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/73714ba4cbef41e1920f16411510d34f : FAILURE in 4m 24s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/62e33999d6344ba4bfad6ab10260ebf6 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4b24e1db9f456b39ac40900aa8a2e106339499c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 17:51:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e28e91b44763cb02a40150bf91cb0bc2c26da5ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 19:57:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af309da4c9ac4209b473ffa3b49c5aed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2189991b7b79461fbda9a4cb559077d3 : SUCCESS in 11m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0564d1948a14994bcbb64c9dd0f175e : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5e48d3fd5f439b8a73ba4d694ca4a6 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18406c37335546c5b726fcaebc607410 : SUCCESS in 7m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcd243d1f1354277b04f842ea7a40389 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6560c76eac654844abfcf7f331998eb9 : SUCCESS in 6m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/23bcceb104e44fef9a4becca3a5ca4b7 : SUCCESS in 1h 07m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d1755d9dfd324554a33c44ce2b623aa3 : SUCCESS in 2h 00m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/22552755c1d6464089f87980224aa689 : FAILURE in 1h 01m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4b32a6198d0546b99702e0584127ee3c : FAILURE in 1h 00m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8bb90ac492d14ad1af660a32ded5d0bf : SUCCESS in 1h 28m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d74b982f5754434f8c599f2506faa60a : FAILURE in 1h 22m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fddc608e36484629804650c9f1a4ea05 : SUCCESS in 4m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e58354d941df4acb96b1470d47338c04 : FAILURE in 4m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b23567db904a4a8eb76da4bc90e08d32 : SUCCESS in 37m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d42bdebb4cf4467ab47f712f0bba14e4 : SUCCESS in 30m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4ba6f0984af841a5ae263b9fb7d74530 : SUCCESS in 35m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/94f949ea496d404fb61988bd3e82b6e6 : FAILURE in 1h 00m 03s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/5f33ab558b364864b520306b47f2cb1b : SUCCESS in 1h 14m 18s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cbff7473bc2e48e894d216198f3603e2 : SUCCESS in 46m 44s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a89fce31c9f7418e8face8aabcbad54e : FAILURE in 5m 10s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b3a9e9d218d348ba9552c01fcbbd7b2e : SUCCESS in 32m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/bb0816b264b14d6a8a7645f44f3895f1 : SUCCESS in 34m 52s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a33aad6f01954ab6bbd73ed73a6c182d : FAILURE in 4m 14s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/66a60319c10140a6a0b38056cc2933d3 : SUCCESS in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9ff610ac0e6cb71b5f472d31be76731d8440369","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-19 20:18:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/72/975372/3/upstream-check-manila/manila-tempest-plugin-ontap-dhss/026b347/ : FAILURE in 13m 01s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/72/975372/3/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/27df1bc/ : FAILURE in 13m 01s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"a0c43a026eea4591550ad0fdf91415879176441c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"real_author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 22:23:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_36741\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"}],"_revision_number":4},{"id":"e85850a394bd73ee129f21ebd78d09587b96ea70","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-19 22:23:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"844996cae77ff9b03f4c0c5e265de1440f922555","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-02-19 22:23:51.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/970786 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"bd958643204e96475b80c1d6011e89eb0f3dd649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 00:20:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f7a7b6f10f94ceabebb2d676c26ccaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4600b99e6f49a99a9cdfb616922dea : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccf189b95d104b86a52b36bfffc7b6c4 : FAILURE in 14m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fa8088dd0c4d76920fa6917b5bc248 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1679862dd9946c2885e53ab00a15a83 : FAILURE in 11m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4df0d85861cf4af2b92c2364796979ee : FAILURE in 5m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59fea14fdceb4ced89be8be14fe341eb : FAILURE in 7m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d0dbcdfe9c4d2eab50fb2eaf6c1ad3 : SUCCESS in 6m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b628cafccb2e42549e7a1a6f5cc74444 : SUCCESS in 1h 32m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/46d209ca7afa4cbdb71bbe150d1b7032 : SUCCESS in 1h 54m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8d2407e483ad4ed693c4b65856ad4635 : FAILURE in 47m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1f7660e9871c460f98c3642a3874370e : FAILURE in 41m 31s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3ff77afd11f3436e8684b80f26489f30 : SUCCESS in 1h 35m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0137246caebd42969519b17098ac1c29 : FAILURE in 1h 15m 19s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc3e791ba1954b4fac3ce16f5193a97a : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89202a0a81464263b3129540e81a2f09 : FAILURE in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c50c6f22e1dd4a5e92c0c020bb762465 : SUCCESS in 31m 04s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1bc6b9cd060048799f61132ec156f39b : SUCCESS in 26m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/984098eb1aae4bf9804af3d137b30f3c : SUCCESS in 44m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4b6b34c720cd40c1bd6162ab93e9e8d7 : FAILURE in 27m 56s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/da30608278994c40ab35b8e1c1ae6548 : SUCCESS in 32m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/759fd93164694e4ea668893fe527ffd8 : SUCCESS in 57m 18s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/545b20e988d34d0a97bad29a06fa9ed5 : FAILURE in 12m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bae3a22c87b74be080f09bf5eed97d02 : FAILURE in 13m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2dd2394d26224138b112d0a563a3a763 : FAILURE in 29m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/105a13b961ab4c45bdf4d6bf2ff46062 : FAILURE in 4m 27s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a905db355b494ccc8404a5b270f293c4 : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a37e42ed57b5a04bd7df490b9d5cbd1b92de512b","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-20 02:14:11.000000000","message":"Patch Set 4:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/975372","accounts_in_message":[],"_revision_number":4},{"id":"be8491fbc504688dddeba3bf2e1e7cbba1b8e559","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-20 18:07:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24c65b026cd3c55a62f4ad486325c73cf0a8662c","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-22 22:36:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cebb531423029f5ff781c4c8b70854f6b95c1804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 00:05:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38a1d2536e7a4861b64d545b83b1c352\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a518a634a1420f8d1ba32682c4e805 : SUCCESS in 13m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c640a5735b2c487fb8fef3804ce1807b : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d742450a04874e00a82bb0b3868aeae0 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b425cfda37d74dd68ef2ebaf97edfbae : FAILURE in 8m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31bc9bea2bba47b2b3bb7a84dbb8b82d : FAILURE in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b0268b43b2a54ac28a7eb249d9293236 : FAILURE in 12m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd7ec0e5af9246c5bdf471ff9f574f73 : SUCCESS in 6m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/44a8b2c248d142f3a3b959dbdd659d96 : SUCCESS in 37m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cb68dcaaf71c447884ce6c0ce03b1a04 : SUCCESS in 56m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/31d81c88d37d47ae9f08bd30ca870acb : FAILURE in 49m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d73ccda1a8a44eca802cd95fdab9cc06 : FAILURE in 53m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0095b65c328842629dd8ca8c48c86545 : SUCCESS in 1h 24m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ea2d8a64001446cba9d5180c95da5191 : FAILURE in 38m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8810905d74a04bc599ecf8bd40159626 : SUCCESS in 2m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/505e9a4f8d9048729b7730e2b9586290 : FAILURE in 6m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4651335de686493cbb1abda3385282ab : SUCCESS in 23m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/181be6c85c024aef8e43682b39d0b721 : SUCCESS in 19m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c226a821643643759dd52f2cc2932336 : SUCCESS in 26m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/70921174cd944fb1866e0bae5f46b524 : FAILURE in 58m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/ef95d72e341a4115aa411bf1645ad1c6 : SUCCESS in 31m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5099ab94fc5f4ba28559a6022812bdcb : SUCCESS in 54m 21s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21037906f9394c91968437b67d4625ad : FAILURE in 4m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/981072ad895b4555909884a8d265a23c : FAILURE in 30m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d8d7e3289b4a4a2ca8f6f5fbbdde28f4 : FAILURE in 36m 17s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f54bab5984424f3f93e8fb03461420c7 : FAILURE in 5m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/6b61bd19632447c787f71ed64f16e19e : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2539cb7502d1e06ed2dfd3c91eba24c10cfa4bf3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-25 14:47:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5a6c14e9419f803181d86c2b88176625b5e78c83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-26 03:22:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"127c80a481394c277be8c6aa2beae50c5c319fb7","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-26 04:35:20.000000000","message":"Patch Set 5:\n\nBuild Succeeded: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/975372","accounts_in_message":[],"_revision_number":5},{"id":"79225a384a247d27b232ca7605eb0a627fe4c81b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 05:07:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cb8f7e105ac47c5a5a39ea5b991ac22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da23eebc10ed4d199ff328a9f57ab534 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0256129c3fd4faf99f5d22baf2407bf : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e017c34d0e884db1a629b232de879321 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0193c21554cc4c929922c7c7a095ed49 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c530fbe4e894d6fabe476eda42ad69d : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d5cfbb57473a4454abbe06c21f3d0893 : SUCCESS in 6m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/863c4d95d14847508fe622452c9db6d3 : SUCCESS in 10m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e5c1bf6619544af8974fca854a500a21 : SUCCESS in 40m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/36374d632bb040a2919502e91bf62c5c : FAILURE in 1h 41m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b32cb7e9a85047b4a494a9843635a609 : FAILURE in 41m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/aff438ee1bf6479792624861431e0a04 : FAILURE in 29m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/96b42121f13b450cbf0304e30af8e508 : SUCCESS in 45m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c0d082a766a947228cc032eea86d1cc8 : FAILURE in 1h 08m 08s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/01db9b0c0038445ca34f91f312a5d2d9 : SUCCESS in 5m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e96754b158e8459d9d671e99b44f63ee : FAILURE in 3m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fc310861a3fd4fc1b27f31623c5240c4 : SUCCESS in 22m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/377a71fb071b4eb3b37e558fb03a4c1d : SUCCESS in 21m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9463eecbf05d401abbc299f3eea1488f : SUCCESS in 30m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5687af74e94547538e3357c71214e54f : FAILURE in 1h 00m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/b7cf3e6102f449c1a43d16c86fb03083 : FAILURE in 10m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e368d23802404ee89451861f111c98c5 : SUCCESS in 40m 35s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76e66ba07dc549348f7b81d3e82cf1f7 : FAILURE in 6m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/818f81e4edcd417c928777e94b1a20dc : FAILURE in 22m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7dde1e3888d04430ad7559324a510b86 : FAILURE in 30m 34s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/39f448e8baab4c4890464649fc6cec0d : FAILURE in 4m 30s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/07bc179b55144b0f894379dbea560ff9 : SUCCESS in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a0c323ce800ccb4fa27221c245785eb0cc8a910","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-02-26 05:18:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b150199dea34904ef76fad64e1dac6fcae280ad9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-26 06:50:37.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9b0a058168a701e20e0c48fe8ecfba737c223a89","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-26 07:01:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/72/975372/5/upstream-check-manila/manila-tempest-plugin-ontap-dhss/2c015d4/ : FAILURE in 1h 19m 45s\n- manila-tempest-plugin-ontap-dhss-rest finger://managesf.services.openstack.netapp.com:7979/9e829d78625841709214d2f7799c987b : DISK_FULL in 1h 50m 59s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"893bdec64a01f492f7e7103a9e1cef12b4b48102","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-26 10:05:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7474471a5f68a7d020242f14b033f389f4985931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 10:06:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7baf5d316d2bceb72be0fc38c9af72a03d59f1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 11:37:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40a7b5a6abb4576b82902c774d7e8e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94aaedaf18874d829bed98a70c9ff237 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0788d6debfb94814b8711f5fd6b873f8 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a367ffcd02c74aef9ee8a547d1ddbb4c : SUCCESS in 11m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90a772cd6087493f9cff857bb899eac3 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bda521084a95419bba6760054a743480 : SUCCESS in 9m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/80a4f8006ef9472982ccf9126f1d3c22 : SUCCESS in 1h 11m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/158c4d146ab845f38c1d54f4322af2b1 : SUCCESS in 33m 16s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8a76536ffe3a40eeb241c08973bbd830 : SUCCESS in 35m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a540fba7698241a98ea7023d27ac1107 : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":5},{"id":"ff48d86391635e7c66fe7c90c105ffb8fd35c754","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 11:37:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6d5c76d8bffa44f3dd3876394457acfb0b99f976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 11:38:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41198bbf9812448da6a68d5f60d06f5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe5ba69674a4a9fadc2077e2ecc03bd : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/265a783fc1484b7d8e6edfe071a18c20 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2c1c36b4fe04af4093eaa1789f486ace4caaf261","revisions":{"1022ae312b37462fc53e4bdd2bb7a564063f6ac6":{"kind":"REWORK","_number":1,"created":"2026-02-02 07:42:31.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/72/975372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/975372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/975372/1"}}},"commit":{"parents":[{"commit":"84ae5b44eaaab7350c76260c366acb44843a305d","subject":"Allow using mount_point_name when managing share","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ae5b44eaaab7350c76260c366acb44843a305d"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 07:40:28.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 07:40:28.000000000","tz":540},"subject":"[NetApp] Fix mount format while managing share","message":"[NetApp] Fix mount format while managing share\n\nWhen managing share on NetApp driver, to use \"mount_point_name\"\nvalue as new junction path, added the \u0027/\u0027 before mount_point_name\nwhen calling NetApp\u0027s volume mount API.\n\nCloses-Bug: 2129743\nChange-Id: I567a3bdbd2650e6b90c7a0f74c30f4d075056f6a\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1022ae312b37462fc53e4bdd2bb7a564063f6ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1022ae312b37462fc53e4bdd2bb7a564063f6ac6"}]},"branch":"refs/heads/master"},"9afd47dd7a750f9ff9d25c2b5875ac109cd3fb12":{"kind":"REWORK","_number":2,"created":"2026-02-03 04:40:44.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/72/975372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/975372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/975372/2"}}},"commit":{"parents":[{"commit":"84ae5b44eaaab7350c76260c366acb44843a305d","subject":"Allow using mount_point_name when managing share","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ae5b44eaaab7350c76260c366acb44843a305d"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 07:40:28.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 23:56:55.000000000","tz":540},"subject":"[NetApp] Fix mount format while managing share","message":"[NetApp] Fix mount format while managing share\n\nWhen managing share on NetApp driver, to use \"mount_point_name\"\nvalue as new junction path, added the \u0027/\u0027 before mount_point_name\nwhen calling NetApp\u0027s volume mount API.\n\nCloses-Bug: 2129743\nChange-Id: I567a3bdbd2650e6b90c7a0f74c30f4d075056f6a\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9afd47dd7a750f9ff9d25c2b5875ac109cd3fb12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9afd47dd7a750f9ff9d25c2b5875ac109cd3fb12"}]},"branch":"refs/heads/master"},"66e94c253c6cb6ffbe2db66f6234dc37125c1fca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-19 01:33:45.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/72/975372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/975372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/975372/3"}}},"commit":{"parents":[{"commit":"407d755b19c831554780f3c18a1634cc28f10d5b","subject":"Allow using mount_point_name when managing share","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/407d755b19c831554780f3c18a1634cc28f10d5b"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 07:40:28.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-19 01:31:21.000000000","tz":540},"subject":"[NetApp] Fix mount format while managing share","message":"[NetApp] Fix mount format while managing share\n\nWhen managing share on NetApp driver, to use \"mount_point_name\"\nvalue as new junction path, added the \u0027/\u0027 before mount_point_name\nwhen calling NetApp\u0027s volume mount API.\n\nCloses-Bug: 2129743\nChange-Id: I567a3bdbd2650e6b90c7a0f74c30f4d075056f6a\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/66e94c253c6cb6ffbe2db66f6234dc37125c1fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/66e94c253c6cb6ffbe2db66f6234dc37125c1fca"}]},"branch":"refs/heads/master"},"fef0023fe25a9258b46fa2db1cef25988d021df5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-19 22:23:31.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"real_uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/72/975372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/975372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/975372/4"}}},"commit":{"parents":[{"commit":"55fb7e03e78f0c9fae7f08344d57ab95629a7383","subject":"Allow using mount_point_name when managing share","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/55fb7e03e78f0c9fae7f08344d57ab95629a7383"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 07:40:28.000000000","tz":540},"committer":{"name":"Eunkyung Kim","email":"ek121.kim@samsung.com","date":"2026-02-19 22:23:31.000000000","tz":0},"subject":"[NetApp] Fix mount format while managing share","message":"[NetApp] Fix mount format while managing share\n\nWhen managing share on NetApp driver, to use \"mount_point_name\"\nvalue as new junction path, added the \u0027/\u0027 before mount_point_name\nwhen calling NetApp\u0027s volume mount API.\n\nCloses-Bug: 2129743\nChange-Id: I567a3bdbd2650e6b90c7a0f74c30f4d075056f6a\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fef0023fe25a9258b46fa2db1cef25988d021df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fef0023fe25a9258b46fa2db1cef25988d021df5"}]},"branch":"refs/heads/master","description":"Rebase"},"2c1c36b4fe04af4093eaa1789f486ace4caaf261":{"kind":"REWORK","_number":5,"created":"2026-02-26 03:22:59.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/72/975372/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/975372/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/975372/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/975372/5"}}},"commit":{"parents":[{"commit":"4913cac9d133dacb313cabd42b2758543fed6e63","subject":"Merge \"Follow up on API v1 removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4913cac9d133dacb313cabd42b2758543fed6e63"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-02 07:40:28.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-02-26 03:20:52.000000000","tz":540},"subject":"[NetApp] Fix mount format while managing share","message":"[NetApp] Fix mount format while managing share\n\nWhen managing share on NetApp driver, to use \"mount_point_name\"\nvalue as new junction path, added the \u0027/\u0027 before mount_point_name\nwhen calling NetApp\u0027s volume mount API.\n\nCloses-Bug: 2129743\nChange-Id: I567a3bdbd2650e6b90c7a0f74c30f4d075056f6a\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c1c36b4fe04af4093eaa1789f486ace4caaf261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c1c36b4fe04af4093eaa1789f486ace4caaf261"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
