)]}'
{"id":"openstack%2Fmanila~976128","triplet_id":"openstack%2Fmanila~master~I6674751c4aab195d638c84fcd228f068b51bccc0","project":"openstack/manila","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-12 14:59:47.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-12 14:46:18.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-12 15:59:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6674751c4aab195d638c84fcd228f068b51bccc0","subject":"api: Remove unnecessary aliasing/brackets","status":"MERGED","created":"2026-02-09 14:53:17.000000000","updated":"2026-02-12 16:01:01.000000000","submitted":"2026-02-12 15:59:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976128-openapi","meta_rev_id":"30c87460d0a73513635dd0ac2c06ef1290be01e2","_number":976128,"virtual_id_number":976128,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 15:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-12 14:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-02-12 14:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-02-12 14:59:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-09 16:42:01.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-09 17:07:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 11:44:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-11 11:44:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"38e4a5dae6b79938768edac4e09623aaec90db56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-09 14:53:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daa112ce20b91cf10648722ff92fe9f255b0a179","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-09 16:42:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/5371dd71f8324d2aa2f6f176c6629762 : FAILURE in 49m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8bdd8bfa94515fb451b6f13b1299ff47957fa733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:07:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a0380d644884b18820f7946308bb850\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a85147ca6f74b4eaadfda7c92caebc8 : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bc5d4bc1a9e46bb970ee4873fb09080 : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79ee5f9bf02a4150abe77d73198d63ff : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24a363bd6bf94521bd493655309e60ea : SUCCESS in 9m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ddabc2e13d5400eac2d6fbbaeea6649 : SUCCESS in 4m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/968ebac320a149538f206a90013aa3a4 : FAILURE in 19m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5d997fcfd53544a98da49781fa9f0df1 : FAILURE in 2h 06m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bf3b3e8a84314729a98bf70961e11e48 : FAILURE in 42m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/be7dcef79cc64cc2b401f1dfad1de4eb : FAILURE in 1h 00m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bf42dc523c394dbfb295de8b7c653d13 : SUCCESS in 40m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b60d3c92f8148b9bb0da115e0b73762 : SUCCESS in 1h 16m 40s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f75fc7cec6024f8e8c47d5b59b317a5b : SUCCESS in 4m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d423762704cb46b9af361e1086e9c946 : SUCCESS in 2m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a938154b466743ccadc5d523e1711449 : SUCCESS in 19m 07s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dac24124c197429fa3e6b3a5876f5517 : SUCCESS in 31m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/662517224ad04ebfb513241525b91a1b : SUCCESS in 53m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f5eef0f6c0604a5a923ed74a3e6dcdeb : FAILURE in 32m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/0fb6d96e2e9d4de78302f0cd5a79de87 : SUCCESS in 1h 11m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e077a99fd8714116837a5ea5a65bd33e : FAILURE in 13m 37s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b26c2d3f815d40148b9cb5f9b6ec6482 : FAILURE in 33m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bfb7133acf4840e5b44031ea8b243f29 : SUCCESS in 33m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/972e040909a0436a8274ee3f05c0ebb8 : SUCCESS in 39m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/e80d36d7d43b45f084a70af3b7bab5a7 : FAILURE in 5m 01s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a8b263cca4794f699bd411c336172d58 : SUCCESS in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99d0281839104d3268d2316e30f34f9df6d427e7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 11:41:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc37926cf74f729307a6a6703887533495b40762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 13:25:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95878d4dd4ed48428df97ac95bdffe14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03339b63fc4e430eac3e5663f2b4369d : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97786783b2044fa387d1cbcb62a73854 : SUCCESS in 32m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8b6bfadb894b438efa3c89ae1b3f2a : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e61d95a34e2c4e679e1ba301eb1eeaff : SUCCESS in 10m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcfe85ded2414de09bd710c9a8c47199 : SUCCESS in 4m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a2025c608f0b4a84879c2926d9388057 : SUCCESS in 1h 24m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d716e00c053547c6abcdc6b37ad5049b : SUCCESS in 1h 41m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4d4f12261d03494885daeff695d880d5 : FAILURE in 40m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/073180fd2186446db57b422445f51009 : SUCCESS in 1h 20m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/22e49e3e68a941eb99585196464a68cc : SUCCESS in 1h 19m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/23fbfbfa27e142e1871233552295c1b4 : FAILURE in 1h 11m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7d990e9ba4494577a7c6d18f4c64a142 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/537ddc6cace54608bd5f97c6b40c53f5 : SUCCESS in 6m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74e7de8ce2c04e8a84d941270ff69ec0 : SUCCESS in 27m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8857752185c440e798ab5168b6d29263 : SUCCESS in 28m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eb73aa7a9c0c4117b44a181251f11f24 : SUCCESS in 32m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ab5742205bc84300bf7059d32b440db0 : FAILURE in 36m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/cc5bbe2e0835459e893c2d69297d707d : SUCCESS in 43m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/244f8b8531a9473c935783550119c4a3 : SUCCESS in 37m 06s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40a26d37c43d41d9b4297e89df4e2759 : FAILURE in 41m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f3cee37ad2bf4fbcbfc2a3d2a8c07daf : SUCCESS in 16m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/288b3b4484fb4c948fc3dbfd742c870b : SUCCESS in 23m 06s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f138c8663f774f868e37d0956d6e939b : FAILURE in 4m 30s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/6a582dc9bcc94d32953fa5e76361b502 : SUCCESS in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"675eb7df9ce33672c1b3fabd4e18e1dc0abcd952","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-12 14:46:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e3e4767e2e0ab360452affb18301e17e6543317","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-12 14:59:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2f6cfe1cebf00ef0f3da9c971972cc1d1623d59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:00:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a2bd4d22da8b40cd49fe8d20cc958a33dd4d80e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:59:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8dfb1a8b1a0492096649736bd4f3f51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3204e597c24f5fb74d1ec4fe605df5 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a701f899a54db2af5cbc61ed218a63 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c123076976348a88baa0cffd6a7896a : SUCCESS in 8m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bc8babd3d254013a0a83db1c1e663ae : SUCCESS in 5m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cbc5b6ba89634a08a22d723af76ad1a2 : SUCCESS in 37m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/28e4802163fe494792736286e7ee4348 : SUCCESS in 34m 53s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9b42195a2710492bb41c57938e1ba600 : SUCCESS in 34m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9e9cbb02c5e04445a9b4257b2ccee8c9 : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f40b78adf327334843fba961e94372b2366f5d6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:59:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"30c87460d0a73513635dd0ac2c06ef1290be01e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:01:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34f828e69a849d5b890c400154e0929\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5105aecfdb924b61adc9e00a0028f7bf : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a707a7b8ff151dc9c29b525c7051f8b167f8f1a","revisions":{"5a707a7b8ff151dc9c29b525c7051f8b167f8f1a":{"kind":"REWORK","_number":1,"created":"2026-02-09 14:53:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/976128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/28/976128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/976128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/976128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/976128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/28/976128/1"}}},"commit":{"parents":[{"commit":"1cea1c65be37b6a68890bba19c0bb2d12670cdfe","subject":"api: Fix schema for quota class sets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1cea1c65be37b6a68890bba19c0bb2d12670cdfe"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-09 13:40:36.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-09 14:53:08.000000000","tz":0},"subject":"api: Remove unnecessary aliasing/brackets","message":"api: Remove unnecessary aliasing/brackets\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I6674751c4aab195d638c84fcd228f068b51bccc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a707a7b8ff151dc9c29b525c7051f8b167f8f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a707a7b8ff151dc9c29b525c7051f8b167f8f1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
