)]}'
{"id":"openstack%2Fmanila~976129","triplet_id":"openstack%2Fmanila~master~Iba6de2d3e2ddb4ae5d73694f01367261371a5e92","project":"openstack/manila","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-12 16:00:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba6de2d3e2ddb4ae5d73694f01367261371a5e92","subject":"pre-commit: Bump versions","status":"MERGED","created":"2026-02-09 14:53:17.000000000","updated":"2026-02-12 16:01:56.000000000","submitted":"2026-02-12 16:00:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976129-openapi","meta_rev_id":"fa2d1c11ac67ec881a173d89e645f413cff2127c","_number":976129,"virtual_id_number":976129,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 16:00:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-12 15:00:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-02-09 18:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-02-09 18:39:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-09 15:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-09 17:00:32.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-09 18:39:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-12 15:00:04.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2026-02-12 15:00:11.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"a890963409a695e687f2ed410b7b7fa9f995f7cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-09 14:53:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b85f79e7a6396046365a1f0e5cf9167e29180a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 15:12:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ba3d7e40d14c91ac8659bbf39784a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9690eaf5a94da8bfe7b509314627cf : SUCCESS in 9m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60c12f621d79481aa0885a5e79473c27 : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92895c7f59d4656af78541f1f40784c : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a82b9ded1e84f51b1c95e99f327a034 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1ecb9bea9b8490fba5b95661fa8df33 : SUCCESS in 8m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0fe6c1c934ad494c9269a7ae422ce674 : SUCCESS in 2m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/deafaf40a02d486a82e1553ef50f4c51 : SUCCESS in 2m 29s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4606934c3b974fb998bc6b5d0ea74e3d : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dbfe6d14ae5feb96aa4dd9db6ac16bf52ec2690","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-09 17:00:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/f1da49604de747fd9fbd2fcd2f0a5cb3 : FAILURE in 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6527e3f72de37e8b20c6b8c12f16786dca6b982f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-09 18:39:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c226d8693fc67d516d4f3e89add4dbcec60de345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:00:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e432e0720753cf1f94a5531ac960faad32bef7ed","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-02-12 15:00:04.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/976128 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"736efbdc8eafaec9b461b84c4c1f9015498f3cb0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-12 15:00:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64412e270b2cd5bd26a435eca6b595ca7109849e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-12 15:00:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b8b6de34671fea32e2508cc332a129a6a03ecb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:00:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7b0cff764704256bbdeea5054db7e96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb671e033bee4d979eacfaf4ca16a2af : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3be91868ca4ee591c84a4207432d7f : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46e1b72004684fd69b8a51cee4186371 : SUCCESS in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f359f4c9cb0148e8b98fd249ed73adf0 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8d7979f79ab10c04e73d16d44a9f028e58a7a117","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:00:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa2d1c11ac67ec881a173d89e645f413cff2127c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:01:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54937e0d46f476e94bdb1a4b93c9f7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96ab582824c4d6e9e54105c57600bc7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"574fc1bdee9d897c9bc9ac9dfbd95b2b01978ec1","revisions":{"574fc1bdee9d897c9bc9ac9dfbd95b2b01978ec1":{"kind":"REWORK","_number":1,"created":"2026-02-09 14:53:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/29/976129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/29/976129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/29/976129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/29/976129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/29/976129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/29/976129/1"}}},"commit":{"parents":[{"commit":"5a707a7b8ff151dc9c29b525c7051f8b167f8f1a","subject":"api: Remove unnecessary aliasing/brackets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a707a7b8ff151dc9c29b525c7051f8b167f8f1a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-09 13:43:58.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-09 14:53:08.000000000","tz":0},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nThe older version used are not compatible with the default Python on\nFedora 43.\n\nBumping hacking (and by extension flake8) brings up 3 new issues that we\nresolve here.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Iba6de2d3e2ddb4ae5d73694f01367261371a5e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/574fc1bdee9d897c9bc9ac9dfbd95b2b01978ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/574fc1bdee9d897c9bc9ac9dfbd95b2b01978ec1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
