)]}'
{"id":"openstack%2Fmanila~976299","triplet_id":"openstack%2Fmanila~master~I3ba06b96130b9120b91150e063a8ac3df54ac76f","project":"openstack/manila","branch":"master","topic":"bug/2141463","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-10 22:08:35.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-03-10 22:08:35.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-06 20:52:21.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-03-10 22:08:35.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-03-10 22:08:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ba06b96130b9120b91150e063a8ac3df54ac76f","subject":"[CephFS driver] Make share deletion idempotent","status":"MERGED","created":"2026-02-10 20:03:57.000000000","updated":"2026-03-10 22:09:40.000000000","submitted":"2026-03-10 22:08:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976299-bug/2141463","meta_rev_id":"2864eb62dcf7bc1a53897c92e5ba6f762a6829e7","_number":976299,"virtual_id_number":976299,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-10 22:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-03-06 20:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-03-10 13:13:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-03-10 20:54:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 20:40:53.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-10 21:34:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-10 23:57:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2026-02-12 17:51:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-05 14:55:25.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-03-05 14:55:25.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-03-06 06:43:57.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-10 12:55:23.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"}],"messages":[{"id":"6e102b8ff00d343ed178a6d263b5c4d3520ac942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-10 20:03:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93075875bb38ab392e643216b06defa92fcfef75","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-10 20:40:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/4de8dcd5b63c414fa4429b497afc7d17 : FAILURE in 36m 29s","accounts_in_message":[],"_revision_number":1},{"id":"86b934025323b35047e9b9d8fb4a4599ce5ac4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 21:34:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/199f96ccfed7499e8116b1ee28209534\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3ad71611248d4477bc1e1a8446bbaadf : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c450cd31edd8404fa7042f79cb752b05 : SUCCESS in 14m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad2aa454eff24738940a1b12f8883090 : FAILURE in 30m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c970b7dc4d0e46f69c9a1a0c7d42636d : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2db7cc1aa9864b879e0e2c87d925ef53 : SUCCESS in 8m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db9b908b8040467484c2138bc6ca631a : SUCCESS in 7m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fbca6746ff944af8863933027c3fd37b : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a17768b595e14fe9b5fbf2086714f66b : SUCCESS in 5m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bd932cfa2e7a4f2dac7ad2ed6d704fb9 : SUCCESS in 32m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/57a6909edc834a3384cfd52c8659da6d : FAILURE in 56m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0ad839c736d340c3b9d2b2f5ea77ca02 : FAILURE in 28m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/02c0032c740b439d9df3d5dfcea248c8 : SUCCESS in 1h 14m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8062ff2f73984829839bacd51cd16cbd : SUCCESS in 1h 22m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bcf105f5025348329e169e1b86a89f01 : FAILURE in 44m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/37bbb32d5c8f4360946b5ad2bfc76549 : SUCCESS in 4m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9a14c2c379040f6a082129951022a85 : SUCCESS in 4m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/46e9f5e592be46058243ba48c6edf821 : SUCCESS in 33m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/10e417e2dae545d8b80cb07b1c4a7b40 : SUCCESS in 20m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2cc8c642f67249439a1091275d807d82 : SUCCESS in 37m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7764d5e34b5346c4aa011ea2719f70ff : FAILURE in 58m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/aa49ac83fb304ce3a488b33caa9f38fa : SUCCESS in 1h 15m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/08724f8e19cb49fcaceed778096945f2 : FAILURE in 13m 23s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00e1839c574147ad931e8e63c094b39c : FAILURE in 15m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1306e92692944d0da815a7c77fb67fff : SUCCESS in 32m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/32d664bee6044bff9ba574f24c459b2c : SUCCESS in 16m 04s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/676b9e88468b4e55aaa1b298a6387291 : FAILURE in 4m 44s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/27416b491b644c1e985d9cf8ea84fe03 : SUCCESS in 3m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06bba032e7a87f6ef873f3e27e05905808afb3b2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-10 23:57:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24e05e49f85b00a6d19d4715166318ad0545ef93","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-11 14:14:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17f9f1482ff1945bbb4d52f941a541d61606c783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 16:48:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d3486c66cce4c68a7a1b404bd8aa860\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/21b7d9765d8f44faa050fa880fd9b46f : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0113be11c4c14da997d1c6799a6be841 : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acd5d3110a494254969a83e3880fd9a5 : FAILURE in 17m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb1d7d36d6544439dd3d0c9f937332d : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cc1b9c256c24bc78d2734d78288d71c : SUCCESS in 5m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8573e706d36b47b4aac388f0bf9608a2 : SUCCESS in 8m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3ab17ef2f1d42f083d667c9c85ab47c : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5deb674f5d44e7b497d5a58db4afe8 : SUCCESS in 6m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0de89fd5e27b41b8a7efcf9a49901144 : SUCCESS in 1h 14m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2967d9ac9dc54e918e35edf03dfa8c0e : SUCCESS in 1h 59m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c00cca13d0e943d8a99cfaa756e43767 : FAILURE in 1h 22m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/fa26190dc0194ae2bd78d2af42c5311e : SUCCESS in 45m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dc5373100f464b79bbbd2e9673002789 : SUCCESS in 1h 48m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/17c4636763e6456c8b72cd352536284c : FAILURE in 1h 00m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b8e692261a749978f88199d5e4b3422 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/348cc611b69c4669a63a8099f742e466 : SUCCESS in 4m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6a9947f42771406ab8baf06616b2fd01 : SUCCESS in 35m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cad80f03872a404cb816944fa9600f51 : SUCCESS in 19m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/64ce0e3051104b5488a0cb4754fa9625 : SUCCESS in 41m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/96d52431ceb045e4994572f26fb160b6 : FAILURE in 32m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2601045968654adcbec1e98ee90c0271 : SUCCESS in 43m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/db6ccc68c10e4a51889d3ee60839a66b : SUCCESS in 2h 31m 22s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b862df45bb8a453283843e65718b9888 : FAILURE in 26m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3c07d9175d584b83a02ccab32463f6ed : SUCCESS in 26m 20s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/308777ee94b44dc78d96e25102f0b483 : SUCCESS in 16m 48s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/7e85426067704161a05d10f91c10e438 : FAILURE in 5m 42s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/832ae643908e434689f489f03ba0600a : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee859e79783ab6f869602a6d9a4c41ca5946181c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-11 17:27:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"648358079687dc2e8f23c4002bb6123529e33de4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-11 19:29:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f7adde944084fe529b0fd72f3337673285be13b0","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-11 20:08:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/286d97dcba114f9eb35e8fc02464b01b : FAILURE in 37m 52s","accounts_in_message":[],"_revision_number":2},{"id":"9766717df55a0c0769314d4f474205508f2d8585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 20:48:58.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e93a6156aa54a6a84c635ef2f5511fe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/886e8f9e03c24287a70e39862f6f5aff : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2a967b9ced4eadb3a035b979734914 : SUCCESS in 13m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adba490dc03d40f9993b107848e1e65b : FAILURE in 26m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa3582e5c29d4d3483e587e850367935 : FAILURE in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a7a5a60c6f0461a82f8defdbba4e35d : SUCCESS in 10m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a53eac31ea84a4caf6d48662310fd59 : SUCCESS in 8m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/641abfe1635449ed92ce95a4c5c872d4 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb421a7905c04628bcaeeb939dc829b7 : SUCCESS in 9m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1fd0bbfb959a48168ef8d2ec7b08f560 : SUCCESS in 1h 10m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7f7e044316a64076b90150ed16541f10 : SUCCESS in 51m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/211f3ed1554f4a9c9b3e46d209c86e79 : FAILURE in 1h 09m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7c64fbf46bf84a2ca282d7e3a773fbf2 : SUCCESS in 1h 13m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6823a647943b49f1ad118838495ddc4c : SUCCESS in 40m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/65b1fd03d23042fc996f6f4adbac2c4b : FAILURE in 49m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e09c95129d364ec8a07b79d4b26ac084 : SUCCESS in 4m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc9bc2796fd64ffcb0193941837a111d : SUCCESS in 5m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8cf4f5bcf7eb4aa69edd00896f81a295 : SUCCESS in 35m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/07a12a64458949dfb99678a2c5de90c9 : SUCCESS in 33m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6eed73452d8742809bfac6daab652c67 : SUCCESS in 35m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0721e648ae384f83b1798a8489410308 : FAILURE in 1h 01m 06s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2f237474645e475ca035736de3a6feea : SUCCESS in 1h 13m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/237d89508c2c4fa5b88cfd93029accc3 : SUCCESS in 42m 15s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2c6cda80f674cbdb13bc0c7e7ddabf7 : FAILURE in 26m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cfdc25b34b4a4bf4b09e35357252ae90 : SUCCESS in 34m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6a0f85b73ea24c8da567b0e79eb4be4a : SUCCESS in 36m 12s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/8973bf466771411c98939a63aedac155 : FAILURE in 4m 08s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3717ce785e3b4a88a98be68eddfaf77d : SUCCESS in 7m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77fc4d2d5157425357c5233629b5efe31ec50121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-12 15:48:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d6c7153ab3059529d294f0477189413105e440eb","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-12 16:25:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/1fbf99d637b24392a89af8d8a5dbf86f : FAILURE in 36m 57s","accounts_in_message":[],"_revision_number":3},{"id":"630e5b65ceb7b2681ceff1ee465ffc81ad92e364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:43:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88ec634b63b44fa4b754716c1ed9bb97\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/26759f60535d48f490b572f9f5e35898 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6769ac49b7ef4b36a11c84d6613e4448 : SUCCESS in 11m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ff9819e7c15445bbc1f33023f76c9cf : SUCCESS in 21m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef4be360e9484cc0958eb959f3e387c8 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c244fe796c3d480d88ada9fcf68e373e : SUCCESS in 9m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e348461beae490a906955dd313af55b : SUCCESS in 8m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6611ab2c8e34663b7962779dffc19c6 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14c36dd37f7945adbbefce3e7d342880 : SUCCESS in 9m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f706715d71624b4b9b6fe3dbf8ea9e3c : SUCCESS in 1h 13m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dc87ae4caf5345b4b0417f6df2897cfd : SUCCESS in 1h 47m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2025f467dff14aefa2b10d1dfc7157ad : FAILURE in 1h 34m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7b880b84afe84c25a6680897006597cf : SUCCESS in 45m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/841b4f497b4d4cbc958eae83f2a68b23 : SUCCESS in 1h 31m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/772b3aa5764b406e83d915c633e11071 : FAILURE in 1h 19m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c6074492137a40e8a752cf718729f3ef : SUCCESS in 5m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e59f9ae31e05452bb78ca60f78686f83 : SUCCESS in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/783fa60e5cef407fb86e779eabda2bba : SUCCESS in 28m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5b9038a65dee476fa6b47a91fec8bef5 : SUCCESS in 22m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/64a2d648dabe4e3880f09fa783cdb92f : SUCCESS in 34m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3b70fb91168546d98111271724d593a2 : FAILURE in 28m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2af75179ae794658877f46fc172e7e9f : SUCCESS in 1h 16m 56s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b3927c2ad8c3471d88690cae4e7be35b : SUCCESS in 43m 34s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83cf5781d68a4a80bfd0ba222890069c : FAILURE in 34m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3a6c46b4314e45c5aa5cd99fe9fc76b5 : SUCCESS in 21m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2eb691cf821d49359ee7b64f258e27ec : SUCCESS in 36m 22s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b46fcb4d636b41bdacd14c153d0fdace : FAILURE in 5m 13s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/58865a4dec024d7f87f1a90030526e35 : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"215fbe91bd5457b1e10b28d06d787ab912eeb907","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-12 17:51:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/99/976299/3/upstream-check-manila/manila-tempest-plugin-ontap-dhss/c2fdb03/ : RETRY_LIMIT in 7m 02s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f3304321efd683b7b00e26e8e13b95bad2539f59","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-03 19:02:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3752982bc2387ca31ee1776764692f846b484aa1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-06 06:43:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7961a4194929c12af2641dbcc9a75cd704a0cea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-06 19:03:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"83dc6dcb173e07ad33b9fa34df213129361f7696","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-06 19:04:29.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9effc892c88e7b0f0d758ca593a719295b352bf4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-06 19:54:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"00a6da634419fdb96591448acf362bfda7aacdef","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-06 19:55:22.000000000","message":"Patch Set 4:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/976299","accounts_in_message":[],"_revision_number":4},{"id":"f49d7dec1f110fcdc8be145f8c1b1a0a8edbdd80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-06 20:42:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"772fcd7278e3190c001cf78ac18d120edc520bd8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-06 20:43:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3482467130292cb85a2144211455a6cba8a60273","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-06 20:52:21.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c1ebfc0482d9e207b8c0ebddd5310c283eaeae8a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-06 20:52:25.000000000","message":"Topic fix-error-deleting-cephfs-driver removed","accounts_in_message":[],"_revision_number":5},{"id":"4b68cc99715d45dd607394262251b720067eb9d3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-06 20:52:29.000000000","message":"Topic set to bug/2141463","accounts_in_message":[],"_revision_number":5},{"id":"5f7ee0110a6f455f88093e0cb5d08c7ba02572e1","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-06 21:56:12.000000000","message":"Patch Set 5:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/976299","accounts_in_message":[],"_revision_number":5},{"id":"8e006d0ec8d7b7498f60bbce143ca003d761a16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 22:50:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f8aaab536240f8be89c8b9b1c1c05a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b092711134b64020bce29a3d5904cd2b : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d5e2b4dca304ff397cdc0c10604c999 : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3178a5c96a2e465b95a3cdbe064084ac : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d0c27c4e62c432bbeef0183cff06f1d : SUCCESS in 10m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a209ee23cbe40538379c74bc88bdae9 : SUCCESS in 8m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4ea0c4167634641b7172283f037e555 : SUCCESS in 8m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae8b2e707881429fa43d0aa61643d2b0 : SUCCESS in 8m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/22b2d2fabcda476eb3b4768cccab399e : SUCCESS in 33m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/59e6b7807d6c4f81b67e7e9e20d5659e : SUCCESS in 2h 02m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/34cb379f6a6145569b3950972b515f32 : FAILURE in 19m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a521b7d456b9444691fbab5d69f2b7a5 : FAILURE in 1h 05m 01s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8e5eb6dfe7a84b98a093f7999c119bc5 : SUCCESS in 47m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/810ba45c574e43ceb53fb2d98bb34287 : FAILURE in 1h 18m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6bfeb03b20a840cc820c4a14e5bcaf02 : SUCCESS in 5m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c2fdacdec2a4ddda00e52248e8e5a9f : SUCCESS in 4m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/55f6221ee02e4fe4a0d8fa50c3c76aef : SUCCESS in 40m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2c87a169590943e3b48b7cc3c3087585 : SUCCESS in 24m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e204293302d74539bacbac1b228d656d : SUCCESS in 52m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9ca37189c0b84599b053925a7794f3f3 : FAILURE in 1h 02m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/2c547a74ae9142e99f80720e6d6aa451 : SUCCESS in 1h 18m 28s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7ba9e294bfef4cac90526d64c3878539 : SUCCESS in 57m 27s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94e499ef35c24701aa043e3bad3761a1 : FAILURE in 7m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f2a6966c57164b799358d7875f4df9c3 : FAILURE in 30m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/187781fe3c564443b909e2235a2b804a : FAILURE in 23m 06s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/cd82a0414c7c450b97d2ca629296d79e : FAILURE in 5m 25s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7ada908e734b4ec4bb80c82023ec5f3f : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"601b56bdb84ca5d591f787fe9a595ab9acc9b7a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-06 22:59:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/99/976299/5/upstream-check-manila/manila-tempest-plugin-ontap-dhss/9f5a7ff/ : RETRY_LIMIT in 7m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"bc0ec3d38c5533b5931a381e83b8bfb0e84863d4","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-03-10 13:13:00.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8a20fceecc79e0be065f9f0caee06cb0bdbf7e45","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-10 19:08:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"724d4a55fc5ea39f60c75ab2c8ab48c31f9935e4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-10 20:54:33.000000000","message":"Patch Set 5: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0858bc78e78a796e7b60cafba1466be0f5ffc898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 20:54:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"371de536ed10e7f8412ebc8971a61a8b4ca255ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 22:08:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265b9e9abd654b4b92bea5045c312d0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb84b0c2ea5d44e8ba8a1fd29c08f258 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/879058cbf69946be8c85f869b07a5167 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f5c20f18d8e46f0ae4c023e9b87cce3 : SUCCESS in 7m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23d77a55026c4e0c84b05f4f05388c63 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a5ee2c41ea44dba8e323a45c85fd65c : SUCCESS in 9m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e37a39ddf3ff451dba345ec54f0f77d4 : SUCCESS in 1h 09m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/60f321faa1d249b0b201cd7ea21c6a02 : SUCCESS in 38m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ee9653744d2c4ae2a968184a7e624f4d : SUCCESS in 33m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6b5cb44f3a6342589f1a6e44123f495e : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":5},{"id":"59c2a98be038e3ad410d8c43b7644f27f4da5622","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 22:08:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2864eb62dcf7bc1a53897c92e5ba6f762a6829e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 22:09:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ada001e98b64b06ab7717f5e3d1a909\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e7966c9c6c4df5baffad1b773dcf67 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa7b238e21cd4378a027f8a5b881323e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b398ffdd7fec2d88cd1c2637c323118df9d75653","revisions":{"9ab3111621ba760d6c717ed24a031e3761f379a1":{"kind":"REWORK","_number":1,"created":"2026-02-10 20:03:57.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/99/976299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/99/976299/1"}}},"commit":{"parents":[{"commit":"ed4b05c514124ab897c10c88c5474b152d7908c2","subject":"[doc] Warn admins about share type mutability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ed4b05c514124ab897c10c88c5474b152d7908c2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-10 19:55:43.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-10 20:03:42.000000000","tz":-180},"subject":"[CephFS driver] Make share deletion idempotent","message":"[CephFS driver] Make share deletion idempotent\n\nThe CephFS driver is failing to delete a share that does not exist\nin the backend.\n\nThis change adds a try/catch block and catches the not found\nexception.\n\nCloses-Bug: #2141463\nChange-Id: I3ba06b96130b9120b91150e063a8ac3df54ac76f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ab3111621ba760d6c717ed24a031e3761f379a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ab3111621ba760d6c717ed24a031e3761f379a1"}]},"branch":"refs/heads/master"},"fa1c77652c41e9420f251067eebc9fa607842640":{"kind":"REWORK","_number":2,"created":"2026-02-11 19:29:48.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976299/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/99/976299/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/99/976299/2"}}},"commit":{"parents":[{"commit":"ed4b05c514124ab897c10c88c5474b152d7908c2","subject":"[doc] Warn admins about share type mutability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ed4b05c514124ab897c10c88c5474b152d7908c2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-10 19:55:43.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-11 19:15:26.000000000","tz":-180},"subject":"[CephFS driver] Make share deletion idempotent","message":"[CephFS driver] Make share deletion idempotent\n\nThe CephFS driver is failing to delete a share that does not exist\nin the backend.\n\nThis change adds a try/catch block and catches the not found\nexception.\n\nCloses-Bug: #2141463\nChange-Id: I3ba06b96130b9120b91150e063a8ac3df54ac76f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa1c77652c41e9420f251067eebc9fa607842640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa1c77652c41e9420f251067eebc9fa607842640"}]},"branch":"refs/heads/master"},"66948e72542b8bd782c0996977c5894afcf956d5":{"kind":"REWORK","_number":3,"created":"2026-02-12 15:48:05.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976299/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/99/976299/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/99/976299/3"}}},"commit":{"parents":[{"commit":"ed4b05c514124ab897c10c88c5474b152d7908c2","subject":"[doc] Warn admins about share type mutability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ed4b05c514124ab897c10c88c5474b152d7908c2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-10 19:55:43.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-12 15:47:57.000000000","tz":-180},"subject":"[CephFS driver] Make share deletion idempotent","message":"[CephFS driver] Make share deletion idempotent\n\nThe CephFS driver is failing to delete a share that does not exist\nin the backend.\n\nThis change adds a try/catch block and catches the not found\nexception.\n\nCloses-Bug: #2141463\nChange-Id: I3ba06b96130b9120b91150e063a8ac3df54ac76f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/66948e72542b8bd782c0996977c5894afcf956d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/66948e72542b8bd782c0996977c5894afcf956d5"}]},"branch":"refs/heads/master"},"e0953f39631d23107ed405866f3a64b254a74e2d":{"kind":"REWORK","_number":4,"created":"2026-03-06 19:03:37.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976299/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/99/976299/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/99/976299/4"}}},"commit":{"parents":[{"commit":"ed4b05c514124ab897c10c88c5474b152d7908c2","subject":"[doc] Warn admins about share type mutability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ed4b05c514124ab897c10c88c5474b152d7908c2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-10 19:55:43.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-03-06 19:03:26.000000000","tz":-180},"subject":"[CephFS driver] Make share deletion idempotent","message":"[CephFS driver] Make share deletion idempotent\n\nThe CephFS driver is failing to delete a share that does not exist\nin the backend.\n\nThis change adds a try/catch block and catches the not found\nexception.\n\nCloses-Bug: #2141463\nChange-Id: I3ba06b96130b9120b91150e063a8ac3df54ac76f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e0953f39631d23107ed405866f3a64b254a74e2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e0953f39631d23107ed405866f3a64b254a74e2d"}]},"branch":"refs/heads/master"},"b398ffdd7fec2d88cd1c2637c323118df9d75653":{"kind":"REWORK","_number":5,"created":"2026-03-06 20:42:50.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976299/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/99/976299/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/99/976299/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/99/976299/5"}}},"commit":{"parents":[{"commit":"ed4b05c514124ab897c10c88c5474b152d7908c2","subject":"[doc] Warn admins about share type mutability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ed4b05c514124ab897c10c88c5474b152d7908c2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-10 19:55:43.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2026-03-06 20:42:49.000000000","tz":0},"subject":"[CephFS driver] Make share deletion idempotent","message":"[CephFS driver] Make share deletion idempotent\n\nThe CephFS driver is failing to delete a share that does not exist\nin the backend.\n\nThis change adds a try/catch block and catches the not found\nexception.\n\nCloses-Bug: #2141463\nChange-Id: I3ba06b96130b9120b91150e063a8ac3df54ac76f\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b398ffdd7fec2d88cd1c2637c323118df9d75653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b398ffdd7fec2d88cd1c2637c323118df9d75653"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
