)]}'
{"id":"openstack%2Fmanila~976320","triplet_id":"openstack%2Fmanila~master~I0fb5babc8ada9fa14f85115192e250c28dec9f2f","project":"openstack/manila","branch":"master","topic":"driver-docs","attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-11 00:28:09.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-02-19 07:30:50.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"38607":{"account":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"last_update":"2026-02-11 10:11:15.000000000","reason":"\u003cGERRIT_ACCOUNT_38607\u003e replied on the change","reason_account":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"}}},"hashtags":[],"change_id":"I0fb5babc8ada9fa14f85115192e250c28dec9f2f","subject":"[doc] Document share driver interfaces","status":"NEW","created":"2026-02-10 23:39:58.000000000","updated":"2026-02-19 07:30:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":16,"unresolved_comment_count":11,"has_review_started":true,"meta_rev_id":"9fdaaf6efdf78c2b5d5b42a6d992be3643dfdab2","_number":976320,"virtual_id_number":976320,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-11 01:44:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"}],"CC":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 23:51:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"state":"REVIEWER"},{"updated":"2026-02-10 23:51:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-11 00:28:09.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-11 01:44:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 11:43:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"}],"messages":[{"id":"edb709b715d3e65e0f96c6ff6891ada5c79a5eb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-10 23:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c84c15271f39dfaf173eb6725ebf3e0c2729e116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-10 23:50:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"655559ce22ca409d38b103cf30133ad1fecbf5f5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-10 23:51:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28b0613aa7336fe0f8dc17e5ae8bd8853de886f3","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-11 00:28:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-alletramp-manila-file-driver build/0db4af8c0c6c4cbfbcaf35afb435d937 : FAILURE in 36m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0e04f569ee4233d368f3536713a11064532ec269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 01:44:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d17de6ade74c63b95e625df22051d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3f947bebebd473cbb672ce6a9417ce8 : SUCCESS in 12m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4825927ab0a7499a859c829a81736524 : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4505dd9f3d64d0aa913e7e55fc9b1e0 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6fc5dd6e1db4122b13ca3c46156cfc9 : SUCCESS in 7m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45d3ce7a0dac4cb38fe46af0e5730694 : SUCCESS in 7m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f078373510314c5982dc16f3fae04be4 : SUCCESS in 1h 19m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c72d368a67044a738f0754bdc264c17a : SUCCESS in 1h 51m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1a3182ee81e54c86b09c4f4d7c6ea82e : SUCCESS in 1h 10m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/53c35dce702842cd918c128eb9afefcf : SUCCESS in 1h 02m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d4f215cbcc804df0a68a292eab031b30 : SUCCESS in 1h 20m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/083f74abcd9e412c985fc6d7736f032d : FAILURE in 34m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f454c6a6baaa41cdaddc8eee1967f7e1 : SUCCESS in 2m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0201d21f7316474da1d258d5e894cebd : SUCCESS in 4m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a9255231a63749419e3986638f2df490 : SUCCESS in 21m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9464d45c36664ca3ac6f05a3f30c4847 : SUCCESS in 19m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7f75790d6dcd4476985ca004dc0af97b : SUCCESS in 27m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/23e35abb02af48e48cf769b0f9e0ccfa : FAILURE in 1h 02m 39s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/883329be9bdf46329652ab813e4ca606 : SUCCESS in 1h 14m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/29bdb63731874187a8306efe5728ef2f : SUCCESS in 52m 21s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/137130dba99f4b4a892995849b6d4ce7 : FAILURE in 32m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e6b526c4e9bb4ac4a52977d579442cdf : SUCCESS in 31m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/19e69e214ce3453d835551b9bc8d8a72 : SUCCESS in 19m 25s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/aba822220ca6459dbf0fcbbb171bfa7b : FAILURE in 4m 39s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7d18402f4c364fd8a74296cab0434772 : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ce808621c220a198e2ba6e82d5d35040fe0cc3d","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-02-11 10:11:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb7fe4af1c5a7232a78e235bd7aa7ede9459d8a0","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-02-12 08:48:57.000000000","message":"Patch Set 2:\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"779fda3d7beca3b8dc072043be1ab402b94ff6a8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-18 11:43:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fdaaf6efdf78c2b5d5b42a6d992be3643dfdab2","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-19 07:30:50.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a4a72d7e1f23010db8d1ceb7331dca5ed715727","revisions":{"8718de38988fd4dfa0e2207a772406d47f88925f":{"kind":"REWORK","_number":1,"created":"2026-02-10 23:39:58.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/976320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/976320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/976320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/976320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/976320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/976320/1"}}},"commit":{"parents":[{"commit":"ed4b05c514124ab897c10c88c5474b152d7908c2","subject":"[doc] Warn admins about share type mutability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ed4b05c514124ab897c10c88c5474b152d7908c2"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-10 23:38:17.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-10 23:39:44.000000000","tz":-480},"subject":"[doc] Document share driver interfaces","message":"[doc] Document share driver interfaces\n\nThis documentation is valuable to call out\nexpectations and behavior, including \"hidden\"\nbehavior, that usually gets called out during code\nreviews of new driver submissions.\n\nAssisted-By: Claude Code\nChange-Id: I0fb5babc8ada9fa14f85115192e250c28dec9f2f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8718de38988fd4dfa0e2207a772406d47f88925f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8718de38988fd4dfa0e2207a772406d47f88925f"}]},"branch":"refs/heads/master"},"2a4a72d7e1f23010db8d1ceb7331dca5ed715727":{"kind":"REWORK","_number":2,"created":"2026-02-10 23:50:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/976320/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/976320/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/976320/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/976320/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/976320/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/976320/2"}}},"commit":{"parents":[{"commit":"ed4b05c514124ab897c10c88c5474b152d7908c2","subject":"[doc] Warn admins about share type mutability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ed4b05c514124ab897c10c88c5474b152d7908c2"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-10 23:38:17.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-02-10 23:50:25.000000000","tz":-480},"subject":"[doc] Document share driver interfaces","message":"[doc] Document share driver interfaces\n\nThis documentation is valuable to call out\nexpectations and behavior, including \"hidden\"\nbehavior, that usually gets called out during code\nreviews of new driver submissions.\n\nAssisted-By: Claude Code\nChange-Id: I0fb5babc8ada9fa14f85115192e250c28dec9f2f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2a4a72d7e1f23010db8d1ceb7331dca5ed715727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2a4a72d7e1f23010db8d1ceb7331dca5ed715727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
