)]}'
{"id":"openstack%2Fmanila~978045","triplet_id":"openstack%2Fmanila~master~I22717b10929b314676b8bc354d3b5db0865c7387","project":"openstack/manila","branch":"master","hashtags":[],"change_id":"I22717b10929b314676b8bc354d3b5db0865c7387","subject":"[WIP] NAE NetApp CI Fixes Depends-On: https://review.opendev.org/c/openstack/manila/+/977290","status":"ABANDONED","created":"2026-02-26 06:54:53.000000000","updated":"2026-03-05 05:30:44.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"b8d2d9671af1968cd02fcad6ec09593f8456de2e","_number":978045,"virtual_id_number":978045,"owner":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-26 08:33:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"reviewer_updates":[{"updated":"2026-02-26 08:22:16.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-26 08:33:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 09:28:58.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"765b261fc7e6cdff8f76b8d0b27f4f7220033d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-02-26 06:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"958fe4a1258907043371771a30bd84d93cf16628","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-02-26 06:55:08.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9d1ec6e8b5f556462d085ff4bc37636fc58762fb","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-02-26 06:55:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55b3185cf890210ea396220869fdb0e469c9ba0f","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-26 08:22:16.000000000","message":"Patch Set 1:\n\nBuild Succeeded: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978045","accounts_in_message":[],"_revision_number":1},{"id":"fd04849ef3239b28a0b3af4132c5c959b8cbedb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 08:33:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a72e42b67a94746abb0fc74ec0aa0bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302aec71a5ad44c3b27e90d91166de34 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b25a82f59ffb4a038b1e43810564ceb5 : SUCCESS in 25m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af5f6c798bb46a18b61388b25ce365c : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2aa70ece5b84e8fb432840b34bb47ec : FAILURE in 11m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc3ac60785fe4df28fe79fafb013baff : FAILURE in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f8403523222444098b3c57c30298c0ec : FAILURE in 11m 16s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2b9c5fdf0680407c83c0630796231848 : SUCCESS in 1h 10m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0c810bdff6fd4644b445d05ab126cc44 : SUCCESS in 49m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/400fe1ca6e574522a47511a77f74b54a : SUCCESS in 1h 27m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/631881fa99e24566be8f876d06d469c1 : FAILURE in 38m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/400e0e5d5e3a46a2af22466d0f72a719 : SUCCESS in 1h 32m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cc1963519b9e4cca88160e4998cd9546 : FAILURE in 1h 18m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3c7db64a02ce4754a581486d93311a1b : SUCCESS in 4m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/362f7126ce6b475891c74688a55d3f2c : SUCCESS in 5m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/df258936caf2481b9b19795fbe66a23c : SUCCESS in 32m 00s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/490608edb35d4534b4aa2d5140977a61 : SUCCESS in 31m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7cf56b4178a44d8e8964bef42a2f79f6 : SUCCESS in 41m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e4782c17b2ff4d289c74051ae0de6cab : FAILURE in 1h 09m 51s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/b0acf9bdd3544cc59aea107d57fb3fa6 : FAILURE in 18m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/dad23eff82de43e9879883d4590a064e : SUCCESS in 43m 42s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd90c7f5941545e7972dad4ee386862a : FAILURE in 4m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d2df263ff99248f48f89d9c53c839aad : FAILURE in 28m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3a0e4e2712fb4d7a8a0d87505f865646 : FAILURE in 12m 34s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0db24795ee0245408f7896a5061e248a : FAILURE in 4m 18s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d30331b0d20041b08ddff1269680b766 : SUCCESS in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cae31f9c76dfa4d82dcc3b28b4c16f13099fd484","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-26 09:28:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/978045/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/3953702/ : SUCCESS in 1h 25m 05s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/978045/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/6cf48c2/ : SUCCESS in 2h 23m 40s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b8d2d9671af1968cd02fcad6ec09593f8456de2e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-03-05 05:30:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a440107170fca5c45aaf9ae9f76bcfb6cc20382","revisions":{"3a440107170fca5c45aaf9ae9f76bcfb6cc20382":{"kind":"REWORK","_number":1,"created":"2026-02-26 06:54:53.000000000","uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/45/978045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/45/978045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/45/978045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/45/978045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/45/978045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/45/978045/1"}}},"commit":{"parents":[{"commit":"4913cac9d133dacb313cabd42b2758543fed6e63","subject":"Merge \"Follow up on API v1 removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4913cac9d133dacb313cabd42b2758543fed6e63"}]}],"author":{"name":"manideepn","email":"manideep.openstack@gmail.com","date":"2026-02-26 06:53:33.000000000","tz":-300},"committer":{"name":"manideepn","email":"manideep.openstack@gmail.com","date":"2026-02-26 06:53:33.000000000","tz":-300},"subject":"[WIP] NAE NetApp CI Fixes Depends-On: https://review.opendev.org/c/openstack/manila/+/977290","message":"[WIP] NAE NetApp CI Fixes\nDepends-On: https://review.opendev.org/c/openstack/manila/+/977290\n\nChange-Id: I22717b10929b314676b8bc354d3b5db0865c7387\nSigned-off-by: manideepn \u003cmanideep.openstack@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3a440107170fca5c45aaf9ae9f76bcfb6cc20382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3a440107170fca5c45aaf9ae9f76bcfb6cc20382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
