)]}'
{"id":"openstack%2Fmanila~978076","triplet_id":"openstack%2Fmanila~master~Idc8a36eded176ad10fbcabd8ba4a6abdb5f49041","project":"openstack/manila","branch":"master","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-26 15:41:27.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Idc8a36eded176ad10fbcabd8ba4a6abdb5f49041","subject":"Add manila review priority and change tracking information","status":"NEW","created":"2026-02-26 14:07:51.000000000","updated":"2026-02-26 16:23:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e293ac31565bbe6f9c6dcd8af024046d23ddc85a","_number":978076,"virtual_id_number":978076,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-26 14:25:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 14:25:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 15:41:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-02-26 16:23:26.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"}],"messages":[{"id":"9c4132885d2aacc27a8d7e8713fe4c870b0f7e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-26 14:07:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c842e410b966e826819cf3842004bd0abcfe8381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 14:25:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c4cc91f8bea4db7935543261a3eb863\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edfec40a53b943ea95ff1b2378c95e55 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f0ed9c442134de3b89cb90ef4d51295 : SUCCESS in 4m 27s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/79d8cc2b3ebc448d80cc189142efaaca : SUCCESS in 4m 58s\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ae7f4506c8734191b391bab2dc6c5038 : SUCCESS in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af25dc6f0c07ac052b23d7f589594c5f10a9a3ab","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-26 15:41:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/978076/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/ed9cb03/ : FAILURE in 1h 13m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e293ac31565bbe6f9c6dcd8af024046d23ddc85a","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-26 16:23:26.000000000","message":"Patch Set 1:\n\nBuild Succeeded: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978076","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d8bf2c134bdc60809b1dd8409477d22cb433da6","revisions":{"8d8bf2c134bdc60809b1dd8409477d22cb433da6":{"kind":"REWORK","_number":1,"created":"2026-02-26 14:07:51.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/76/978076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/76/978076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/978076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/978076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/978076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/76/978076/1"}}},"commit":{"parents":[{"commit":"f6586efd54b17f18fb9ffc56c11bb1fe878947ec","subject":"Merge \"[NetApp] Fix mount format while managing share\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6586efd54b17f18fb9ffc56c11bb1fe878947ec"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-26 14:01:55.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-26 14:07:42.000000000","tz":-180},"subject":"Add manila review priority and change tracking information","message":"Add manila review priority and change tracking information\n\nThis change introduces information regarding where the OpenStack\nManila team keeps track of the ongoing changes on the repository\nand prioritizes the reviews.\n\nChange-Id: Idc8a36eded176ad10fbcabd8ba4a6abdb5f49041\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8d8bf2c134bdc60809b1dd8409477d22cb433da6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8d8bf2c134bdc60809b1dd8409477d22cb433da6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
