)]}'
{"id":"openstack%2Fmanila~978319","triplet_id":"openstack%2Fmanila~master~Iadefd06f68abab4b6996c2b527ed64934e1cb34a","project":"openstack/manila","branch":"master","topic":"bug/2142898","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-03 15:00:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-03-03 18:18:36.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-03 18:10:31.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-03-03 18:18:36.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-03-03 16:24:54.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-03 18:18:36.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-03-03 18:18:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadefd06f68abab4b6996c2b527ed64934e1cb34a","subject":"Adopt to new castellan","status":"MERGED","created":"2026-02-28 13:02:46.000000000","updated":"2026-03-18 17:57:58.000000000","submitted":"2026-03-03 18:18:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"978319-bug/2142898","meta_rev_id":"4b0e65ef33904b24775d7fbdbb039dd5ada93d20","_number":978319,"virtual_id_number":978319,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 18:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-03-03 15:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-03-03 16:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-03-03 16:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-03-03 16:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 14:16:02.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-28 19:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-01 12:05:29.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-02 11:48:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-03-02 11:48:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-03-02 11:48:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2026-03-02 11:48:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-02 15:54:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-03-02 15:54:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-03-02 15:54:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"}],"messages":[{"id":"3810577f948a0f982a1808a2f78d1a20d596029a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 13:02:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cd3568841ac49b759d8f50d791a349a6120fa1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 13:06:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9c39c1d1656f8bcde6fe896f200ff87026be9eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 13:07:47.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"26c7cf3270e7bce28ae5fe4cbc34387c2268c82a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 13:14:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b9766a0dc7ab0c3104c44b462045d8185c44175","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-28 14:16:02.000000000","message":"Patch Set 3:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978319","accounts_in_message":[],"_revision_number":3},{"id":"159993842bacc2572a21504892f196bc6460c9e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 19:05:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd812b12fab24f35b0142fbaae3f9dd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ae02ce93194718ac095c4debade17a : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2539509dca84aaca571568df58259c1 : FAILURE in 34m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414e6505fed74f64a106dcb2bb03f84c : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c2d507197f64ab78f29d79b8bc968a9 : SUCCESS in 9m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c51394d9916473b8a3f9196e513dc91 : SUCCESS in 8m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b85b2bf8aa1744cf83819b28401a8458 : SUCCESS in 12m 50s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bf9bdf57098d42eaa4e70aea1e6cc33c : SUCCESS in 1h 12m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1473a39310e04b4896867d541e27f4de : SUCCESS in 2h 11m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1eeacd23673b490eb15959f0c76990a0 : FAILURE in 55m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a38ef198e7224149bc19a3a2f7ed1d92 : FAILURE in 28m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/603cce54140e464891a939b852447108 : SUCCESS in 54m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a89535fc38a942c69e59c417db13a882 : FAILURE in 1h 18m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a363e12316e047aebbc5b453e27408a6 : SUCCESS in 3m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/448948a0c8a74510b820647d02f61461 : SUCCESS in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/59df2cc472a5403594cc79a2ba23b264 : SUCCESS in 35m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f5eac75b2b0d4208878df02b9c588df4 : SUCCESS in 32m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0ac89692d98e4c2b89b90b4650ca5cc3 : FAILURE in 43m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa3dc7d832b746d283fdcf1f589d9e53 : FAILURE in 57m 58s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/3303020b11144724ac8cc6e126acd795 : FAILURE in 19m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/df38d63d4219415980ebe4916da0aed2 : SUCCESS in 54m 11s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a374b6ca1d5f49b4ab66bcf789cf0f2b : FAILURE in 8m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/fd117390d89f437fb861baabab74213d : FAILURE in 29m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4a0457f08ecf44c3b58762b4039af58c : FAILURE in 27m 14s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/93bf72ae67ac48f2b26e18a6e0c9b55d : FAILURE in 6m 18s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ea001baef1414cbba235b0991ec6a5e4 : SUCCESS in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d3ad9632925c7585e86b9af3cbe52dc81406c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 04:19:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f269d656c7bfe133b3d2da590dc8d8c798b04ff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 04:21:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"49f377778a642c45874c74e9a49ab36831020ffb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 04:21:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b8e8000e1537215cb9dbae56492c13f744e4a5e","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-01 05:15:00.000000000","message":"Patch Set 5:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978319","accounts_in_message":[],"_revision_number":5},{"id":"2fa7772573b2a2b88b072fb0f13cb439af76ebb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:30:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"575d8df1da38137a4ef0b11a78f4f7e1cafb1f74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:35:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ba3075068ecc3aa30c12f31f185d20522077b82b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:46:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"839e8c201490ed71761ce7be3413c5967141ff5b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:47:00.000000000","message":"Topic 978319-patch6 removed","accounts_in_message":[],"_revision_number":8},{"id":"c3b97a8a49f58de0dca6342fa5d7fd4d31e86d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 07:09:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2487ca1f6a8858cbc2339001f956a9d590f39717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 07:23:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8344a163a828ca66bbdb0d14816dfc18b9d435da","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-01 08:11:19.000000000","message":"Patch Set 10:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978319","accounts_in_message":[],"_revision_number":10},{"id":"7164049889db43306df9fbe2281e811ca0681fec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 08:49:54.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"18d4364d685b104d784c280402c60f15307c4c35","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-01 09:39:27.000000000","message":"Patch Set 11:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978319","accounts_in_message":[],"_revision_number":11},{"id":"9700dbc8d54f5a25fd6d7eba367994b32bbcfe13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 10:43:33.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66cbb176b4a4e75b1e1ffce04f373e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f5762c50a640f2b114da702e2b3693 : SUCCESS in 11m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/938db5e8273d4ebb9ea21f9fa4d89634 : FAILURE in 22m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2b2c3d32334cd8a21bf78c0d566d91 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fac947466f64315a5fd26e1259bdf8b : SUCCESS in 11m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1fcda5625d34073a5079ef7410237ee : SUCCESS in 5m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0153f5557fa84dbea830427627bea231 : SUCCESS in 9m 22s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/061e4a1031f343c6afc1b7c09fe2bd4e : SUCCESS in 1h 20m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/efe402ba062b42f58a07b92227ce2fc3 : FAILURE in 1h 50m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/824c9996db914163bbf5078347f7d52b : FAILURE in 1h 02m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/552084b43c3648beb630fcddf686e8e9 : FAILURE in 27m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bc20378fd7664f1bb4eb6c495fd2d1d0 : SUCCESS in 1h 20m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e2c897d1e8a741dcb286cbee09860376 : FAILURE in 1h 11m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/65b83cebda574d6c80945b6fbd8d2140 : SUCCESS in 4m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55f2a008fada4457bebef47e42336bc4 : SUCCESS in 4m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e5d6c4026df4462786ce038bc1800243 : SUCCESS in 36m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8d9802b7d94f404989430aaf8f050c6f : SUCCESS in 20m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/64fd3f8c72e341ab8534f55ae02a30ce : SUCCESS in 30m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ce669ba8846e4f1c8b073cb30d70a321 : FAILURE in 30m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/9da28c8768384072a0c032acf92f6a5b : FAILURE in 18m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c1f4b18a671c4a01a7c40fbc6c176b8d : SUCCESS in 54m 47s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c843ea1e0c6942a7b53d28b3f86ccc25 : FAILURE in 4m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/122fa788e6f043b491745ca75af5ec5a : FAILURE in 30m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a0e38c7dc5694a69ad732f0e7d2d4762 : FAILURE in 20m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0c71d89d50d24d17861a6574b6fbda90 : FAILURE in 4m 16s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0313515d8be0437382061c523faae7c7 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"400123bc1c26669682928993672a4fc347108404","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-01 12:05:29.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/978319/11/upstream-check-manila/manila-tempest-plugin-ontap-dhss/86373e0/ : FAILURE in 1h 20m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"3600f6290ddcc37e9039d15c97eb512378abdeab","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-02 11:48:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"90531e1f1fb049023f913d17f704f6ffc29c9c25","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-02 12:06:16.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f9d9c285b2f1f4dc37bb19be432b072ea14a5cff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-02 15:47:42.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"6a6e9457f7058f830e9da2004ed6de14a940ab66","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-02 15:49:00.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7e0bf8d993c18b5a3d91ba5990ef80e37d5deb1f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-02 16:14:55.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a78d25ce5f67598bdc1e41fc9c6c08823159a72d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-02 16:23:08.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0a59df29f52938a59150760848b7ed5771f139e8","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-02 16:42:05.000000000","message":"Patch Set 12:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978319","accounts_in_message":[],"_revision_number":12},{"id":"9a0aa3551105b68a1b6c55dc1508b22f6a6824f3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-02 18:28:54.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"982b3f3f9ef6df2564c5f3edfea1529303e6122d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-02 18:57:47.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fa728c2c0529fa2ad44fc27509b6c1b18f0b7a69","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-02 18:59:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"23e47825d6f982b66383da8227c67c8eeab7b537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 19:11:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1b5d246564747e3a12a60238212a9f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f66e4c11174655a8b91484cb4bfa3d : SUCCESS in 13m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2171d8eb923433dba3d9b0cc2af1bc6 : FAILURE in 32m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8df91877db4da290b627a71254c5b1 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7002a40835b445269b45e3f0fa423a99 : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51e1811983f9477b8a61a4bb2ff567f5 : SUCCESS in 7m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7ca0070197943ee907fb077729e7117 : SUCCESS in 12m 01s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/667f316adb5b4d91a28106b996a9952e : SUCCESS in 33m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/30453147fc6c44938e89aada10cca519 : SUCCESS in 2h 17m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/53a75c21c2154f08bde393103d783372 : FAILURE in 1h 00m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b6d6efe7d7684402afddbac3e3eab527 : FAILURE in 34m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/75a486be1dca458fac11e87b80dfac3a : SUCCESS in 1h 32m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/768360d77e3a4f9ea74860a4d372467d : FAILURE in 42m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2ecf52d42cf042b5a0890b6f77916cb6 : SUCCESS in 3m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/151bf97b2dc2426b9db3b12f28812a98 : SUCCESS in 4m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a6b4a9ee60f45fab134650240483620 : SUCCESS in 31m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7d4054f533b646c097877116579ad99e : SUCCESS in 30m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eae2c2608cd949889e1024cb1c18fdd4 : FAILURE in 46m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/16c417ee2a354c73b0ff836d617858d7 : FAILURE in 50m 37s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/ead2ed5b24e84b7880b0926098c28044 : FAILURE in 11m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/db5591f73e804db09dd9997dc9b3a55f : SUCCESS in 1h 01m 15s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b358872fe134f6c8337f542ac3f4b97 : FAILURE in 7m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5b04c9c118dd4386bd879ef4ce7659b6 : FAILURE in 13m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8385592463c146309c2230459bb38941 : FAILURE in 30m 30s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/e05a7e2edf274426ac0cb3eb0a600898 : FAILURE in 3m 27s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1d31c2ccab9148718b361da75110bcb5 : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b55cec86c998692de48ea7102f039d7027d548b8","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-03 05:22:32.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8fbbaa1e920536538814a93aef5d67519ed42aed","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-03 06:11:19.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"023be63dcdaf0d1847de53319c41d6d9ea6eee30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 07:40:00.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b231559838dc4f8a8eae8a34d342b3e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f91d45dbe3c74b8b973b536244f2392c : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31b9cb68410f4617b148bdd45bfb7a61 : FAILURE in 20m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df863344edb74ec7bbe5c431fe240a61 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fac8fa6f8664f8eb36a32c49ba87e80 : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98f2f2a7625d4ea2adc0f895a0c6d998 : SUCCESS in 8m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7943a3a27dfa4bc18c141b45efd61048 : SUCCESS in 8m 48s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/98245ce7595449ecbf47dbae8db6dd34 : SUCCESS in 51m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2e12ff53c54742bdb7618e3113af669d : FAILURE in 55m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/113a4154a3154da3bb1c54b9c8d0c7bd : FAILURE in 49m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ecbc9f220695480bb11e25d784b3e37b : FAILURE in 30m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/064f73dac2664a74b1b8feebf3109aa4 : SUCCESS in 1h 02m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8fc14f4bc1d64dae8264c5a84dd06e3e : FAILURE in 1h 23m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/346f2db15307418aa2433ea650c40715 : SUCCESS in 3m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/43313e49813c43b397bc70a9e9272a14 : SUCCESS in 4m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1c3059ebde834536a27289f228559186 : SUCCESS in 42m 37s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3acb76fa98f0423b9bf664b0b6557947 : SUCCESS in 30m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c3f3c9896177427d9b07b24fbf3014a9 : FAILURE in 29m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7ad511bb579d4dc2b4214d692b74345c : FAILURE in 27m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/74fd3fca2c354acbb3ad6e488c529d91 : SUCCESS in 46m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5067d997c5ed4c3eb3c2be0e9c62cd1b : SUCCESS in 55m 27s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9714cd80c5934ce79cc3e006eac92e3c : FAILURE in 4m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e769391d9c2743a5a155991df07022ed : FAILURE in 28m 16s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/15e7b8f59d4f4cd5962f72a5143d533f : FAILURE in 30m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/6712e18eefb343449bc6c6c2a7638363 : FAILURE in 4m 17s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4ec56de847734a768f69f9092333d54b : SUCCESS in 7m 08s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5e6d45278aa8dcdfc34f051d4c06b36fa7e2b71b","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-03 10:23:21.000000000","message":"Patch Set 12: -Workflow Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"43379a9cd3f5d71f7a2698239913177ec2a8b263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-03 10:33:57.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"15fd23006e599012af010cd92668178bdcde789c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-03 10:54:09.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"270ad0bbd12b5d8b5a5a8778bab54150e6d6240e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-03 11:51:51.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"70183c39da5b43e8f0c599caeb0867f1fe4de3b8","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-03 12:09:15.000000000","message":"Patch Set 13:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978319","accounts_in_message":[],"_revision_number":13},{"id":"65741606362879289ffac68a1e58e8eaeaa6f175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 12:13:18.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa7e3eebe0484a19b0113d30215e004e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a6d130c0a04de4ad8d74cf83c780dc : SUCCESS in 11m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b20cfc4d1fb747fb9ef6dd1f6487a40b : FAILURE in 32m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88023e266f344f0bb64c261079f1a78e : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c9036fe231b4e359131fcaa6835a8cf : SUCCESS in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea34ef37489444fc8f6d53e88894279f : SUCCESS in 8m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a3908b7976c4c3fbdbba4ad0f850af8 : SUCCESS in 10m 46s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e5985cd1d0bb452da909a2bf2ced5ee6 : SUCCESS in 36m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3c925c1c39d34a158e83ce678edfe8b3 : SUCCESS in 1h 36m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0736e1ea7cd64c2b95aaaaddf568729c : FAILURE in 1h 35m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8bd8a613c57c4e5f8c08cb1ce34c8c1c : FAILURE in 1h 02m 10s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8fba3c173a894898a1e538ba9800752e : SUCCESS in 48m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4db1adf23bfc437f9f38bbfe0c344aa0 : SUCCESS in 1h 12m 54s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f451eb7c698b474488d56023926d6602 : SUCCESS in 3m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85b395bfab3d423285a935776f77a86a : SUCCESS in 2m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e37f02349b3e494da2874abd7d9d8a6e : SUCCESS in 33m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ca3e5c1464d8467891fcdc79c43e00c9 : SUCCESS in 31m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/548f5c7074c343b786f3913212ecd959 : FAILURE in 38m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/419290bd1630447a9653794f3a027ca0 : FAILURE in 1h 00m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/cdfb5c4f75424a4e82d3d8a291ff5967 : SUCCESS in 1h 07m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4dc5784f1b2d4290a5551ba4b4ddbaae : SUCCESS in 1h 04m 29s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78382e8ae65d41bda212d113e75b6044 : FAILURE in 11m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/da961d716e684903802ccd40343a1dc1 : FAILURE in 17m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6de3d68d4a4546918a8543f77dfa2992 : FAILURE in 27m 36s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/6625a24c1c5c4c2ea46af883db396521 : FAILURE in 4m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/cc76b22dd0b84aaa83c6176783efa467 : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8d0af99fb3f19d498e3ad4622be51f1a26c13bdc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-03 12:15:48.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"bc9aee7a7357ac7279d8bb3917b46a5d42f5ee00","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-03 13:00:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1e4d1e87a0a7c3c6ad9e66b8a996dca35cf8d427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 14:45:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca8fe2fde114dbaa59d7c19cac99b18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e572d045717d42b9ad8c0fcbfdab4873 : SUCCESS in 11m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e76dc290eec24d348ee4b9606fcd65f1 : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b13e5deb62e46debe7058cd3e760be7 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d923c48a00a44c0581557f813b4d55b9 : SUCCESS in 6m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d0567be863142b8890235f32c1e8ff8 : SUCCESS in 7m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a4c3ae5e3a534d51a32dbcb53c75255f : SUCCESS in 9m 59s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e8dccae87ec54202b26bf3030d43c992 : SUCCESS in 1h 18m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c05eafffe5634a46a36a18284e6ada62 : FAILURE in 2h 22m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1e1a24326ba94f73bb980fc650f03338 : FAILURE in 1h 47m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ae1a9585987a4e4198bfa68280f9d0b1 : FAILURE in 44m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d2535d29e015453cb8cab5856fa9c879 : SUCCESS in 45m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c8d8ac4493b84d7bad365f8a3b2d2d20 : FAILURE in 1h 15m 21s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/14517f1c2936461bbd6e4405a2efa3f1 : SUCCESS in 5m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98cc70b65fc040a9ba83bcfaa951bcf9 : SUCCESS in 4m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fa6c96e9a7da47c1972b13a59b5e7003 : SUCCESS in 21m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9cb3f48f8d754636a714d9ccfac3bb2a : SUCCESS in 31m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/36d4616d968248e68312e8438d97ffbd : SUCCESS in 41m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0e54692bb02f4342862db16de500ed05 : FAILURE in 1h 04m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/3175a390010d44a1b50563f4b181e10f : FAILURE in 39m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fac05eba19894e8cbbd3b068f3204073 : SUCCESS in 1h 07m 45s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/612bacca5d2541b69ae78b72ec84a9c0 : FAILURE in 7m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/07c1cf29e6d143d581ae67a493190f3c : FAILURE in 28m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b2cdd0bb1a814f0391c43d02aa6ec02e : FAILURE in 29m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b6f07883a37a43e5b77a82b3a3bc96cf : FAILURE in 2m 38s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3d82116cdc3549a09269bb578ed905db : RETRY_LIMIT in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b56b1034399ef2c2b97634deac5934bf960ff081","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-03 15:00:15.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"55dde84104ffa01d4cc11d6c688c98d59e82f131","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-03 15:00:50.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ac749b856ea00c014902f520fda7ec31cc51a933","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-03 15:30:12.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/978319/13/upstream-check-manila/manila-tempest-plugin-ontap-dhss/e32ca4b/ : FAILURE in 42m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"0d2aaf98e15bd214787ac0850e0248a2ca510e34","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-03 16:24:54.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b7c7ada8b052f080b63ea0dd6ea8ac61fb4cc05b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-03 16:56:24.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"13d475ffa54918c5b82881dcb3a9c4f5083d8a1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 16:56:46.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"f93977ce1180eaf24862a145716b20b3e6d71481","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-03 16:57:14.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"f59b78b8cf22fa5dde0320aa5616985c947d0de7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-03 16:58:39.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c2bcd1f4d3be82f1ca4c7a3ba8ab4f9712d0f42c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-03 18:10:31.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"376a3d24fc75e82d9f07dba7c6eb1429914a476f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:18:36.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ce0633590244f5ade975fd6fd391f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87646cd0811a42f39d544d4234b3f24f : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d3a5298f384ab9a631ebb586df768e : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/511a4d271eba4fa6a10e63df592515b8 : SUCCESS in 10m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d7d97db413d4df186a1d8b79d18c43c : SUCCESS in 11m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/113a5e3fbe6d4764bd46158d1c421ccc : SUCCESS in 1h 09m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/94ff2c278b43409fbade11e877e0ec26 : SUCCESS in 36m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2e696e66cebf4c47bf74cfc262f3d09e : SUCCESS in 20m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9706a16628324ea880718d7709a9ec80 : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":13},{"id":"6bf35d41f7edccf56c466505f00210e4f9fad654","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:18:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"38f2c38f44bd87a00ab2d248506ef422f6bc631a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:20:05.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a240b4fd687a40148ba786b57e851b29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa768ce86c5e4c799b10f72003716896 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":13},{"id":"7f7d9a0389a59fcef42e8b1979b7648a5852b42c","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-03-17 05:38:22.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6f6085a6a8572009c10895ed9ed72f7d57ad4650","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-18 17:46:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4b0e65ef33904b24775d7fbdbb039dd5ada93d20","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-18 17:57:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"c10154bc7997dbdef7b50ebb3f5a1b13b28118c7","revisions":{"cce2965dbd1c73820c2bbd85e8de71e388814791":{"kind":"REWORK","_number":1,"created":"2026-02-28 13:02:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/1"}}},"commit":{"parents":[{"commit":"f6586efd54b17f18fb9ffc56c11bb1fe878947ec","subject":"Merge \"[NetApp] Fix mount format while managing share\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6586efd54b17f18fb9ffc56c11bb1fe878947ec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:33.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint url\nor secret url by its own logic, because underlying barbicanclient just\nignores the base url part but uses only the last uuid part[1]\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new API access to secret url to get\nsecret href. This is considered to be robust and should be efficient\nonce manila completely migrates to castellan \u003e\u003d 5.6.0\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cce2965dbd1c73820c2bbd85e8de71e388814791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cce2965dbd1c73820c2bbd85e8de71e388814791"}]},"branch":"refs/heads/master"},"5e53a888a62afa62737dfd2585a49e6afbdb399f":{"kind":"REWORK","_number":2,"created":"2026-02-28 13:06:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/2"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:05:59.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint url\nor secret url by its own logic, because underlying barbicanclient just\nignores the base url part but uses only the last uuid part[1]\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new API access to secret url to get\nsecret href. This is considered to be robust and should be efficient\nonce manila completely migrates to castellan \u003e\u003d 5.6.0\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5e53a888a62afa62737dfd2585a49e6afbdb399f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5e53a888a62afa62737dfd2585a49e6afbdb399f"}]},"branch":"refs/heads/master"},"a3b85cb17b03550a3d240d6ebc8775e0d240d078":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-28 13:07:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/3"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:06:34.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is considered to be robust and should be efficient once\nmanila completely migrates to castellan \u003e\u003d 5.6.0\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3b85cb17b03550a3d240d6ebc8775e0d240d078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3b85cb17b03550a3d240d6ebc8775e0d240d078"}]},"branch":"refs/heads/master"},"83fe4ba516532e8ec1c08e136e588cb0badbcd7c":{"kind":"REWORK","_number":4,"created":"2026-03-01 04:19:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/4"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 04:19:40.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is considered to be robust and should be efficient once\nmanila completely migrates to castellan \u003e\u003d 5.6.0\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/83fe4ba516532e8ec1c08e136e588cb0badbcd7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/83fe4ba516532e8ec1c08e136e588cb0badbcd7c"}]},"branch":"refs/heads/master"},"75919634157cbb14fa9020e9e6bfe3f1b7b96813":{"kind":"REWORK","_number":5,"created":"2026-03-01 04:21:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/5"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 04:21:13.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is considered to be robust and should be efficient once\nmanila completely migrates to castellan \u003e\u003d 5.6.0\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/75919634157cbb14fa9020e9e6bfe3f1b7b96813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/75919634157cbb14fa9020e9e6bfe3f1b7b96813"}]},"branch":"refs/heads/master"},"cecc06bd34db82af31cef05ec4ff313c1189d7f3":{"kind":"REWORK","_number":6,"created":"2026-03-01 06:30:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/6"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 06:29:58.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is considered to be robust and should be efficient once\nmanila completely migrates to castellan \u003e\u003d 5.6.0\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cecc06bd34db82af31cef05ec4ff313c1189d7f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cecc06bd34db82af31cef05ec4ff313c1189d7f3"}]},"branch":"refs/heads/master"},"c67102ba5382b62333c7f93447d425acfe2edc90":{"kind":"REWORK","_number":7,"created":"2026-03-01 06:35:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/7"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 06:34:05.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1]. This follows the same approach and\nremoves the redundant logic to create full url.\n\nNote that encryption_key_ref in share metadata is now set to UUID, not\nto full URL. The existing share with URL encryption_key_ref is still\nfunctional because barbicanclient can handle both.\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c67102ba5382b62333c7f93447d425acfe2edc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c67102ba5382b62333c7f93447d425acfe2edc90"}]},"branch":"refs/heads/master"},"463b5766fdc65d175b5c86aa7865d3aadd11f00c":{"kind":"REWORK","_number":8,"created":"2026-03-01 06:46:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/8"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 06:46:27.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is considered to be robust and should be efficient once\nmanila completely migrates to castellan \u003e\u003d 5.6.0\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/463b5766fdc65d175b5c86aa7865d3aadd11f00c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/463b5766fdc65d175b5c86aa7865d3aadd11f00c"}]},"branch":"refs/heads/master"},"2bb1e44894d0108cfe539fa352e504f6f64015a2":{"kind":"REWORK","_number":9,"created":"2026-03-01 07:09:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/9"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 07:08:44.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is more robust than implementing endpoint detection in\nmultiple layers, and allows us to ensure that the requested secret can\nbe accessed early.\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2bb1e44894d0108cfe539fa352e504f6f64015a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2bb1e44894d0108cfe539fa352e504f6f64015a2"}]},"branch":"refs/heads/master"},"893865893fb25399c0138293ff706a72b80557c1":{"kind":"REWORK","_number":10,"created":"2026-03-01 07:23:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/10"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 07:22:51.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is more robust than implementing endpoint detection in\nmultiple layers, and allows us to ensure that the requested secret can\nbe accessed early.\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/893865893fb25399c0138293ff706a72b80557c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/893865893fb25399c0138293ff706a72b80557c1"}]},"branch":"refs/heads/master"},"e9f88aeec004c0fdb3ff5db8642a35df171f92d9":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-03-01 08:49:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/11"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 08:49:09.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is more robust than implementing endpoint detection in\nmultiple layers, and allows us to ensure that the requested secret can\nbe accessed early.\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nCloses-Bug: #2142898\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e9f88aeec004c0fdb3ff5db8642a35df171f92d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e9f88aeec004c0fdb3ff5db8642a35df171f92d9"}]},"branch":"refs/heads/master"},"76513ff9211375b77683f405da2d9ddb52a8b40e":{"kind":"REWORK","_number":12,"created":"2026-03-02 15:47:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/12"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-02 15:47:27.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is more robust than implementing endpoint detection in\nmultiple layers, and allows us to ensure that the requested secret can\nbe accessed early.\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nCloses-Bug: #2142898\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76513ff9211375b77683f405da2d9ddb52a8b40e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76513ff9211375b77683f405da2d9ddb52a8b40e"}]},"branch":"refs/heads/master"},"c10154bc7997dbdef7b50ebb3f5a1b13b28118c7":{"kind":"REWORK","_number":13,"created":"2026-03-03 10:33:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/978319/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/978319/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/978319/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/978319/13"}}},"commit":{"parents":[{"commit":"9f499ff74881daef98502ced89377791eb990a6e","subject":"Merge \"Netapp: Add support for QoS type specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f499ff74881daef98502ced89377791eb990a6e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:02:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-03 10:33:47.000000000","tz":540},"subject":"Adopt to new castellan","message":"Adopt to new castellan\n\ncastellan 5.6.0 removed its internal logic to create full endpoint base\nurl or  secret url by its own logic, because underlying barbicanclient\nuses only the last uuid part[1].\n\nThis follows the same approach and removes the redundant logic to\ncreate full url.\n\nNote that this introduces the new access to secret API to obtain secret\nhref. This is more robust than implementing endpoint detection in\nmultiple layers, and allows us to ensure that the requested secret can\nbe accessed early.\n\n[1] a227bd9c4c29707239a46211af1af276df2a1464\n\nCloses-Bug: #2142898\nChange-Id: Iadefd06f68abab4b6996c2b527ed64934e1cb34a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c10154bc7997dbdef7b50ebb3f5a1b13b28118c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c10154bc7997dbdef7b50ebb3f5a1b13b28118c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
