)]}'
{"id":"openstack%2Fmanila~978321","triplet_id":"openstack%2Fmanila~master~I5b3bd29f8a35e980a767b36caadf1e4461f58e2a","project":"openstack/manila","branch":"master","topic":"bug/2142898","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-02 18:19:49.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5b3bd29f8a35e980a767b36caadf1e4461f58e2a","subject":"Enforce castellan \u003e\u003d 5.6.0","status":"NEW","created":"2026-02-28 13:06:26.000000000","updated":"2026-03-03 12:58:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aadc35e85b613134e78c71fc3e33abf32dc670ff","_number":978321,"virtual_id_number":978321,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-03 12:36:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 13:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-28 15:06:39.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-03-01 12:29:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"e9426b7ab1d3b33e1559908034e5a7d898989c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 13:06:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71eafd4cc20d894d452dae5953b7d39009fa8f5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 13:07:47.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"92058caeb7e030e19e3c4f52a79ea066500e74b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 13:11:40.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/requirements/+/977783 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"5d3c4053977be6ff5b61f4b56f4765d6fd806e37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 13:14:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ba04f600c37e178ef5e1a5005b5629e62f6b2e13","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-28 15:06:39.000000000","message":"Patch Set 3:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978321","accounts_in_message":[],"_revision_number":3},{"id":"f771acb0fae03124671c95155148d3da3cf1e9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 18:48:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67ccbbf6c6544b9b8f732baa50e3ef2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4709b7d02a4bb58574b2030ef68acd : SUCCESS in 12m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9476f880f91441978345914a1da28da3 : SUCCESS in 30m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013114a4b287471fb3acd8496e4e2dad : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3aad7a8d4b22407aa45568fffcc8321c : SUCCESS in 8m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9282c2a49a184e39bcedcb1594e6cb4e : SUCCESS in 9m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/248261f05c60474e9bec3cf4164357af : SUCCESS in 8m 05s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b0dbd0be4a7436e9895828555b3c87f : SUCCESS in 3m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0fcc6aa6457441fda3d47f88c94f9ab0 : SUCCESS in 1h 06m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/db3b81c3fa094055bc278c01b9352fbf : SUCCESS in 1h 28m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9225d2bf297d4b5da1d6cde89bb9152e : FAILURE in 36m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d06e68108a2844f696d8e0ae59930e83 : FAILURE in 1h 01m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8d566d55ec174defa0de573ce79ebb82 : SUCCESS in 1h 24m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/086c3e69bf9f49b89d92822f84f4c350 : FAILURE in 58m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/461fc87c878b410fa8d9f65f192f332c : SUCCESS in 4m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4ef15a76bcc0493eaa113ae9a594f1c4 : SUCCESS in 7m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/24dea28f728c4cb0bbf551aec7143da2 : SUCCESS in 33m 23s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1624ac4cbbd748e091843d828670648d : SUCCESS in 19m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/92531b4ad5344942bc268bd86e029fb0 : FAILURE in 29m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1e2687c6295c41fabd2685a84653bcbe : FAILURE in 30m 45s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/32da9703a6554a419ec8d991e1ddfa01 : FAILURE in 11m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b800a2a7de314997b3cc343c8386fcc3 : SUCCESS in 56m 07s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a5774d8f61a437fb7ebdeeeb82ee5f2 : FAILURE in 14m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5a1491621ad34c05b621fde9da99bac3 : FAILURE in 28m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6032d3a2a3a34566bde8c6c326b60f3d : FAILURE in 27m 22s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/2c43c1cc245e47b9b0c0fa73a6fdf180 : FAILURE in 6m 41s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/01661bac07c44df9844a2b664f7fcae2 : RETRY_LIMIT in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc1cd6e722320f7e74d95c33fc857e9019697f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 05:12:42.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cb7714e12ee6db5671d258e1fd30295dde52ff05","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-01 06:02:58.000000000","message":"Patch Set 4:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978321","accounts_in_message":[],"_revision_number":4},{"id":"2e1ae54c192829302bd5022b4c938ddc37f9e059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:30:12.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c49f054d04e3e662177dcc97c6a5f98d34f6beb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:35:42.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"17e6394938c5f4c28322fa7f10dd39f335493080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 06:46:46.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/requirements/+/977783 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"4d70cee79b07ef3b7f097b30aa9c559db06c8813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:47:05.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6f1f4d6133329ce62a792f5027bf1c92062886e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 07:09:46.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"9b82aadab83d96007efbf2f99c3ce2a14e8e72da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 07:23:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"2a270dfe25b941118aff82d7bad88af70e97dab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 08:49:54.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"1c9eebdbf1880ab13610506d968f23318b1c3b15","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-01 10:27:36.000000000","message":"Patch Set 10:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978321","accounts_in_message":[],"_revision_number":10},{"id":"6197faa43fdf30f4d456b5813c16dbfd039457e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 11:07:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa81bb30328d49f1a146eae919db89b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1080081ecbc945809a80b5cf92a5d89b : SUCCESS in 11m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc90b8e2526242c5824d3a23ed9fc38e : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb491825e868498aa67462607873770b : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25d0a6047cae440a8fe1341db6a1fa51 : SUCCESS in 9m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd2b8aed713f4ca38eca69ea0bb22158 : SUCCESS in 11m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36a3d3be45664259ad0797e59eb01a6f : SUCCESS in 13m 12s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/d020bcfbda8749e6af619d2a425b0ab0 : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5394c84307ae4f39aa312928ebd3c69e : SUCCESS in 1h 21m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/50717317d32b43ac94315aad1ee8b9df : FAILURE in 2h 11m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b04d5d19a83548cd9d5617b62f7cbd19 : FAILURE in 53m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/201b93d6e2f24945ada9f873332d53e1 : FAILURE in 54m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5b4ce7f2d53e440eaa1bc0904ccce1ac : SUCCESS in 1h 23m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/365dfb04d4a74abbba881852c666ef30 : FAILURE in 1h 27m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/36b0e2db0b2144a1a3c977e44e534b53 : SUCCESS in 4m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fca2c71ccb9a4cf296430cc69f6c86e7 : SUCCESS in 6m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a14c6ef7b14f40a2a61537285a57e145 : SUCCESS in 32m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/83d958bff9df495c8c6008bd52b39b3c : SUCCESS in 30m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9d17983224ce418cb98dbb8f898f98aa : SUCCESS in 39m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/57baecbe0aa8406cb4fb8a9a567905ed : FAILURE in 39m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/13983be907394db7bbf41c0a5caa10e7 : FAILURE in 9m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/def1e76822504ebd870ec2d6df41a3bd : SUCCESS in 1h 00m 10s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/024260a9a05c4c2aabbe5218878dfc1b : FAILURE in 4m 40s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2c78b21570a94588bb6ce6825878a86b : FAILURE in 33m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2d6c426ac55d428bb37d444f595bd7e6 : FAILURE in 18m 40s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/8207103dfa5546ae9038d49188cfec0c : FAILURE in 4m 30s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/755cf0cfd71b439c9473c97e94d685d2 : RETRY_LIMIT in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dbdc341c56e5b6c226f5bb73497ac39ebd63c934","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-01 12:29:13.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/21/978321/10/upstream-check-manila/manila-tempest-plugin-ontap-dhss/a0f574a/ : SUCCESS in 1h 13m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"923db3ba94dde98f5b2e8e77bf85f7d49d92172a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-02 15:49:08.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"3430764190007e97a7ab24109212a5e9bccec9d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:14:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50ea24524cc14865b0b13bd7f96d247d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c660b2cc5f5043d692e1ce5c5f48cd63 : SUCCESS in 12m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fac4a537da14ad185b4d64cfa56c637 : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eef154333414e3f8f2041800930c843 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14a68d5c11254222920225bdb504c120 : SUCCESS in 10m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f806f019934647ac833192fed33f184e : SUCCESS in 9m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/971b5757c3104f94942f79b1f9a20805 : SUCCESS in 12m 16s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/ddf4b2bf30a24b0b8b7e133a5ca30ec7 : SUCCESS in 3m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/808b467c85954600ad32ed1d27267c62 : SUCCESS in 1h 06m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6d52dc2b2de444eea618b1864fabd3bf : SUCCESS in 52m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c9939f866c8a4fcaba4f08245437d23f : FAILURE in 1h 07m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5536cc8c7fd74820af7d9ce367f8a75d : FAILURE in 27m 52s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c18da6374fc24194aade9aa1be366409 : SUCCESS in 1h 19m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b847e0c709aa49269cb134837ea5deb5 : FAILURE in 1h 24m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/85fc9c65f8044b99ac44d24cc536ea7c : SUCCESS in 3m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4bb4c4fc4e8d492eb92e670f7dcf6647 : SUCCESS in 6m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/45ab2ca306ce4a40af2511dc41992055 : SUCCESS in 34m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8c841205cb16489fad48a3f3c04324c7 : SUCCESS in 25m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d9e2c6c1e6e1402099253ba2204e2ab2 : FAILURE in 44m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/017d657de7644b0fa519f4e413ec578a : FAILURE in 1h 03m 46s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/d568f56ce1564fac87d6b027dcf3b567 : FAILURE in 9m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bbe424937b9e4c62856997139f80de54 : SUCCESS in 50m 47s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f121158eb05a4c22b261eb7aabf7585b : FAILURE in 12m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/400d2b8040ee4223813f5162c6cc7a28 : FAILURE in 17m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e7b4cfe025994958ab60e6d0b19a29c6 : FAILURE in 23m 31s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/6c08eb8710b6461681c0ce4062796665 : FAILURE in 5m 34s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0f1608d7421e4634a462efc708441c13 : RETRY_LIMIT in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bce7e60032a8eef9173094cf006828f5e195f726","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-02 18:19:49.000000000","message":"Patch Set 11:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978321","accounts_in_message":[],"_revision_number":11},{"id":"6d42cf1278cbb7c56b551852a5fafa4620c6d8b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-03 10:34:24.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"f5f6d99594712412d3df76e0fb7779685a878949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 12:36:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f375c6974b284c0b8d572b197cd04206\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2436c5d24966473194c301665730b666 : SUCCESS in 11m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cf1dab1458c4180bcc0ac3d041618ee : SUCCESS in 34m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e9bf42ab7e48a8beb031230bb200ab : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a925c76efbe946c197d26222e3a6da64 : SUCCESS in 9m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91802f7771904f639f465f6dd5b29b26 : SUCCESS in 9m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d42c078ee0c48d1946be7058f9a52f6 : SUCCESS in 10m 28s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c7d259462c942ef82e79071feab61de : SUCCESS in 2m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/efd8b63c38bf4a3598c383629fdabe9a : SUCCESS in 1h 18m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3d292ef088cc4a5cb5c5b15e2a959324 : SUCCESS in 1h 58m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1db78de83e734f298ba17667ef2329c9 : FAILURE in 40m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f7f3bad79cd949a793828a787273ce76 : FAILURE in 53m 38s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/151d7615bae04e72b8b3e20dfb217cd3 : SUCCESS in 1h 16m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/16beedd00e9c454cad07e69c26480fa2 : FAILURE in 1h 17m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4010ed66a1254eb2b36d942e7560c424 : SUCCESS in 3m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a43744ca8afd453b8c906bf9caa9c18c : SUCCESS in 7m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4baaeb2f2889478d9170b805ad77f497 : SUCCESS in 33m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/aee1bffceb744dd28a4a84b495be02ae : SUCCESS in 33m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e8e0c1a5054d40cd8ff5da52348b7eaa : SUCCESS in 36m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d50cfbe4552b48afad1b21f384f7a80e : FAILURE in 34m 24s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/780eb09293d24029b9b486a021431820 : SUCCESS in 1h 17m 28s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/96597acffac7468f8f3872646acafc85 : SUCCESS in 44m 18s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a179651b4784dea9d69bbbee883a3af : FAILURE in 12m 11s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c5234bb12b964ab4b51814a3ed17a979 : FAILURE in 30m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d0b3d34ce42e4a95acf28b8256029f74 : FAILURE in 26m 18s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/07da52f07626407381aed170b5d8ace2 : FAILURE in 5m 01s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/6de1f40a70e74c5f8914794ab6674d06 : RETRY_LIMIT in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"11859bc046905ed0f654f870eb92ab1a1bd13efa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-03 12:50:29.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/21/978321/12/upstream-check-manila/manila-tempest-plugin-ontap-dhss/d8f01b3/ : RETRY_LIMIT in 7m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"aadc35e85b613134e78c71fc3e33abf32dc670ff","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-03 12:58:22.000000000","message":"Patch Set 12:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978321","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"fc8914e74f1aee9fde1846ed0f458358fb6ab95d","revisions":{"34bda3086afd2206a166a3ab58de59d8d7387179":{"kind":"REWORK","_number":1,"created":"2026-02-28 13:06:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/1"}}},"commit":{"parents":[{"commit":"5e53a888a62afa62737dfd2585a49e6afbdb399f","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5e53a888a62afa62737dfd2585a49e6afbdb399f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:06:17.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/34bda3086afd2206a166a3ab58de59d8d7387179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/34bda3086afd2206a166a3ab58de59d8d7387179"}]},"branch":"refs/heads/master"},"36c6a3b8cd5f653d4c951b1ae7ae96bb2ab5cfb6":{"kind":"NO_CHANGE","_number":2,"created":"2026-02-28 13:07:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/2"}}},"commit":{"parents":[{"commit":"a3b85cb17b03550a3d240d6ebc8775e0d240d078","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3b85cb17b03550a3d240d6ebc8775e0d240d078"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:07:25.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/36c6a3b8cd5f653d4c951b1ae7ae96bb2ab5cfb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/36c6a3b8cd5f653d4c951b1ae7ae96bb2ab5cfb6"}]},"branch":"refs/heads/master"},"b43b65af569641ce9c9e8504ece11cc3f921258f":{"kind":"REWORK","_number":3,"created":"2026-02-28 13:14:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/3"}}},"commit":{"parents":[{"commit":"a3b85cb17b03550a3d240d6ebc8775e0d240d078","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3b85cb17b03550a3d240d6ebc8775e0d240d078"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:13:54.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b43b65af569641ce9c9e8504ece11cc3f921258f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b43b65af569641ce9c9e8504ece11cc3f921258f"}]},"branch":"refs/heads/master"},"1f616120bf423764df7eeeeac2c1dd92d96243a0":{"kind":"REWORK","_number":4,"created":"2026-03-01 05:12:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/4"}}},"commit":{"parents":[{"commit":"75919634157cbb14fa9020e9e6bfe3f1b7b96813","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/75919634157cbb14fa9020e9e6bfe3f1b7b96813"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 05:12:32.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f616120bf423764df7eeeeac2c1dd92d96243a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f616120bf423764df7eeeeac2c1dd92d96243a0"}]},"branch":"refs/heads/master"},"9e50749f6b2013ef4e40f7bfb262aed2f26b3adb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-01 06:30:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/5"}}},"commit":{"parents":[{"commit":"cecc06bd34db82af31cef05ec4ff313c1189d7f3","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cecc06bd34db82af31cef05ec4ff313c1189d7f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 06:30:01.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9e50749f6b2013ef4e40f7bfb262aed2f26b3adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9e50749f6b2013ef4e40f7bfb262aed2f26b3adb"}]},"branch":"refs/heads/master"},"5ffb53f0c38546abd5d079f0f60a5b0efda6fff9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-03-01 06:35:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/6"}}},"commit":{"parents":[{"commit":"c67102ba5382b62333c7f93447d425acfe2edc90","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c67102ba5382b62333c7f93447d425acfe2edc90"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 06:35:33.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5ffb53f0c38546abd5d079f0f60a5b0efda6fff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5ffb53f0c38546abd5d079f0f60a5b0efda6fff9"}]},"branch":"refs/heads/master"},"60eb8b0da7b8ab8e2a49b8140dc6a2979bea68db":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-03-01 06:47:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/7"}}},"commit":{"parents":[{"commit":"463b5766fdc65d175b5c86aa7865d3aadd11f00c","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/463b5766fdc65d175b5c86aa7865d3aadd11f00c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 06:47:05.000000000","tz":0},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/60eb8b0da7b8ab8e2a49b8140dc6a2979bea68db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/60eb8b0da7b8ab8e2a49b8140dc6a2979bea68db"}]},"branch":"refs/heads/master","description":"Rebase"},"7ef3e4fd7b7a1b1677b8422426755af7d9906e56":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-03-01 07:09:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/8"}}},"commit":{"parents":[{"commit":"2bb1e44894d0108cfe539fa352e504f6f64015a2","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2bb1e44894d0108cfe539fa352e504f6f64015a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 07:09:34.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7ef3e4fd7b7a1b1677b8422426755af7d9906e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7ef3e4fd7b7a1b1677b8422426755af7d9906e56"}]},"branch":"refs/heads/master"},"a7443fb4a2f6830cc01e4d548275903d0b169bbc":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-03-01 07:23:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/9"}}},"commit":{"parents":[{"commit":"893865893fb25399c0138293ff706a72b80557c1","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/893865893fb25399c0138293ff706a72b80557c1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 07:22:56.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7443fb4a2f6830cc01e4d548275903d0b169bbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7443fb4a2f6830cc01e4d548275903d0b169bbc"}]},"branch":"refs/heads/master"},"5eb614e0f5cc9018fb3ab84361fcb357b1797fda":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-03-01 08:49:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/10"}}},"commit":{"parents":[{"commit":"e9f88aeec004c0fdb3ff5db8642a35df171f92d9","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e9f88aeec004c0fdb3ff5db8642a35df171f92d9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 08:49:22.000000000","tz":540},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nRelated-Bug: #2142898\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5eb614e0f5cc9018fb3ab84361fcb357b1797fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5eb614e0f5cc9018fb3ab84361fcb357b1797fda"}]},"branch":"refs/heads/master"},"eaca68635756bbe0483703c4cfb80236f0e152fa":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-03-02 15:49:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/11"}}},"commit":{"parents":[{"commit":"76513ff9211375b77683f405da2d9ddb52a8b40e","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76513ff9211375b77683f405da2d9ddb52a8b40e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-02 15:49:08.000000000","tz":0},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nRelated-Bug: #2142898\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/eaca68635756bbe0483703c4cfb80236f0e152fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/eaca68635756bbe0483703c4cfb80236f0e152fa"}]},"branch":"refs/heads/master","description":"Rebase"},"fc8914e74f1aee9fde1846ed0f458358fb6ab95d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-03-03 10:34:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/978321/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/21/978321/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/21/978321/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/21/978321/12"}}},"commit":{"parents":[{"commit":"c10154bc7997dbdef7b50ebb3f5a1b13b28118c7","subject":"Adopt to new castellan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c10154bc7997dbdef7b50ebb3f5a1b13b28118c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 13:04:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-03 10:34:24.000000000","tz":0},"subject":"Enforce castellan \u003e\u003d 5.6.0","message":"Enforce castellan \u003e\u003d 5.6.0\n\nTo avoid duplicate logics about href composition.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/977783\nRelated-Bug: #2142898\nChange-Id: I5b3bd29f8a35e980a767b36caadf1e4461f58e2a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fc8914e74f1aee9fde1846ed0f458358fb6ab95d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fc8914e74f1aee9fde1846ed0f458358fb6ab95d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
