)]}'
{"id":"openstack%2Fmanila~978331","triplet_id":"openstack%2Fmanila~master~I218eaa3795d45fd8c67fdd3919e74ed03dc25136","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-01 08:34:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I218eaa3795d45fd8c67fdd3919e74ed03dc25136","subject":"Use bare uuid for secret_ref","status":"ABANDONED","created":"2026-02-28 16:21:12.000000000","updated":"2026-03-01 08:46:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"068503129ccbe00b5b3f2061cb126d9c3553e31f","_number":978331,"virtual_id_number":978331,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-01 07:09:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-03-01 08:34:21.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 17:16:15.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-02-28 18:47:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-01 06:51:41.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"}],"messages":[{"id":"ad9accc7fc2cb7d9570086244202894449eb3f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 16:21:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20d1c65dea4e5f81241c37665337491fe1e431bf","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-02-28 17:16:15.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978331","accounts_in_message":[],"_revision_number":1},{"id":"fd97d1666e2386d112d91d5be59dce6a7d514e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 18:47:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d97190cdfea4fba81e0b02f65ec3acd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3dbf2ec8d3048419e91a213f1d0a6ca : SUCCESS in 14m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3df72d37d8fe434ab80854263a693bc7 : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f799d40f0754c1b90ca6402236e23d2 : SUCCESS in 7m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e7bc1b842c4446fbced62e827b85646 : SUCCESS in 6m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f943fe412ab1487ba2d65f8fc718a45c : SUCCESS in 10m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b45d2bcd5bd0499d88b4c23d2fc1f535 : SUCCESS in 14m 09s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b532bc898c314cfabab703056124442b : FAILURE in 31m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1be173258b4c47d294302d7a81c538b7 : SUCCESS in 1h 00m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a60410b536b649929f8f7cf3452c14d8 : FAILURE in 1h 14m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/26892b11c31c4ce49eabb83d2a8b8fc2 : FAILURE in 58m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4edceec53848468f8ccb8c11bccc4bb2 : SUCCESS in 1h 22m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0239800f45c34b86a6e92f1283c5f558 : FAILURE in 1h 08m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e7019b5eeeea428bbb5cbcc9ad03e9b0 : SUCCESS in 4m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48020205a0294fd3b7fb5df48583e4cf : SUCCESS in 4m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/85b9dde4f647455cbf8ece674f0af7c4 : SUCCESS in 23m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8527a52039534f0281cb7e53852146ec : SUCCESS in 26m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71e77098e4044d60b57a39265cd0e536 : FAILURE in 44m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f74f0c25a09d4b69b1b71060a1c0efaf : FAILURE in 1h 02m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/8721f218ea164f40b6023b729e797056 : FAILURE in 19m 49s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ceb9830f50fd4d51a1788aa3e890c14c : SUCCESS in 59m 25s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff9a4de5e52441978d8730db67f62579 : FAILURE in 12m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/04ef5523636b4f66a49a5e579b73dec5 : FAILURE in 18m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e9b809d182464a0fb758bd544a032180 : FAILURE in 20m 57s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a5d3dee046c648b6adc633d2b1a74ff3 : FAILURE in 4m 24s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ee4e02843cd14da694db8b5f989fd0f0 : RETRY_LIMIT in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f4c598cdee0ec5753154d3835c5e8b11bbb21b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 05:12:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f30a1d726c76907cd20beca2696f8a286f3958f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 06:30:19.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/978321 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"30148ad5fa710694df416b79c43191b15f219b66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:35:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"6b33b3706709a2a6e4bc469f6bedd594386cddfc","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 06:47:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"1f62a6d75c6a1d618bece9b59de93f04db1a92cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 06:47:28.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila for 978319,8","accounts_in_message":[],"_revision_number":2},{"id":"d22931e7d342f0d31bc1edb2022e3087bca27b14","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-01 06:51:41.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update openstack.gerrit/openstack/requirements\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"ac5483dc3e7d8efe768184a86fcb8b7004e16629","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-01 06:52:43.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update openstack.gerrit/openstack/requirements\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":2},{"id":"c887a82a40ebffdc3d5f54c73f76cdf0e2a5ba33","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-01 06:53:46.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update openstack.gerrit/openstack/requirements\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8b7b619b269e572aafa003522ae04b5067f64d66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 07:02:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cd8a671540bac102896e1d51e2885aa8d219f561","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 07:05:43.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"f28ef21f2acd998627f6cfe32a794741955f3b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 07:09:55.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/978321 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"6a4ae8679997f21fcf90da07e6d354482eea8edd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 08:34:21.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"068503129ccbe00b5b3f2061cb126d9c3553e31f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 08:46:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"120d10fd78952d255519960130a61936da1a8cc5","revisions":{"0e23d72c17a3ceec81d9ffa6298567e6498a24b0":{"kind":"REWORK","_number":1,"created":"2026-02-28 16:21:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/978331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/31/978331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/31/978331/1"}}},"commit":{"parents":[{"commit":"b43b65af569641ce9c9e8504ece11cc3f921258f","subject":"Enforce castellan \u003e\u003d 5.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b43b65af569641ce9c9e8504ece11cc3f921258f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 16:18:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 16:21:01.000000000","tz":540},"subject":"Use bare uuid for secret_ref","message":"Use bare uuid for secret_ref\n\nbarbicanclient can handle both full URL and the resource UUID since\n4.8.1. When a URL is given as a secret_ref, it ignores anything but\nthe last UUID element.\n\nBecause all secret resources are managed by barbicanclient we don\u0027t\nhave to pass the full URL.\n\nChange-Id: I218eaa3795d45fd8c67fdd3919e74ed03dc25136\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0e23d72c17a3ceec81d9ffa6298567e6498a24b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0e23d72c17a3ceec81d9ffa6298567e6498a24b0"}]},"branch":"refs/heads/master"},"b00bf1024cdb7eff9f6ae97183b8e68681fa8566":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-01 05:12:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/978331/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/31/978331/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/31/978331/2"}}},"commit":{"parents":[{"commit":"1f616120bf423764df7eeeeac2c1dd92d96243a0","subject":"Enforce castellan \u003e\u003d 5.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f616120bf423764df7eeeeac2c1dd92d96243a0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 16:18:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 05:12:46.000000000","tz":0},"subject":"Use bare uuid for secret_ref","message":"Use bare uuid for secret_ref\n\nbarbicanclient can handle both full URL and the resource UUID since\n4.8.1. When a URL is given as a secret_ref, it ignores anything but\nthe last UUID element.\n\nBecause all secret resources are managed by barbicanclient we don\u0027t\nhave to pass the full URL.\n\nChange-Id: I218eaa3795d45fd8c67fdd3919e74ed03dc25136\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b00bf1024cdb7eff9f6ae97183b8e68681fa8566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b00bf1024cdb7eff9f6ae97183b8e68681fa8566"}]},"branch":"refs/heads/master","description":"Rebase"},"120d10fd78952d255519960130a61936da1a8cc5":{"kind":"REWORK","_number":3,"created":"2026-03-01 07:02:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/978331/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/31/978331/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/31/978331/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/31/978331/3"}}},"commit":{"parents":[{"commit":"60eb8b0da7b8ab8e2a49b8140dc6a2979bea68db","subject":"Enforce castellan \u003e\u003d 5.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/60eb8b0da7b8ab8e2a49b8140dc6a2979bea68db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 07:00:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 07:00:31.000000000","tz":540},"subject":"Use bare uuid for secret_ref","message":"Use bare uuid for secret_ref\n\nbarbicanclient can handle both full URL and the resource UUID since\n4.8.1 . Using UUID allows us to reduce risk of exposing internal URL.\n\nNote that it seems that NetApp driver requires a link to full secret\nURL. This is handled by additional call to obtain full URL, following\nwhat was done in the rest dummy driver.\n\nChange-Id: I218eaa3795d45fd8c67fdd3919e74ed03dc25136\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/120d10fd78952d255519960130a61936da1a8cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/120d10fd78952d255519960130a61936da1a8cc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
