)]}'
{"id":"openstack%2Fmanila~978780","triplet_id":"openstack%2Fmanila~master~Icb02c880c305ad380b9dc9c0f96747640f396b51","project":"openstack/manila","branch":"master","topic":"bug/2143047","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2026-08-11 14:05:25.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-08-11 12:42:22.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-08-11 14:05:25.000000000","reason":"Change was submitted"},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-08-03 09:55:55.000000000","reason":"\u003cGERRIT_ACCOUNT_38371\u003e replied on the change","reason_account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-08-11 14:05:25.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2026-08-11 14:05:25.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-08-06 13:24:28.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-07-20 17:24:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Icb02c880c305ad380b9dc9c0f96747640f396b51","subject":"NetApp: Support force delete volume via share type extra spec","status":"MERGED","created":"2026-03-03 16:16:24.000000000","updated":"2026-08-12 21:44:47.000000000","submitted":"2026-08-11 14:05:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":60,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"978780-bug/2143047","meta_rev_id":"01ded00dbaa0b0f9b5ac0c2f0feb02cea1268695","_number":978780,"virtual_id_number":978780,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-11 14:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2026-08-11 12:42:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-08-03 09:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":2,"date":"2026-08-06 13:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2026-08-11 12:42:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 17:07:11.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-03-03 18:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 20:26:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-09 10:11:43.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2026-03-12 16:08:22.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-03-17 14:03:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-03-17 14:03:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-05-05 06:30:34.000000000","updated_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"CC"},{"updated":"2026-05-14 11:04:08.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-07-16 16:26:14.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-20 17:23:57.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-07-23 08:58:10.000000000","updated_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-08-12 21:44:47.000000000","updated_by":{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"},"reviewer":{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"},"state":"CC"}],"messages":[{"id":"08017308f6d9fc9b73e91f2eecfb59ed3c618b2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-03 16:16:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db4c8e2b806803edad56fd319d747f08dfed333f","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-03 17:07:11.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978780","accounts_in_message":[],"_revision_number":1},{"id":"234786e25afa4e82bcaeac438ee702e7cdfbb96e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:22:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f34d3128a517427780522ed41b07be0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4de4f98731463099d397ed7d4b83a5 : SUCCESS in 11m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91d9fd388a1c4096b51e71958de80d61 : FAILURE in 22m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd577508c319422cbbf38a8657b5f69d : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f87a386b6b3041b08419fe4291a04827 : SUCCESS in 6m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1526c3b9d3b94bd2b8cb168bac8e11da : SUCCESS in 5m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6b1b174107a411ba21fa3c20f38085c : SUCCESS in 10m 49s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7c256a0def8747c0a48fc171cb2b337a : SUCCESS in 1h 22m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1a4157c74d2441308654ed26e0467456 : SUCCESS in 2h 05m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8c3e94182e1e4876b803437920a2662f : FAILURE in 40m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ffec40d203c1470e862a7c0aa86668b3 : FAILURE in 58m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/46acc4e954fe42c5a1aafd85ab1c9ec8 : SUCCESS in 1h 27m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5065a8bfc012421fb8b016b7eaffc876 : FAILURE in 40m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/daf329602fa74d93a8ceb63b757c7127 : SUCCESS in 3m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/18e5602d637d4fe8a1940fd89da12d55 : SUCCESS in 3m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b314fefbfc714db8a0d6f32591b086b6 : SUCCESS in 32m 38s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9570411856014a9b954adce22c864b36 : SUCCESS in 32m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/16af636f5e744777a10c925b271094ce : SUCCESS in 28m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f0797b7c7d0f4341b4bb4f783ef9c164 : FAILURE in 32m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/d52206c6d5ea44a99995fb7d447eb7ab : SUCCESS in 1h 21m 05s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f4ed7921dde1435e9b2aefa47b11cf5a : SUCCESS in 57m 24s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3b883d0eb574f579cf4a006dd3229ad : FAILURE in 5m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dac12a0be7204e35bd083c8db772e1d5 : FAILURE in 18m 25s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dc8a6f3739d5431882c1915f9079efec : FAILURE in 14m 22s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/9284c488038245a6a6d58922b0ebbf0e : FAILURE in 4m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2545d05775374217a67d49f8b21ee091 : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0470c0cd43f9bbde93a88353b71d361c1175456f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-03 20:26:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/832950b/ : SUCCESS in 1h 35m 41s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/768a932/ : FAILURE in 2h 03m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6058789517924a0c6a25b2f66f62af9ab195db2f","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2026-03-09 10:11:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c2c3ce406518e26f2862d8f54a48c9121cb690df","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2026-03-12 16:06:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f9cecf2ef353be8b621b4dd2c750b02b4102009","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-12 16:08:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"39d89c3de122900696f6392b5d4075da7a0ec635","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-16 14:01:50.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2538b9cbe41dcd1f90bb649cf18b5df71a3a1f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-16 14:01:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b4b3d050afc7c2597ca6a85ab8b02b2052b5438","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-16 14:50:26.000000000","message":"Patch Set 2:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978780","accounts_in_message":[],"_revision_number":2},{"id":"5efa0f3aa7b7155a872c6e67d442810a470da424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 17:29:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a595858ee0364e0592391f1360ab364a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5103f16bdeaf4fb389d9ee04999dfaed : SUCCESS in 12m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/605cb7367c514920818de95b0582c989 : FAILURE in 33m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66c290eb6164e2cbdc4b67e563f3732 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aed53fa9912b4c259cde7d609e400e8b : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa211e66879f488fa457f75735aaead0 : SUCCESS in 9m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6bad3b5818e540a8a520f3fb2554588c : SUCCESS in 12m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d495453752948e28907896199a5c5a9 : SUCCESS in 6m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c5cefdb2ba7b4101a04fb2755cbef0e6 : SUCCESS in 1h 31m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e95eeccdfb5248508c4dbef974fb7345 : SUCCESS in 2h 00m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/afbd14a81a534b18a412f5788a4a5ef8 : FAILURE in 1h 08m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/91c7885aceb844db9dc7adeb2f95b005 : FAILURE in 1h 01m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/69792fbde8504edfa541b6e105acd865 : SUCCESS in 1h 29m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a8ae27a304024b4d8e8e25ad45dca35b : FAILURE in 55m 08s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b8bc168aad73451c8f7f5a22e9af442e : SUCCESS in 5m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b0e3fdd9ebf84ed083c238598f5bac68 : SUCCESS in 5m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9bbc5f7d90ec409c89d0c106bce97ecb : SUCCESS in 34m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/87b47819a2fc4e99a17de9fba9faad1a : SUCCESS in 37m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07c6b48081054279b7ecf04267787baf : SUCCESS in 48m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1e1de35c209c45b08a8f4686a7a4e77d : FAILURE in 57m 06s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/c2d036220f344d56bdfcbc7d441d3e13 : SUCCESS in 2h 54m 33s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/07bb5ae33dc741468160d555a5630404 : SUCCESS in 51m 40s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40df1780f3fe4dc5886a450913bd5171 : SUCCESS in 1h 02m 12s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ba37d9a03a784219ba0eeb89b33f5d53 : FAILURE in 25m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/63e1337e5d2a4752bb0854e590794a32 : FAILURE in 28m 34s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/55935f1d970a4f00a23bdd3326731439 : FAILURE in 5m 14s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/6fbee6b5f8cd4cdf8af7a1bc4bebf363 : SUCCESS in 8m 49s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b5b0b80a0cb546d2968f6aef184c3669 : POST_FAILURE in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0977b14d76fdd0cd4f55c19949619f3b9ce39c3b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-16 17:46:02.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"bceef56c6d9f0b40ebe25d9ed2b1f29e875430ea","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-17 09:30:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3cbc671ea6f0b135cb4e4a7f0c0a069567934042","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-17 09:41:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b9af424c869a6fbceac6d424108e53ac807ca8a","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-17 09:45:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e664a3c5e7369fc5edd2df060045e2e1386857cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-17 11:55:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"29458cb696ebf0da974db3b7dafbb958f78874ce","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-17 12:07:24.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b1f55cf0c4f737974662b551999573865e9c0ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-17 12:08:34.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"bdd7f5a388a0379f4a0e8d86091a55c40125c7e9","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-17 12:30:02.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"97694e43997a4f851d79b9b9bc0696d181fb28b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 13:32:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17f93492d58c4b0fba9c859227bf701c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b942584bf14fb4a6985f3f1af19372 : SUCCESS in 12m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/313dff1ed6f646048184b19e7a4e2a88 : FAILURE in 23m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4ccbd7ce2f42c2853e48a2ece2fbb8 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4484829e346e463288cabf8653388399 : SUCCESS in 8m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7844d64c0af4c9f9dc8445cdb549769 : SUCCESS in 8m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7383802cc2cb48a2b99a9dfe59cc9b54 : SUCCESS in 14m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/406b924334294f7ebe3f58d8f9cdaeeb : SUCCESS in 9m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/13f7b4cd48d04274a91c468219820579 : SUCCESS in 37m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/03ddb0a8e4f94ec0b4b81ed5fadccd95 : FAILURE in 1h 02m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/794992c02c4845888b5e6acc3091ab38 : SUCCESS in 39m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0ea371f444f44faa978081d8be459470 : SUCCESS in 1h 09m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/58ecacea820c45388c9eb2c98f9eba18 : SUCCESS in 1h 14m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d159c6dc239b4a0d874d1c47198d43c1 : FAILURE in 34m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fdeff8b9006d4a8e93f5695754a7f422 : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fad603c8e7aa4af2928499619ccdc915 : SUCCESS in 5m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9429ef6ec96646929bac47a910e1d034 : SUCCESS in 36m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f942d291a5f04129807d1eccf4095bea : SUCCESS in 31m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0fef1e84618e4d74ad76031bce194494 : SUCCESS in 32m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/552c0a581eee4a4b8e73648e34b20d6d : FAILURE in 58m 12s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/680af828909f4e72a8a3d7b9046363f1 : SUCCESS in 1h 17m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2a41336b9ceb406cbc853f4ddf16cb26 : SUCCESS in 40m 54s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60ad412dc1854bedbe958e05de3e9946 : SUCCESS in 56m 32s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/517c74e0b782430f8e8ffbb2433d9b10 : FAILURE in 15m 08s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b0436ebf528f435dadbb428907eb4f9c : FAILURE in 31m 13s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/0eb9563fc7a94336a025063d4532ed14 : FAILURE in 4m 35s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/d7dfdd5e8c7d4dbbad2827d757a6be5d : SUCCESS in 8m 48s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c265f80515bf4b8ea51368898a018e21 : POST_FAILURE in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9d37f830a7f8fdb5e2352d17a86aa7bf5e2187fc","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-17 13:38:52.000000000","message":"Patch Set 4:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978780","accounts_in_message":[],"_revision_number":4},{"id":"6e5a9791acba75964855792c1973825198274b32","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-17 14:03:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e8c79bc92078489695f5c94dd064b0701a04be3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-17 14:05:17.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"ff305cc62b0242a640cafcbc081ff1fcea42e19d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-03-31 07:52:15.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ff1e0ae44efee4cfc2b42589841c621479bfc1ee","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-31 08:48:31.000000000","message":"Patch Set 5:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978780","accounts_in_message":[],"_revision_number":5},{"id":"6584d6d2e50e3f1621bca67c11f528f1dcc38f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:23:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3453e100b128418597cb4169510d6c03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94903d2853294923a858e5f931f9f9ae : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa9c766f465144d188ef3eadd5a82c40 : FAILURE in 36m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd678875f19411e99ea6256caaaa37f : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bff215c52494e9a963d2e90b5a97de9 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc8a136e8a404de48c876b070246ee6f : SUCCESS in 11m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be17c0eee9274c6daa7ce857b97b7431 : SUCCESS in 6m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7927398aefb447fbbf2f431b3fa916d : SUCCESS in 10m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/010fd6bb3c924ccc8dcc231afaf2ec14 : SUCCESS in 1h 24m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c3259355ac3a469aabbecbee6aa25381 : FAILURE in 2h 22m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/995248856cd4429e8e09644f80534a76 : FAILURE in 58m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/85e94f135a34476c9734b20c2ae6c7c4 : SUCCESS in 58m 26s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2ad43ffd133947a5a36e9f018e89c936 : SUCCESS in 1h 33m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9308b73361bd4210ba76c0156bc33a35 : FAILURE in 1h 13m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f4bc18e2df7e4a3f829b896ff0afa144 : SUCCESS in 4m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6c3adb04efa40bdb8695c2973957ea7 : SUCCESS in 4m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/88e5dad2e57c4a299110e864cdadbff9 : SUCCESS in 35m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f98433a4ddce4e88afe032cdae10de25 : SUCCESS in 30m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/014fd2eee8734a958adf68e207307028 : SUCCESS in 41m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/81c06e30f04148709bc112951595d8d5 : FAILURE in 1h 00m 16s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/6370fb84dd6c411696db470b85cae1c1 : SUCCESS in 53m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a0f31bb1ef6c4807b012a738d605c790 : FAILURE in 42m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6185e9d7c05a40dba4231e87d4f3b728 : FAILURE in 30m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9d01b018196047f2abecc00082b22a4f : FAILURE in 31m 09s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c438bfe10b7f43d3abe26dfbdb228fc0 : FAILURE in 2m 32s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/a46c7258f02943d49449ac29843e923c : SUCCESS in 8m 14s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/51abe3940c6343fc8599b7994641e7e7 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"13eb6d01a043dadbcb03364f7d0b70dba465fcf2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-31 10:33:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/5/upstream-check-manila/manila-tempest-plugin-ontap-dhss/ea6738f/ : RETRY_LIMIT in 7m 07s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/5/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/70ff843/ : RETRY_LIMIT in 7m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"26c85c12c5c8f6528b53b788f131fd518f9d27b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-04-06 12:28:58.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8f935a3e04ad0cea45f62f645c845b9273cf3e2c","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-06 13:45:04.000000000","message":"Patch Set 6:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978780","accounts_in_message":[],"_revision_number":6},{"id":"992a01019531d8d5ea009a7f27549c5b3b0beef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 15:34:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e228b75f7e34c5c9d12a195a600f297\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f18ad738204c74b99127fb48811760 : SUCCESS in 9m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0f1f5db133a40b390445e2a1aa45809 : FAILURE in 22m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98a291353674582aef851f10b35e306 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79053808cfd6408dbbc814d7116c5237 : SUCCESS in 9m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e6e52ee26fb4a7a9f68c9d6720c5e78 : SUCCESS in 6m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/263fb0d77fa8453f9fc23a5634012601 : SUCCESS in 11m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73b93b0e4b704cf99c7f06d715915d1e : SUCCESS in 8m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d0749ef2758048f69778a70d594e3273 : SUCCESS in 32m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4697c11d87d244a49942d940f06b6f09 : FAILURE in 2h 58m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c6710f0b05f4432f8b01041a17d6de0b : SUCCESS in 1h 24m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/99e6a53b8f5c42d0874f20eb67bfd8a3 : SUCCESS in 31m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9c58db7ecf264b97bccc9e23f50561aa : SUCCESS in 49m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7a6cf9a063994950b7fc7d2716182d43 : SUCCESS in 52m 43s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7fb1d1e5871d454180c56c7b3f42da4a : SUCCESS in 4m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b152983cf5b4308bed33fb626c1255a : SUCCESS in 3m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9ec6f57ca13b4d4ea99169d5449f6928 : SUCCESS in 21m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/08cf5331ba7740638c8130ad64e525ac : SUCCESS in 26m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/72fc2016841b4a8181c6f4a9b44a8b03 : SUCCESS in 37m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d79a64843066487fba0e97d5cecc79b2 : FAILURE in 45m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/156cfdc1b6364d12b5be8521ae86a624 : SUCCESS in 49m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f34b9fcc1acd4cef941653608ed7d877 : FAILURE in 34m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/946c3f0f64874238b9453b67fb49daa8 : FAILURE in 18m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a5dc7cb0c9824312afbedb2d047a4a5e : FAILURE in 28m 14s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/007bdb3d6d6c4cffbc7a8c1f4237b5ee : FAILURE in 2m 55s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/f4c8ebab4cf74aafa2e00e9389b1813b : SUCCESS in 8m 06s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1b6de9db4f8b457f8fe85d477ddc35c2 : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f5b1b8cf5c8588cf2c2f60155574c67d3b36dd28","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-06 17:34:23.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/6/upstream-check-manila/manila-tempest-plugin-ontap-dhss/ecb073a/ : SUCCESS in 1h 23m 56s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/6/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/d74b679/ : SUCCESS in 1h 57m 40s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/6/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/fd2dd9a/ : RETRY_LIMIT in 8m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"306a8d2584d0e78c7063b2ca9f25acc52640744e","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-05-05 06:30:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"239a091ce45bfd325a8c04bfeaaf425b01edb310","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-05 08:41:03.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/6/upstream-check-manila/manila-tempest-plugin-ontap-dhss/b617403/ : RETRY_LIMIT in 23m 35s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/6/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/85ae35a/ : RETRY_LIMIT in 15m 47s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"2198c48d62b3dd3194c8d6a3ecad03a451b2b676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-05-14 11:08:33.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"a554c662c6c561775912eec704e519229f08c109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 13:09:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1cec34c20c94609833912e67ae1127a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3be99b5a924198992a8e90fff9bc1f : SUCCESS in 12m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e78738e97124c9cb949c0676b54d46d : FAILURE in 27m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cbb048ba4a14713857fbfc2fca943b0 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/273cf509a1a44afca5a2cb7f5d480fde : SUCCESS in 8m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27aa67c6a4984e8f9ace1d5dfb08b0c9 : SUCCESS in 6m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c6d6d9517c846cd921026b6a9984156 : SUCCESS in 11m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5648cec236dd46b095b01b04dde255a5 : SUCCESS in 4m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/260fb0dc51ae4ef7a5131efa510b06ac : SUCCESS in 44m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/047a6c368b6b492fbfd6d9ba1617a413 : SUCCESS in 1h 40m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a74dcae799d94f09b23bfdee247ee96c : SUCCESS in 1h 21m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/59eae1db629c497ca0a9d3e53e29ddab : SUCCESS in 27m 54s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/87d8653b9eed429baedb00b040a831a0 : SUCCESS in 1h 49m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/02c7bfe5ac86444489c89f1ace9b89de : FAILURE in 1h 15m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/26dac470bce14d15a9a1f93610f61239 : SUCCESS in 6m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e578ed6e1a894c2dab7ee76f3dad879b : SUCCESS in 6m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/28a2a9bc80ba485ab09141a7f31e82a9 : SUCCESS in 31m 38s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/99c29348549d42c0a0e304aba00710ed : SUCCESS in 29m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1637e65b46fa4f3caee4fa8fc3a82e73 : SUCCESS in 34m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b81aa16cf0a743b698c556c8257c86aa : FAILURE in 12m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/67aa05cefb6a4a5ea394a7c85464f899 : SUCCESS in 1h 25m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/07729b073bfd4f1cac011d29fa809dcd : SUCCESS in 1h 10m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ca43a6158b4144f4b3c82191980c13fd : SUCCESS in 21m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0c7cfbc36135401981e13bb2371e7a9e : SUCCESS in 35m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/5d991865e2e54993b863a8d60a51b2a9 : FAILURE in 5m 25s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/bb4bef8bb37e4d249ecc563852b1135a : SUCCESS in 7m 52s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/9be3143605dd4c9699413e21d90aa669 : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ad87620881954f467a8dc297764d59a67f7a038e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-14 13:19:38.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/7/upstream-check-manila/manila-tempest-plugin-ontap-dhss/48cb903/ : RETRY_LIMIT in 7m 10s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/7/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/c93697e/ : RETRY_LIMIT in 7m 05s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"40e2f1ff87ac9e0d46b9cfb07a2cbcbba70c725b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-05-19 14:12:36.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"20880b5ada692893e94ce08908665e46d0bd166c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:43:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad07500f63094bda8171367158ee29ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc4d79eed834c8aa7a53f0ae6a33369 : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e76b65cc55e48fc8a53f087197654ba : FAILURE in 31m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b344c8f04cdc42b087cd3fb6e398f736 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a773e88d61004635bf8eac01f3344c23 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8af81672d4234c14a9a70fef71120f9d : SUCCESS in 7m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/51d6f2ba86e848acbfdbcae9927df243 : SUCCESS in 13m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b67b192009244ac6b25a80985ee3bb9e : SUCCESS in 7m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a1e347a52db04b1b9a490b2963ad7afd : SUCCESS in 1h 13m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f2a481cd4c094d51a501b3da9126ae6c : SUCCESS in 49m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d0b6e8b3e58b4ebea46cb1de4c729545 : SUCCESS in 1h 28m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/633558b003a94caaaf91517d21e16935 : SUCCESS in 1h 04m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b6ea7b59729345b5823432ba906cf062 : SUCCESS in 47m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f3ab180c6874296922722c800fecd10 : FAILURE in 34m 24s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bcc51e4230ce462c9567ad07a63b5207 : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f68710ad24240b5b5d1f3055a981bcb : SUCCESS in 4m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/452dc3384221466a8953498f7326ca1c : SUCCESS in 31m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cf535e1644234f95a06dae37b8bb3b78 : SUCCESS in 19m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0e1de615b2a54627817d878046d3f8a7 : SUCCESS in 31m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/37056f24c34448298b07bd3abd27e002 : FAILURE in 18m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/083be599b9744ee8bb60e13477e3704f : SUCCESS in 1h 16m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/39d53e1d5d1f43c29042595f067d2035 : SUCCESS in 1h 08m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a6da1510e79e446a8a51fff884d102c3 : SUCCESS in 33m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0d1909d0c00b4960ba7dee5f820b4112 : SUCCESS in 28m 49s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/bab491038dd1472ca9d2e61bd4111461 : FAILURE in 2m 34s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/924bcf7deeb347afbfba037af0d6eeb7 : SUCCESS in 5m 55s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/dc9cfaadd8844a0d8889aa69e76f5c2c : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"41fa7ae4f30b4e844459e9c76cde6d513374d8ef","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-19 17:32:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/8/upstream-check-manila/manila-tempest-plugin-ontap-dhss/a2bd090/ : SUCCESS in 1h 13m 40s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/8/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/bad39b6/ : FAILURE in 1h 46m 12s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"0aed1b937c5170da86cf04820f7b1c2f3bf230a3","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-05-26 14:13:54.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"13c695d89d14f2da04c19166c04b48f443e9311a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-05-26 15:45:01.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"20d2ac8beeef3eb41887452fec8ffeeb7a3fb53a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-05-26 15:45:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0be271371290c55ebd35168c449826d5f0ab9972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:37:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08727b1774f94c179eff95813bfbe770\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bf63848d11491cba4d7189e60a9f72 : SUCCESS in 12m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9ed5968a0dd4967a62907ec39235096 : FAILURE in 25m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895e1e6daa524087bbad9c6965e2ebfe : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baeb34ad4ec04d29b8a4628d83ac751d : SUCCESS in 8m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ccb14c4586c401ab3888d3f739faf11 : SUCCESS in 7m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13782d72154f4c32a61fc034a127c88f : SUCCESS in 12m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4541d90f38bd4451b6245ab2386c0264 : SUCCESS in 8m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/afc3fb965bf04645bbe30f7cfc6b8a08 : SUCCESS in 1h 10m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6965c7af86f84b61a2c5e561c21a7c43 : SUCCESS in 1h 46m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/044777cf744747309d87cf8b4f57a97d : SUCCESS in 1h 20m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/11ec5f102b284ff7858d76e420a682fc : SUCCESS in 28m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cbf9bffc750f4ab4be1770cfef313928 : SUCCESS in 1h 37m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9d00408e92134154b7ba981893f8581a : FAILURE in 1h 08m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0535e9786b624f51a6b05b5506c1a8d0 : SUCCESS in 4m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c4d58595a614caba6882ef10fadaaa7 : SUCCESS in 4m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1ed6a4832a4f4f629954c7165e961dbd : SUCCESS in 30m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f827a99361b846d3bbc08540cc553538 : SUCCESS in 22m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/22c1a791b9f44aed95b52b5210e1c4ee : SUCCESS in 42m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a7aac872a60045ecb99317da59bd2e52 : FAILURE in 19m 30s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/a3bbfe17d77e4cd29792ffaebdc0bb53 : SUCCESS in 58m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/13e3b471dcec421cbdc1a70ed9c324d7 : SUCCESS in 53m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1414084343b0480cbf31cf143815763e : SUCCESS in 36m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3a8c9a9fae1b44659bb2a65d1b2913e4 : SUCCESS in 20m 11s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/694dd9567ea142118f80c5032c07dcde : FAILURE in 4m 59s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/c133ce5bbda4498fbfded3ac2bcfef4c : SUCCESS in 9m 20s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3ca75c5bf2b74c05b15a4231c879e205 : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"86bb4de4341d151e5fdc4810b6fc8dfb7895625c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-26 19:05:05.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/9/upstream-check-manila/manila-tempest-plugin-ontap-dhss/ddd6b93/ : FAILURE in 1h 17m 33s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/9/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/4965ffd/ : FAILURE in 1h 25m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"0755271529128355040b6518faebfe453dc7c3bc","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2026-05-27 07:26:14.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"76449710d4f35c60c1b68d408e63847819ed2a99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-05-27 09:36:54.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"49a8a7e293a399ad4498b9bdb1b7c1b95dcf3654","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-05-27 09:36:56.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0584908bc2e4bfe05035f715d2808a3afe312bfb","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2026-05-27 10:35:36.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"6e8bd4a6f358fdf4935e2375336e1ec0788ce179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 12:02:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d137707b8a4edaa5efa74bc12c6661\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019a23949e6d47a3b5ee87d1c87fdb4f : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cd7ff95fb4e4847b3a3d74bc454ac8c : FAILURE in 26m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41230fd167046b1a3c59d6fec0686b4 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e092214ce8c6490892a405750b283480 : SUCCESS in 7m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a684386a893345f48b256b6ab2e53202 : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1af0d1bd45a41acba3b561b4b86d24a : SUCCESS in 12m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e887124859249bbabe5f0f53debd3b1 : SUCCESS in 9m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4c374f586e30498a9326e6904d16aacd : SUCCESS in 57m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/70d9e7c609754609b396cbf8e9893c04 : SUCCESS in 2h 09m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f613440418454c829987445ea35963cb : FAILURE in 2h 12m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/01bd88798e104af9b1b29ec27535e222 : SUCCESS in 58m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a0965fed3f314ab69b2a2a637c815985 : SUCCESS in 1h 26m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5d26da16be544080816b03a786acf90b : FAILURE in 1h 04m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3324fa1382364c57a4346cd9670295c3 : SUCCESS in 4m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/474ae62e77b4487abc1f9f7e75f4188d : SUCCESS in 5m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/806682baee0a43b1b3ddc663c051033e : SUCCESS in 24m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7f400301307441bf86360db5023e8b65 : SUCCESS in 24m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f92426a035be40ceb71353912a5d3ae1 : SUCCESS in 42m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/77058c5d54c5416e8aa44ce044893708 : FAILURE in 10m 13s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/1fe4af6a51b3453395cb39991701a9b2 : SUCCESS in 1h 12m 27s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5088bd36b0904d5eaca30281f76cfea9 : SUCCESS in 1h 02m 17s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2c3e93430e2f4e0db74819cae2a44b7f : SUCCESS in 33m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fd01d0f9ff374fba852e46472346a463 : SUCCESS in 36m 17s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/4f398a8366c8420883a000b254e4118f : FAILURE in 4m 37s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/f2f500b9afbc431085b1e493af29731a : SUCCESS in 9m 05s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/9a480ec153b244e8bb4fb561b64e3b44 : SUCCESS in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"67a9646d5862d80329daf316ffc2a177a1fe32ac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-05-27 13:38:08.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/10/upstream-check-manila/manila-tempest-plugin-ontap-dhss/c4c0e38/ : SUCCESS in 1h 22m 29s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/10/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/ab41cb6/ : FAILURE in 1h 33m 26s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"577c7be18c9dde979b11550200120dc4f1e21751","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-02 11:49:07.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"aded02c5842a66ab9d76b044e50a58c6ce58137f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-06-02 16:12:17.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d62cd23765ff7cb79cff8779269931923dcc823a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-02 16:33:05.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"3e50f54b66380ca433e17c355c56fe68315e4620","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-02 16:33:06.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0280ff9f2ddf8e68bae669024f9339c638fe8cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:10:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f648d421cb6c49f49d4436726fc9f0ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7dcac076f1d4357b3b1d4213b48f8a8 : SUCCESS in 13m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edef7d12c4f442169bb222c15125cb22 : FAILURE in 26m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d773038fe0434a6e8bcab3110d6ac5e9 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e2685aed6564c25b187512f6faa1f95 : SUCCESS in 9m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af16bde00a404368a6a2810a4511c1e6 : SUCCESS in 5m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ee66b8e62094f98b3ad023ff8a8136c : SUCCESS in 13m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f7cf34eef234442b298ada8d904e7a1 : SUCCESS in 10m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/53242b4e7d4c45b88b100c91fdaf7392 : SUCCESS in 35m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d698e48da9d44454847aaba8e70edfdc : FAILURE in 2h 31m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/63bbd8579dba4624b2ddec06141063d7 : SUCCESS in 1h 52m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/440213fa836d485e9cc2ceac463f1579 : SUCCESS in 44m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7aace00ab15c46cc8b4671a2c4894349 : SUCCESS in 1h 30m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3173ac8f58dd4fca903329576c150872 : FAILURE in 1h 28m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a18f7f520c6243b1a217db13a3c52dee : SUCCESS in 5m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/61856121b8b24260b06e84f423ded9a5 : SUCCESS in 5m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1ceee71e14de42c68857091cd5a7a871 : SUCCESS in 35m 42s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/506fd736164241fb866056e929a8ffa2 : SUCCESS in 28m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/98afd7e9dc494c23b8b4b5f8a54cccb0 : SUCCESS in 39m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e53c647fdfd2467ea5f8ae61463c4a6d : FAILURE in 9m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/1d2d1c4ba56a459b818afba3af563c85 : SUCCESS in 1h 16m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cd3d1ee3d3274ee794d8e004b3d6d417 : SUCCESS in 50m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/797adf53a6bb4de78b56dd2aa3e1417e : SUCCESS in 15m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fd665cf80242423f88b042311791cc95 : SUCCESS in 26m 46s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/25a3ab5ac993427a951aeab400aaba01 : FAILURE in 5m 17s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/fd5ad5058fae4d7aaaebff72eb3078a6 : SUCCESS in 9m 03s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ca598ce00ea140e7833b6e22a653d3ed : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d550af5807b69bc20711c39811a7c8f85e93f258","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-06-02 21:43:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7598e4f78815fc05973596d803a19a726575b581","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-03 03:14:28.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/11/upstream-check-manila/manila-tempest-plugin-ontap-dhss/998f093/ : RETRY_LIMIT in 8m 05s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/11/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/2abf8f9/ : FAILURE in 2h 00m 44s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"2df60bd35da3fd41b86f0fb45e22a2edd68291e6","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2026-06-03 05:56:23.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"375cd1c0e87fece3667ba9b7377ae26bae77ec3a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-03 12:00:02.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fde48116011ff53fa622664c2a865be65d0bde73","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-06-04 15:14:46.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9ad302a8ec5ff6e2c0dffb5ae4e850ad74602f2e","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-08 12:36:54.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fa28f239a321de37364d170365e032ce49872b98","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2026-06-08 12:57:40.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3af3af908b3f2587a5028174ad19cd90efb8079a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-06-08 14:22:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dd55ce1a87afaae0f4696d0b5d528f51bfa7f128","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-16 07:34:13.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"92e20b825b11acb4c782716f8e67a71bf883e691","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-16 14:33:01.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4514090bddd4e2665f92f17db2426ca6f9d95d22","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-16 14:37:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":11},{"id":"46bce5023d26e22089cc9dd0290560bad9005069","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-16 14:37:49.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b1dfd8b57d384795787f7f816c63313b8571f00b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-16 16:04:51.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"fe738feb6a4b75b4008d034ea9f7d6d9d60e6599","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 16:26:14.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/916266fc94724461b4dfcfff25c4ae87/","accounts_in_message":[],"_revision_number":12},{"id":"c997ec5d07663a5e8951bd4482e333d48d662aaa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-16 16:26:43.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":13},{"id":"903425c7c538a0ac180ea496d1ab6b9b1773bd28","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-16 16:28:53.000000000","message":"Patch Set 13:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":13},{"id":"e92eec055ba05bcecfd05a13fd6a2e50a69c80cf","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-16 16:29:04.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"49b9c49b4cc653e732579cfd93328e90700561f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 16:49:50.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/e523237b0b21480eb900471170f1e09a/","accounts_in_message":[],"_revision_number":13},{"id":"f078e1782de587779b070290b4802da4ddc3a2d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 17:10:20.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e523237b0b21480eb900471170f1e09a\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/959caa14e18f46bcb3ecac352bac9f3a : SUCCESS in 32m 05s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/60d5f4bb4c3e4fa4bc45a66f08e6eef3 : FAILURE in 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8151ef48185a8a11f2e7fd6d7e23c4ea2453b126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 18:36:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3024e76e3aa044f0b0abe889e3811544\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ced0e883e4420dbad26d7ad49e1387 : SUCCESS in 12m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01028b16dc31407eb37f270dbff9244e : FAILURE in 30m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c32adebb97343a68c491a11c78a6f7f : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4bc42c908264108881703354d19d0c4 : SUCCESS in 8m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1bb8e748e5143798c5d4c32379739d4 : SUCCESS in 10m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f9558d72ef5c46ccb7490988fba04109 : SUCCESS in 11m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7737112a41a24c3c99d322e5cb42484d : SUCCESS in 10m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5efa1e38372f42b5b4ab473c02f2f78a : SUCCESS in 1h 11m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8ddc0e99c9a34c70bb657e0502d7bd96 : FAILURE in 2h 03m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d74af975df2c4b5aaacd0702337b04ca : SUCCESS in 1h 17m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/550b1ca0c5674ffcb1a0c726aa8e6832 : SUCCESS in 1h 14m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6fa82db2656045579f52626ac69e88db : SUCCESS in 1h 37m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3362a599f0fd45f09fc8dd15d6acc253 : SUCCESS in 1h 26m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/aaf90065fb664da0b9adeb208f6ce269 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22e76ee8d2364f2da60eea6ec8ddcf47 : SUCCESS in 5m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f117cf28549449619247aa75faa18c09 : SUCCESS in 36m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/98c86e10059b4589ac2c6fb4e1c80216 : SUCCESS in 32m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3779468779d24fbe994c2e6075e7b44b : SUCCESS in 35m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/78da6c5f83724c238619bbaf29075e50 : FAILURE in 55m 06s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/6662089abfd34284b20913a71def6946 : SUCCESS in 1h 18m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/61c7b61ce205441b93424d137e4c42a5 : SUCCESS in 1h 17m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7b6cec1206f14a98b7ab57be06097f84 : SUCCESS in 40m 22s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/97ce35f3fd1b4c5c8f1162ba9af6ff9f : SUCCESS in 23m 56s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b07506c24cec41ada7173ddffda0ed7b : FAILURE in 4m 33s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/465e68e437ca436cb81c2e538db4b823 : SUCCESS in 9m 33s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/82bc936e691f4ae3a13b6862fee6ed3d : SUCCESS in 8m 42s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"c7f77286e30cfa4cf2c03a4c0acb92a8f6303067","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-16 20:11:08.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/13/upstream-check-manila/manila-tempest-plugin-ontap-dhss/9cf5206/ : FAILURE in 1h 23m 40s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/13/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/765366c/ : FAILURE in 1h 32m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"36af9797b9a6299be4defae0412648f7e1afdf72","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-07-16 20:53:23.000000000","message":"Patch Set 13:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978780","accounts_in_message":[],"_revision_number":13},{"id":"022c16bc35a6966eba28e1662d0ccc690e55ff12","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:23:57.000000000","message":"Patch Set 12:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/161/ : ABORTED","accounts_in_message":[],"_revision_number":12},{"id":"1343887dfba826e8e43020dc0c2fecedef28dc70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:24:13.000000000","message":"Patch Set 13:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/162/ : ABORTED","accounts_in_message":[],"_revision_number":13},{"id":"73f83f7201a64cbe33448da724037fe5f379fe78","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-07-23 08:58:10.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1ac65f47930f8c24ea0bde5c83b45a37d4da9ffc","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-23 14:08:12.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0f3b2f2d28649a2c3757cc187e43ee11c3d975cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-30 13:30:51.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"134a051f2a8bdae161791aa2b0292bc9bf62b73a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2026-07-30 13:30:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5547cd0f8dcac6c4a03938a62874232775beccd6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 14:09:33.000000000","message":"Patch Set 14:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/d4fd4c05ff4a4a658743337a6842d97d/","accounts_in_message":[],"_revision_number":14},{"id":"a647458d5ba114faf06fe0e416686006fa6a1ec4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 14:19:59.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d4fd4c05ff4a4a658743337a6842d97d\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d15725b14fc1413995c785ba93d74944 : SUCCESS in 36m 44s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53ad889d60cc4de79e3bb4b303ef6eb9 : FAILURE in 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"674ade9d7e8cc77864ed6bb913f880c9efda8ceb","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-07-30 15:25:20.000000000","message":"Patch Set 14:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978780","accounts_in_message":[],"_revision_number":14},{"id":"58dbe52798d866bbb4a52517fd0d9995d14b3dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:17:47.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99c39352c18a4528850b4eb36696f6af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929fdbfa16cb47f1be6ce2aa7c2d383f : SUCCESS in 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5554af931f634ba88db8e00aa1a5576b : FAILURE in 28m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2aab733453f41e697d354563740eaa0 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/864e7eb692194083b2d45c7625c13690 : SUCCESS in 8m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76d843f580514c73a3c72b9abdf1ddc9 : SUCCESS in 5m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ac940d687c4543589357c7f1026a3287 : SUCCESS in 9m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db286851c8204258ac13cc97be651ccd : SUCCESS in 9m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d8e14a92caa24524a9a98ad9bea9e608 : SUCCESS in 1h 11m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ee1193e45181487abe43bdf8197d50c5 : FAILURE in 2h 36m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/537bb154b45e4135b0e931eb6d9015aa : SUCCESS in 33m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8f96fe926ef5416487881597159ade0a : SUCCESS in 1h 14m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5c8e58e803494694a24f703a76680471 : FAILURE in 2h 15m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/87cd8218008c44278a65b9002d708276 : SUCCESS in 1h 23m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0bdb2c1b48cb45e08cd17bd4f65ae740 : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e8ea7634a2d4c578645b0dca1f14bb6 : SUCCESS in 4m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/87c6d508e2a84c13a7738fda3fff0bc8 : SUCCESS in 31m 20s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/068da626dd12436392d16c8cbbb8642d : SUCCESS in 32m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/78d610c9809544c3bbbb9529420cf967 : SUCCESS in 38m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1883048a65ff43f5a49663a5e3fd52c6 : SUCCESS in 55m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/cb31a0cd26c44b478076a545e21486a3 : SUCCESS in 48m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/59a003e1f99e4c018e603bb603baa604 : SUCCESS in 1h 00m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7e40483a83f641c6a32ee3a6ce701fdb : SUCCESS in 18m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ebc1541c13044ea0a1b520cee014f99e : SUCCESS in 34m 48s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/cf0122bb685a4fe1964dc14121e737f6 : FAILURE in 4m 02s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/d20e302f4cfc49acbd5f7443aae4f68c : SUCCESS in 8m 10s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a08119e960b843728f3d2b4e69c1288c : SUCCESS in 8m 36s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job manila-tempest-plugin-lvm-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"932e0dac3fd7b6a31e7489c5264580da83031edc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-30 16:26:51.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/14/upstream-check-manila/manila-tempest-plugin-ontap-dhss/5b5459d/ : RETRY_LIMIT in 7m 30s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/978780/14/upstream-check-manila/manila-tempest-plugin-ontap-dhss-rest/e8c5913/ : RETRY_LIMIT in 7m 31s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"108f219e7e8ab516dc527e4d11a08f114bc49d74","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-08-03 09:55:55.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6c8d1dc89101c23ee02582f99bbbd6c0a4bf3550","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-06 13:24:28.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8d6d51f8be076a3336589b102338886a9fe9f123","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-08-11 12:42:22.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"450dbf5d4dad3cd336afe2947c6a791e49316872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 12:42:52.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"cc4a8da8e614f2ad6bb560a9107295d7430c5d8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 14:05:24.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e409ff9fa754c10a87ae368b9608a91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9840b8ea9ab84b08a231e745da4d08f5 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08da4d7fdd4c40f78d60813da6eb217c : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edd9d2768b9744939619036743ebe714 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01568382e27c4790893bfd93df2e0e03 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57201559c8854176a54bac645b5bc1cf : SUCCESS in 9m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/af6b61f280ad4a51b9043e62518cd517 : SUCCESS in 1h 17m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c40a416c9aa543e79ce8b826d906b642 : SUCCESS in 32m 59s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/0aa3b4f5cdb944f4bf184cd47a538d66 : SUCCESS in 33m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/338fd6749b0142bc9018e5cb8f6424b7 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":14},{"id":"5d58ee0664d9ed18a59ac4f34e3f51e4ab46d870","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 14:05:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"41ff418f591a26d3785807da8f9cd8470404b052","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 14:07:06.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891e60649b6949f8830349b079c9bf3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304a513b4d4c4a668f00e855a86c8557 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06787d4dea4c4552b32997bd2134bc49 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":14},{"id":"01ded00dbaa0b0f9b5ac0c2f0feb02cea1268695","author":{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"},"date":"2026-08-12 21:44:47.000000000","message":"Patch Set 14:\n\nExaScaler Manila Driver https://d3krvozkty9bs0.cloudfront.net/manila_build_142/index.html : SUCCESS in 204m","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"43c858dc75b44e5cef512d3d4d30c30fd9472826","revisions":{"ea711df9392ac749dd3e6310a1f6776275348c11":{"kind":"REWORK","_number":1,"created":"2026-03-03 16:16:24.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/1"}}},"commit":{"parents":[{"commit":"773dc6c428854dc4d3ae27b986f9f404cef01eb1","subject":"Merge \"PowerScale documentation update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/773dc6c428854dc4d3ae27b986f9f404cef01eb1"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-03 16:12:57.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time\u0027 which can be configured\nto non-zero value to specify hours. If share alive time is less than\nconfigured value, share will be force deleted. Config value 0 signify\nshares will always go in recovery queue.\n\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ea711df9392ac749dd3e6310a1f6776275348c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ea711df9392ac749dd3e6310a1f6776275348c11"}]},"branch":"refs/heads/master"},"b63e02bb34be7f00912bc0a4b91c6551120ebbd5":{"kind":"REWORK","_number":2,"created":"2026-03-16 14:01:51.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/2"}}},"commit":{"parents":[{"commit":"c0151ac7c67737b20822936ebd744235512ff8e8","subject":"Correct classifier typo in pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c0151ac7c67737b20822936ebd744235512ff8e8"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-16 14:00:52.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time\u0027 which can be configured\nto non-zero value to specify hours. If share alive time is less than\nconfigured value, share will be force deleted. Config value 0 signify\nshares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b63e02bb34be7f00912bc0a4b91c6551120ebbd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b63e02bb34be7f00912bc0a4b91c6551120ebbd5"}]},"branch":"refs/heads/master"},"82b093d0e68e95f77e722ad4c0c8f5d8185fe547":{"kind":"REWORK","_number":3,"created":"2026-03-17 11:55:46.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/3"}}},"commit":{"parents":[{"commit":"28eddeddbb6cc2cfb2957e23a8d8e63c3c82e191","subject":"[Doc] Annotate max api microversion in Gazpacho","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/28eddeddbb6cc2cfb2957e23a8d8e63c3c82e191"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-17 11:33:29.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time\u0027 which can be configured\nto non-zero value to specify hours. If share alive time is less than\nconfigured value, share will be force deleted. Config value 0 signify\nshares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82b093d0e68e95f77e722ad4c0c8f5d8185fe547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82b093d0e68e95f77e722ad4c0c8f5d8185fe547"}]},"branch":"refs/heads/master"},"74c4af3db219c99facc5e007a2985a432493b944":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-03-17 12:08:34.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/4"}}},"commit":{"parents":[{"commit":"28eddeddbb6cc2cfb2957e23a8d8e63c3c82e191","subject":"[Doc] Annotate max api microversion in Gazpacho","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/28eddeddbb6cc2cfb2957e23a8d8e63c3c82e191"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-17 12:07:49.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/74c4af3db219c99facc5e007a2985a432493b944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/74c4af3db219c99facc5e007a2985a432493b944"}]},"branch":"refs/heads/master"},"94456b15ddb7499ae13ca0fb259206354e62407e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-31 07:52:15.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/5"}}},"commit":{"parents":[{"commit":"7a7b225371ae05386207707cd55fe237a6e2ed92","subject":"Merge \"Fix service listing with a wrong status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7a7b225371ae05386207707cd55fe237a6e2ed92"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-31 07:37:24.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/94456b15ddb7499ae13ca0fb259206354e62407e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/94456b15ddb7499ae13ca0fb259206354e62407e"}]},"branch":"refs/heads/master"},"4fe7d9875a07817e36685c141d085cc389c7ba71":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-06 12:28:58.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/6"}}},"commit":{"parents":[{"commit":"54ac755d93d6c4128ddb8814b5d7f3906234fff4","subject":"Merge \"Fix indentation in share replica API docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/54ac755d93d6c4128ddb8814b5d7f3906234fff4"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-04-06 12:10:55.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4fe7d9875a07817e36685c141d085cc389c7ba71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4fe7d9875a07817e36685c141d085cc389c7ba71"}]},"branch":"refs/heads/master"},"e0490d9dfefea6878825f08646a6991383310ca0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-14 11:08:33.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/7"}}},"commit":{"parents":[{"commit":"bfa1c563e2349f55a663fcf6dd7aee2d9f6c31fe","subject":"Merge \"Use systemctl command to manage service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bfa1c563e2349f55a663fcf6dd7aee2d9f6c31fe"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-05-14 11:04:54.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e0490d9dfefea6878825f08646a6991383310ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e0490d9dfefea6878825f08646a6991383310ca0"}]},"branch":"refs/heads/master"},"4462683f689ce6e7655a86f44473befeec0bb33a":{"kind":"REWORK","_number":8,"created":"2026-05-19 14:12:36.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/8"}}},"commit":{"parents":[{"commit":"2a78e94a652ddd7e2b666d74967f49122dfe4ba8","subject":"Merge \"NetApp: add share_id, project_id, name, type, proto as volume tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2a78e94a652ddd7e2b666d74967f49122dfe4ba8"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-05-19 13:30:41.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4462683f689ce6e7655a86f44473befeec0bb33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4462683f689ce6e7655a86f44473befeec0bb33a"}]},"branch":"refs/heads/master"},"9702f9b1324fcd567e712c2ea194ac4a3ce74df5":{"kind":"REWORK","_number":9,"created":"2026-05-26 15:45:01.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/9"}}},"commit":{"parents":[{"commit":"4a392b68344dd3e3010ea3c0c74681d81589872c","subject":"Merge \"user-guide: Switch manila CLI to openstackcli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a392b68344dd3e3010ea3c0c74681d81589872c"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-05-26 15:44:53.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9702f9b1324fcd567e712c2ea194ac4a3ce74df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9702f9b1324fcd567e712c2ea194ac4a3ce74df5"}]},"branch":"refs/heads/master"},"a0b0c8434972543123ad1955517208d00ed1fb15":{"kind":"REWORK","_number":10,"created":"2026-05-27 09:36:54.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/10"}}},"commit":{"parents":[{"commit":"4a392b68344dd3e3010ea3c0c74681d81589872c","subject":"Merge \"user-guide: Switch manila CLI to openstackcli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a392b68344dd3e3010ea3c0c74681d81589872c"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-05-27 09:36:46.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a0b0c8434972543123ad1955517208d00ed1fb15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a0b0c8434972543123ad1955517208d00ed1fb15"}]},"branch":"refs/heads/master"},"54b4ad73c2d1e1988eb7a1ffa0fc6a4b4ac3d410":{"kind":"REWORK","_number":11,"created":"2026-06-02 16:33:05.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/11"}}},"commit":{"parents":[{"commit":"a6bcad3b38a69d2e267b42282cc54595d48f39cb","subject":"Merge \"Improve share snapshot and access list performance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a6bcad3b38a69d2e267b42282cc54595d48f39cb"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-06-02 16:29:42.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/54b4ad73c2d1e1988eb7a1ffa0fc6a4b4ac3d410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/54b4ad73c2d1e1988eb7a1ffa0fc6a4b4ac3d410"}]},"branch":"refs/heads/master"},"84ace8fcfb98832014725bdaa3a788d760b60729":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-16 16:04:51.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/12"}}},"commit":{"parents":[{"commit":"19f3340ad8da51d0bddd080874a4606d9d1a164e","subject":"Merge \"NetApp: skip efficiency settings on data-protection volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/19f3340ad8da51d0bddd080874a4606d9d1a164e"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-03-02 16:24:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-07-16 16:01:51.000000000","tz":0},"subject":"NetApp: Support force delete volume","message":"NetApp: Support force delete volume\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. But this is global\nconfig option and for debugging the value is kept close to higher limit.\nHowever, sometime large number of shares created and deleted within\nshort span of time and if such shares goes to recovery queue, the volume\ncreation limit hits. So these volumes dont need to stay in recovery\nqueue at all i.e. they can be force deleted.\n\nAdded config option \u0027netapp_force_delete_time_minutes\u0027 which can be\nconfigured to non-zero value to specify minutes. If share alive time is\nless than configured value, share will be force deleted. Config value 0\nsignify shares will always go in recovery queue.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ace8fcfb98832014725bdaa3a788d760b60729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84ace8fcfb98832014725bdaa3a788d760b60729"}]},"branch":"refs/heads/master"},"1919cc45ac9445bbc5c0eb5aad746a5bc09edcad":{"kind":"REWORK","_number":13,"created":"2026-07-16 16:26:43.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/13"}}},"commit":{"parents":[{"commit":"19f3340ad8da51d0bddd080874a4606d9d1a164e","subject":"Merge \"NetApp: skip efficiency settings on data-protection volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/19f3340ad8da51d0bddd080874a4606d9d1a164e"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-07-16 16:20:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-07-16 16:26:16.000000000","tz":0},"subject":"NetApp: Support force delete volume via share type extra spec","message":"NetApp: Support force delete volume via share type extra spec\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. However, sometimes\nlarge number of shares created and deleted within short span of time\nhit the volume creation limit.\n\nThe force delete window is now controlled per share type via extra spec\nnetapp:force_delete_time_minutes. If the share lifetime is less than\nthe configured minutes, it is force deleted bypassing the recovery\nqueue. Value 0 (default) disables force delete. On force delete failure,\nfalls back to regular soft delete.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1919cc45ac9445bbc5c0eb5aad746a5bc09edcad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1919cc45ac9445bbc5c0eb5aad746a5bc09edcad"}]},"branch":"refs/heads/master"},"43c858dc75b44e5cef512d3d4d30c30fd9472826":{"kind":"REWORK","_number":14,"created":"2026-07-30 13:30:51.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/80/978780/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/80/978780/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/80/978780/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/80/978780/14"}}},"commit":{"parents":[{"commit":"280e481ce4d6c289b96a0aa9d91341d434475d1c","subject":"NetApp: fix replica promote failing for shares with mount_point_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/280e481ce4d6c289b96a0aa9d91341d434475d1c"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-07-16 16:20:32.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2026-07-30 13:30:43.000000000","tz":0},"subject":"NetApp: Support force delete volume via share type extra spec","message":"NetApp: Support force delete volume via share type extra spec\n\nWith delete_retention_hours, NetApp backend deployment can decide how\nlong shares/volumes will remain in recovery queue. However, sometimes\nlarge number of shares created and deleted within short span of time\nhit the volume creation limit.\n\nThe force delete window is now controlled per share type via extra spec\nnetapp:force_delete_time_minutes. If the share lifetime is less than\nthe configured minutes, it is force deleted bypassing the recovery\nqueue. Value 0 (default) disables force delete. On force delete failure,\nfalls back to regular soft delete.\n\nCloses-bug: #2143047\nChange-Id: Icb02c880c305ad380b9dc9c0f96747640f396b51\nCo-authored-by: Maurice Escher \u003cmaurice.escher@sap.com\u003e\nSigned-off-by: Kiran Pawar \u003ckinpaa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/43c858dc75b44e5cef512d3d4d30c30fd9472826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/43c858dc75b44e5cef512d3d4d30c30fd9472826"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
