)]}'
{"id":"openstack%2Fmanila~978839","triplet_id":"openstack%2Fmanila~stable%2F2025.2~Id714b4ae20512b819dcc461a9c91316fe11fd00c","project":"openstack/manila","branch":"stable/2025.2","topic":"bug/2143129-stable/2025.2","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-16 17:09:46.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-16 18:40:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id714b4ae20512b819dcc461a9c91316fe11fd00c","subject":"decorate wrapper method with functools.wrap to keep func name in hooks","status":"MERGED","created":"2026-03-04 02:32:46.000000000","updated":"2026-03-16 18:42:35.000000000","submitted":"2026-03-16 18:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978839-bug/2143129-stable/2025.2","cherry_pick_of_change":978699,"cherry_pick_of_patch_set":2,"meta_rev_id":"4c955b8b097ae6ed3472c2d5da46250d594268ef","_number":978839,"virtual_id_number":978839,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-16 18:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-03-16 17:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-16 17:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-03-16 17:09:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-04 02:32:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":35979,"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","username":"sylvanld"},"state":"CC"},{"updated":"2026-03-04 03:20:49.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-03-04 04:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 06:08:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-16 17:07:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-16 17:07:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"a70bcd1c63d3e9834e03ebf2a142ba055cad7b4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-04 02:32:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f20cb4b7bcf93c4b1b987fb0a293536102b73e38","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-03-04 03:20:49.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/978839","accounts_in_message":[],"_revision_number":1},{"id":"83c53e5725363ba39bb9314fd87d7eab00c6baed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 04:58:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3887024315fe4424b2814739e2583e59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9cee290be14cfba39eb1a58e07c5cd : SUCCESS in 7m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0620ad86731a444db9381b4bed304857 : SUCCESS in 31m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e091f40d21b446e79307678321804ef7 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a1e730754c24361b7cd609ff4e4021b : SUCCESS in 9m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5e19cf970f248f4ae9cdc67f0e216b5 : SUCCESS in 6m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2de05d063dfc47db8e0816b339f6cd9e : SUCCESS in 15m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5824936c82dc4a5783fed6ffc813536d : SUCCESS in 8m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ded84657acc04447bd9aeefe9930d936 : SUCCESS in 1h 04m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9ed7d991bf1d40979c7dc4722b104a98 : SUCCESS in 2h 00m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e7a4f625eb0e4fcb8c43b40660ab106c : FAILURE in 1h 24m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c47a6bc2dde94e34affbe1c29192d2b4 : FAILURE in 52m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e27abe15b9514a8da49cecfd0c54a5ef : SUCCESS in 1h 20m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c51be86b6a3147ec898b5324a4a9fbb4 : FAILURE in 11m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0ae2e2a4f91c4de39de817cf7a25347e : SUCCESS in 6m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34a56969fd1e44c88aa8eca426a7e22c : SUCCESS in 5m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/507e1a7b937e41fc847e9c33a80626ca : SUCCESS in 40m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/61ecf475bb734e46937b741fb8ff2589 : SUCCESS in 43m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d022772bd91f49cb90d0730b31c59a5a : FAILURE in 33m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/57cba98daff94df68afdb8ff75f7b687 : FAILURE in 13m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/71851c57397b4d0b9954597d8dcf8264 : SUCCESS in 56m 20s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/411885938b7549149416684b8c7ef83a : SUCCESS in 46m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d83e81ebd3b64a4499df53dd391a2ad3 : SUCCESS in 32m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c39021fe1fe34efdbb509121d35911b1 : SUCCESS in 41m 04s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a348b967c0d240dd8e0cc51dae47a5c6 : FAILURE in 3m 02s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/fb27425e41a34a82a57327de8454c8e0 : SUCCESS in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0833377d0fb0344f37cb805a4787e4bbbe30634","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-04 06:08:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/978839/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/c17f105/ : SUCCESS in 1h 07m 19s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6698ef623a13b73f58e67573b40f39e060aafee2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 17:07:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1224c18ffb8c364e5e5b68c9cd46c6a5d588738f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-16 17:09:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec0a49f930d2e7ef7633b2584fb3e03dd3066c96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 17:10:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24bc653c9d200e4e4f08f88a22934639f38feee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 18:40:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e88ff531bd43ce846112b7f5f32f9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a07e5202634d7c9d991eba8ee6c0d5 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c49a8538a04c9eb42354ce296d2cff : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/667ce2a405df436f8407f30e465f1f43 : SUCCESS in 10m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab94d4c81757457cac2a01d99e9b8eb2 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f997486cb13f4053b6e5acafc335d81c : SUCCESS in 5m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d2562e28f6504e4fa3352a75a4e44058 : SUCCESS in 1h 24m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cba9021f3af342268b20f7e9e69cca55 : SUCCESS in 37m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a4cff615ce66463cac8445d51441b142 : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"42893c801dc8727602dec815ac85998a78f2d29d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 18:40:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c955b8b097ae6ed3472c2d5da46250d594268ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 18:42:35.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e908dfb7fc5f477b93417aa234443216\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d350e6e1c31c4890bf5d0f35f77e51a9 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f10d290b2aac442f802e22a4f3ab1410 : FAILURE in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36080eb16057cbff23d8c32c1123ca3dc2218503","revisions":{"36080eb16057cbff23d8c32c1123ca3dc2218503":{"kind":"REWORK","_number":1,"created":"2026-03-04 02:32:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/978839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/978839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/978839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/978839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/978839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/978839/1"}}},"commit":{"parents":[{"commit":"4d5004ead3ffde0a6c5df9e35315254f85c9501a","subject":"Merge \"Fix rally jobs\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4d5004ead3ffde0a6c5df9e35315254f85c9501a"}]}],"author":{"name":"Sylvan Le Deunff","email":"sylvan.le-deunff@ovhcloud.com","date":"2026-03-03 16:34:56.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-03-04 02:32:46.000000000","tz":0},"subject":"decorate wrapper method with functools.wrap to keep func name in hooks","message":"decorate wrapper method with functools.wrap to keep func name in hooks\n\nCloses-Bug: #2143129\nChange-Id: Id714b4ae20512b819dcc461a9c91316fe11fd00c\nSigned-off-by: Sylvan Le Deunff \u003csylvan.le-deunff@ovhcloud.com\u003e\n(cherry picked from commit 6df5b420821522ddc7773bca5e8fa9da031448cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/36080eb16057cbff23d8c32c1123ca3dc2218503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/36080eb16057cbff23d8c32c1123ca3dc2218503"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
