)]}'
{"id":"openstack%2Fmanila~980392","triplet_id":"openstack%2Fmanila~stable%2F2026.1~I6dbfe9cc10c8b1d5d92a3ef6f0ccaf68507e0fb6","project":"openstack/manila","branch":"stable/2026.1","topic":"create-2026.1","hashtags":[],"change_id":"I6dbfe9cc10c8b1d5d92a3ef6f0ccaf68507e0fb6","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","status":"MERGED","created":"2026-03-13 07:15:13.000000000","updated":"2026-03-13 08:31:57.000000000","submitted":"2026-03-13 08:31:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980392-create-2026.1","meta_rev_id":"12b4c68a44e17078f6bbc37afa742d686d622633","_number":980392,"virtual_id_number":980392,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-13 08:31:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-03-13 08:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-03-13 08:14:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 07:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-13 08:10:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-13 08:14:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-03-13 08:14:37.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"b9f6b2d6c35d2ad5c1578280aa585814aa62b24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-03-13 07:15:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c37c34fa6c74e4b003860c07ff918407e6f31bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 07:54:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd39d29e5c6c4231940e9faf9f7f6dd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0cff9630b14617a4531b4112cac3ef : SUCCESS in 13m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7ba7d7b748a4d53b01651a2bbb3cd25 : SUCCESS in 32m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53453f616a2e49218f64c3fc30cb3d29 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6394875f15004afaaf6d37f1d9594e2e : SUCCESS in 9m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b453830e0b124091afba9e22504b24d5 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/117d103486294073924004e57cc08577 : SUCCESS in 9m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0ea81ee933d5471c8214d57b2ca966ce : SUCCESS in 4m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/90b664f468c54acfbcee2fef1b2c7883 : SUCCESS in 5m 21s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/7a46a349dd974926ac64f96f107488c0 : SUCCESS in 8m 07s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4cf6aa28b67542eaa29726a9edf75b02 : RETRY_LIMIT in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cf0922ddd2e1af36cf45c7cf27829bf4d3f76f9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-13 08:10:50.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"691f78a84390ea444d33132a7ce2da387f14e085","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-13 08:14:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df5abcc6b5687cdd8498ddef4a7ddfb0c63bc918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 08:14:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"003a348ad8063d64f2e26f9cbb02ce2bc2c72bb3","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-03-13 08:14:37.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/980391 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"b74123fa0d33170806d6c8e121124f29e2907e0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 08:31:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55304259b9464b8d930c578ebd508706\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d2b3b5f0b74987b82b171161531e57 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c173e0fc2d4ad5ba950e2dcbcc0c3d : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d96ef5856434481beb11dc41cbe45fc : SUCCESS in 6m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff5a096ff899476eb5f3b40ff939cf14 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/088a32f977b447c58e84d978bb24dfdc : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b28b1d8e5ef7fda6f20e9dd9a5572de221e127d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 08:31:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"12b4c68a44e17078f6bbc37afa742d686d622633","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 08:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e1ce1546bb41dda1ccc527b0da2306\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a586ed66d7964566b7dd5d1f9b63bc1f : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2fa2355e3b4472b92c0c164c61ba9dd : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c8789a1d6dccd7578315a633a7b8bc7c62be9f8","revisions":{"9c8789a1d6dccd7578315a633a7b8bc7c62be9f8":{"kind":"REWORK","_number":1,"created":"2026-03-13 07:15:13.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/92/980392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/92/980392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/980392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/980392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/980392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/92/980392/1"}}},"commit":{"parents":[{"commit":"a051aac25ac2239b31333e5e7ce5ef6697070215","subject":"Update .gitreview for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a051aac25ac2239b31333e5e7ce5ef6697070215"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-13 07:15:12.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-13 07:15:12.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I6dbfe9cc10c8b1d5d92a3ef6f0ccaf68507e0fb6\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9c8789a1d6dccd7578315a633a7b8bc7c62be9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9c8789a1d6dccd7578315a633a7b8bc7c62be9f8"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
