)]}'
{"id":"openstack%2Fmanila~982136","triplet_id":"openstack%2Fmanila~master~I1fd6923f470539ccde0d1fe4c5d5110f2ffa859d","project":"openstack/manila","branch":"master","topic":"bug/2146259","attention_set":{},"removed_from_attention_set":{"38946":{"account":{"_account_id":38946,"name":"Leslie Mwendwa","display_name":"Leslie","email":"lesmwendwa@gmail.com","username":"lesmadeit"},"last_update":"2026-03-25 17:37:34.000000000","reason":"Change was submitted"}},"hashtags":["outreachy-2026-prospect"],"change_id":"I1fd6923f470539ccde0d1fe4c5d5110f2ffa859d","subject":"textapi-ref: Fix type for group_snapshot_links parameter","status":"MERGED","created":"2026-03-25 17:00:12.000000000","updated":"2026-03-31 20:08:10.000000000","submitted":"2026-03-25 17:37:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982136-bug/2146259","meta_rev_id":"0bcbc4539d2fd33c793a2091c0c218b8a21b9ad3","_number":982136,"virtual_id_number":982136,"owner":{"_account_id":38946,"name":"Leslie Mwendwa","display_name":"Leslie","email":"lesmwendwa@gmail.com","username":"lesmadeit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-25 17:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-03-25 17:25:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-03-25 17:25:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-25 17:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-25 17:25:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-25 18:45:23.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"30fae4d2b8d2dfb0dce15002898abfc495ecf67e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38946,"name":"Leslie Mwendwa","display_name":"Leslie","email":"lesmwendwa@gmail.com","username":"lesmadeit"},"date":"2026-03-25 17:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d61c1da44a0515f9b10c261b49d9e56934e0edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:19:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c65cfa63b1d494da3e79f9056a835af\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b382f4da2e324d2ea27da05f2594c529 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c3a19a48889447fa020ae7dcdbf3baa : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da9c4d484543476b9ab23870f11a0ce7 : SUCCESS in 4m 43s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a09db567ee45486083c5c7ada363674f : SUCCESS in 4m 36s\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/646f072cb29b4283878ecd09a0fd757a : POST_FAILURE in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0f6588f1c6b33d8a5d6ec0c43ab352dcc7df9d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-25 17:25:29.000000000","message":"Topic fix-group-snapshot-links-type removed","accounts_in_message":[],"_revision_number":1},{"id":"06f88b164b36277f0ed7376f5d0f1424037d9e4c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-25 17:25:34.000000000","message":"Topic set to bug/2146259","accounts_in_message":[],"_revision_number":1},{"id":"be75b52c488f8d5eda7cee12b428228fda1a817e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-25 17:25:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8cc736268b92e18a7acab7aafe8f063f1dceede","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-25 17:25:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"382fb4c5541c2c1875bbd71bca1e3aec46996d00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:25:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aacdc25b343b0a0af667cf56db45e287b9ff462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:37:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81cec840a8464a47bf77774f8a8e3eab\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/905556c218b04976a2bf6f4e4c492d55 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669e1a08317f4bcaac5bc6e16f28d674 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1a1581757b47b4a91a93969fcf4f74 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"59233938b689d80dea1877c8a1acd83aab63383a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:37:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2084aca40341c5ede6c25f3a92abb16a9bc41609","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:38:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a7007054ac34e4887158c96737465fb\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/fde5b79daa3545a8bccb25225bb56a99 : SUCCESS in 50s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0d5a347fb247ce9e033aef32defabe : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"08ad5c40c2f982e07989f3e5854c9bd7b5f286b9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-25 18:45:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/982136/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/30f33bb/ : SUCCESS in 1h 24m 23s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"0bcbc4539d2fd33c793a2091c0c218b8a21b9ad3","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-31 20:08:10.000000000","message":"Hashtag added: outreachy-2026-prospect","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f82eef7da60c90d69b941d415474ae1b6e7b417f","revisions":{"f82eef7da60c90d69b941d415474ae1b6e7b417f":{"kind":"REWORK","_number":1,"created":"2026-03-25 17:00:12.000000000","uploader":{"_account_id":38946,"name":"Leslie Mwendwa","display_name":"Leslie","email":"lesmwendwa@gmail.com","username":"lesmadeit"},"ref":"refs/changes/36/982136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/982136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/982136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/982136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/982136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/982136/1"}}},"commit":{"parents":[{"commit":"0e56a348aa9a0300d9eec1e673d614e27e0e6ae4","subject":"Fix typographical errors in Manila documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0e56a348aa9a0300d9eec1e673d614e27e0e6ae4"}]}],"author":{"name":"Leslie Mwendwa","email":"lesmwendwa@gmail.com","date":"2026-03-25 16:56:34.000000000","tz":180},"committer":{"name":"Leslie Mwendwa","email":"lesmwendwa@gmail.com","date":"2026-03-25 16:56:34.000000000","tz":180},"subject":"textapi-ref: Fix type for group_snapshot_links parameter","message":"textapi-ref: Fix type for group_snapshot_links parameter\n\ngroup_snapshot_links was documented as type: string, but it should\nbe type: array because links are arrays of objects.\n\nCloses-Bug: #2146259\n\nChange-Id: I1fd6923f470539ccde0d1fe4c5d5110f2ffa859d\nSigned-off-by: Leslie Mwendwa \u003clesmwendwa@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f82eef7da60c90d69b941d415474ae1b6e7b417f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f82eef7da60c90d69b941d415474ae1b6e7b417f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
