)]}'
{"id":"openstack%2Fmanila~983749","triplet_id":"openstack%2Fmanila~master~Ie7b29c3f080c3c6e560c7f2de7e36b2da05f8809","project":"openstack/manila","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-08 18:13:26.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie7b29c3f080c3c6e560c7f2de7e36b2da05f8809","subject":"wip: Look up barbican endpoint from initial service_catalog","status":"NEW","created":"2026-04-08 15:28:43.000000000","updated":"2026-04-08 18:13:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c3fe8dc1fec32baf019d9b93d882f15721e0ae1","_number":983749,"virtual_id_number":983749,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-08 18:13:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-04-08 16:42:43.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-08 15:29:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2026-04-08 17:31:58.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-04-08 18:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa7f9b340ecead5f0aaecb67c5dd409123701385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-08 15:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95e9e70555417d89a7fb9f0c941effec0b3800e0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-08 15:29:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6acea1abc8d54ef8d355f247b0b66b4521d74dfd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-08 15:35:14.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c94620d3e1e7a51318d9b26144d7d4a47c745ceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-08 16:01:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"71b66b743b6d70950314cebe9e96c4b61a176825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-08 16:25:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cd7c4a3af6f419fc2aeeeb77a9918b1ed3fc63a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-08 16:42:43.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7983eb6687ceddc8fc7b531cec9696b0cf69f3b5","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-08 17:31:58.000000000","message":"Patch Set 3:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/983749","accounts_in_message":[],"_revision_number":3},{"id":"1c3fe8dc1fec32baf019d9b93d882f15721e0ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 18:13:26.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39f453138d4447809afdfe9bbe5c35c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e703728d23de47cd864b84eb091372a2 : SUCCESS in 10m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be479f7a0cde4569aab4bc57dbaf800c : FAILURE in 17m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4f278cf32b4071b6657ae744e11f89 : FAILURE in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b027e5e69b94dd4af7daf23e8d91252 : SUCCESS in 9m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcb5523b47d54c83b8bcc7f5489283c9 : SUCCESS in 7m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98ffda354e72433793d42bf8e9c970fe : SUCCESS in 12m 31s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ca63229835d744dd95e30f0fb43ad5e2 : SUCCESS in 35m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4d7f45dc163c42f7901460586ddac02c : FAILURE in 56m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6c063f98f8914b34bc29ffa4ce678bf7 : SUCCESS in 1h 42m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b5313c0847cd4e6592eec976710fcd1f : SUCCESS in 1h 04m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/fce3d3f1de494e388c68a640e05d75f5 : SUCCESS in 46m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/17f49644a355438cacf639f08d1a4b0c : FAILURE in 1h 08m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e04de43473a748888aa0511e81820216 : SUCCESS in 3m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06d4ef9f41874b9099c9bfe8297952d9 : SUCCESS in 4m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3982d3dbeb15432c8a1206f96f100ad6 : SUCCESS in 39m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6e33c1d4cca74a73956cc56cf9a4990b : SUCCESS in 33m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7acc47185d0546ee86c97b11c31f7234 : FAILURE in 34m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/099b0025857740fba688f16d2cb34655 : FAILURE in 49m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips-py311 https://zuul.opendev.org/t/openstack/build/b4eaf54c0c7b4de582e4deb72f0ee95a : SUCCESS in 1h 32m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/942735d773d5429584615bd207240117 : SUCCESS in 46m 19s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/eafcac372a3c4643b9379b2a318f2e82 : FAILURE in 14m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ccb0626c82f24cdca3715c8600d1a0a8 : FAILURE in 30m 42s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/1599fd2335fb43e3817d2956c480e708 : FAILURE in 2m 47s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/7aa0a4c33a284d54bad8c10c7017056c : SUCCESS in 7m 55s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/88eee87e7b5d4d289272733d222741de : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"105122e1b91078d7612f4290ac237b2f4daa0191","revisions":{"f60f4759c599c4ffe7abb8bfa8c2d844bb83657f":{"kind":"REWORK","_number":1,"created":"2026-04-08 15:28:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/983749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/983749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/983749/1"}}},"commit":{"parents":[{"commit":"773dc6c428854dc4d3ae27b986f9f404cef01eb1","subject":"Merge \"PowerScale documentation update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/773dc6c428854dc4d3ae27b986f9f404cef01eb1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-08 15:27:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-08 15:28:35.000000000","tz":540},"subject":"Keep barbican endpoints in service catalog information","message":"Keep barbican endpoints in service catalog information\n\n... so that barbican key manager can leverage the information obtained\nonce, instead of querying the same from keystone API again.\n\nChange-Id: Ie7b29c3f080c3c6e560c7f2de7e36b2da05f8809\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f60f4759c599c4ffe7abb8bfa8c2d844bb83657f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f60f4759c599c4ffe7abb8bfa8c2d844bb83657f"}]},"branch":"refs/heads/master"},"3bda4801f1a9e6cc445cf10aedc82f1121f067ee":{"kind":"REWORK","_number":2,"created":"2026-04-08 16:01:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/983749/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/983749/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/983749/2"}}},"commit":{"parents":[{"commit":"201ac9b5bbdb9265dc0e442716709796b920fc6d","subject":"Merge \"Migrate grenade resources.sh from manila CLI to openstack CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/201ac9b5bbdb9265dc0e442716709796b920fc6d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-08 15:27:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-08 16:00:43.000000000","tz":540},"subject":"wip: Look up barbican endpoint from initial service_catalog","message":"wip: Look up barbican endpoint from initial service_catalog\n\n... so that barbican key manager can leverage the information obtained\nonce, instead of querying the same from keystone API again.\n\nChange-Id: Ie7b29c3f080c3c6e560c7f2de7e36b2da05f8809\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3bda4801f1a9e6cc445cf10aedc82f1121f067ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3bda4801f1a9e6cc445cf10aedc82f1121f067ee"}]},"branch":"refs/heads/master"},"105122e1b91078d7612f4290ac237b2f4daa0191":{"kind":"REWORK","_number":3,"created":"2026-04-08 16:25:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/983749/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/983749/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/983749/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/983749/3"}}},"commit":{"parents":[{"commit":"201ac9b5bbdb9265dc0e442716709796b920fc6d","subject":"Merge \"Migrate grenade resources.sh from manila CLI to openstack CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/201ac9b5bbdb9265dc0e442716709796b920fc6d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-08 15:27:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-08 16:25:10.000000000","tz":540},"subject":"wip: Look up barbican endpoint from initial service_catalog","message":"wip: Look up barbican endpoint from initial service_catalog\n\n... so that barbican key manager can leverage the information obtained\nonce, instead of querying the same from keystone API again.\n\nChange-Id: Ie7b29c3f080c3c6e560c7f2de7e36b2da05f8809\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/105122e1b91078d7612f4290ac237b2f4daa0191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/105122e1b91078d7612f4290ac237b2f4daa0191"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
