)]}'
{"id":"openstack%2Fmanila~984151","triplet_id":"openstack%2Fmanila~master~I2aefa202f91a6f6edf3d3b2c03351eacb47f4308","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"38967":{"account":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"last_update":"2026-04-12 15:09:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2aefa202f91a6f6edf3d3b2c03351eacb47f4308","subject":"doc: document encryption_key_ref prerequisites and microversion","status":"ABANDONED","created":"2026-04-11 12:28:39.000000000","updated":"2026-04-12 15:09:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e6843c98a838dbb8821236266187839efb690823","_number":984151,"virtual_id_number":984151,"owner":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-11 12:43:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-11 12:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-11 13:21:44.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-04-11 14:07:41.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"11b3f77f8a251367f040043cb4d46ee9ec25bd4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-04-11 12:28:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"848e052e62222c2bfda397e42ea0b1388688d62e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 12:43:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75f56c19d734a2588978924b03cc098\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09afe4e70c164643a439d2837977d103 : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3561dfa937384ff9b14aae112842e90c : SUCCESS in 3m 38s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/48a7278e11a144f3b1d7a422f9460df3 : SUCCESS in 5m 42s\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/fbb314951d844fd0998346088948468b : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc756ba7f8050d3f2bce5f592b939dfa0b9e44b4","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-11 13:21:44.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/984151","accounts_in_message":[],"_revision_number":1},{"id":"34e16f0c721a39d217898e9cfdff370617283a3d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-11 14:07:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/984151/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/ec2e180/ : SUCCESS in 1h 22m 58s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/984151/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/e11ba6a/ : RETRY_LIMIT in 8m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e6843c98a838dbb8821236266187839efb690823","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-04-12 15:09:48.000000000","message":"Abandoned\n\ni had errenously reported a manila-germane bug in the python-manila client lauchpad","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e9300eafb219f223d0b74dc1296c56dff26ba81","revisions":{"1e9300eafb219f223d0b74dc1296c56dff26ba81":{"kind":"REWORK","_number":1,"created":"2026-04-11 12:28:39.000000000","uploader":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"ref":"refs/changes/51/984151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/51/984151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/984151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/984151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/984151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/51/984151/1"}}},"commit":{"parents":[{"commit":"201ac9b5bbdb9265dc0e442716709796b920fc6d","subject":"Merge \"Migrate grenade resources.sh from manila CLI to openstack CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/201ac9b5bbdb9265dc0e442716709796b920fc6d"}]}],"author":{"name":"Insitu","email":"insitupeter@gmail.com","date":"2026-04-09 22:36:11.000000000","tz":0},"committer":{"name":"Insitu","email":"insitupeter@gmail.com","date":"2026-04-11 12:19:52.000000000","tz":0},"subject":"doc: document encryption_key_ref prerequisites and microversion","message":"doc: document encryption_key_ref prerequisites and microversion\n\nThe existing documentation for creating shares with an encryption\nkey reference was missing critical prerequisites:\n- No mention of required API microversion 2.90\n- No mention that only NetApp ONTAP (cluster mode) supports it\n- No mention that driver_handles_share_servers\u003dTrue is required\n- No explanation of share vs share_server encryption modes\n- No guidance on creating the prerequisite share type\n\nCloses-Bug: #2147668\nChange-Id: I2aefa202f91a6f6edf3d3b2c03351eacb47f4308\nSigned-off-by: Insitu \u003cinsitupeter@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1e9300eafb219f223d0b74dc1296c56dff26ba81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1e9300eafb219f223d0b74dc1296c56dff26ba81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
