)]}'
{"id":"openstack%2Fmanila~984239","triplet_id":"openstack%2Fmanila~master~I600be7c8f7cfe0060790e4e2d6f1386f20d9f5c5","project":"openstack/manila","branch":"master","topic":"unit-tests-rpcapi","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-07-24 05:45:27.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-07-24 05:27:53.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"38471":{"account":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"last_update":"2026-07-24 05:45:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I600be7c8f7cfe0060790e4e2d6f1386f20d9f5c5","subject":"Improve test coverage for share/rpcapi","status":"MERGED","created":"2026-04-12 14:33:20.000000000","updated":"2026-07-24 05:46:31.000000000","submitted":"2026-07-24 05:45:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984239-unit-tests-rpcapi","meta_rev_id":"a74e621415ff98f38a12642e7df814a51e1fac06","_number":984239,"virtual_id_number":984239,"owner":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-24 05:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-07-24 05:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-07-24 05:27:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-12 15:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-12 15:19:26.000000000","updated_by":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-04-12 15:19:26.000000000","updated_by":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-04-12 15:21:13.000000000","updated_by":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"reviewer":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"state":"CC"},{"updated":"2026-04-12 16:35:19.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-07-19 07:22:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-20 17:23:20.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"5f94ee64aecf744d97b04b885c1e5ef359ac3887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"date":"2026-04-12 14:33:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4691ba68eb59a14159ef2928361df5b8207de311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 15:12:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b73236563b4002b8cd4d37033982dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a92785ac1d4d79bdca07a5636389c1 : SUCCESS in 12m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/022e1f3e85e6499092c6cb18c69fb1a9 : SUCCESS in 33m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca27fe44476e4726bb90f8d3ec1f726a : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eacd79e7362c46159793dcc88149f1a6 : SUCCESS in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26ad58b990b94ea398146b4dfe940dc7 : SUCCESS in 8m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/938354845bef4b67b76dd1c97ac61827 : SUCCESS in 11m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/75dffd59f6614a05b3e7290e184cb177 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/39224be19ec14bd3adc33a723780d348 : SUCCESS in 5m 20s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/fa27cd1e02924821bac4eec371759b8f : SUCCESS in 8m 15s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ab23305a73b54d93b6b8a07f0cbedf70 : SUCCESS in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"383373189f8aa8de30610108ff0c915bfce60849","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-04-12 15:21:13.000000000","message":"Patch Set 1:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/984239","accounts_in_message":[],"_revision_number":1},{"id":"45c15b358ef3ca03bdab2db3cf7c2348cb0bf301","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-12 16:35:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/984239/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/f2dc74f/ : SUCCESS in 1h 21m 29s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/984239/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/f81e19e/ : RETRY_LIMIT in 8m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"301c55ddb9e758234940dc5162dfe599235d334c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-15 14:48:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23bdcf778876f64fc5d1e0519c70f4f9646831cd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-19 06:16:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4711b9118f58eae2b03a8cec39b5d53415565517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"date":"2026-07-19 07:02:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b6fa7fdb012f23c357674abbdae6acc5f19a9fe0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-19 07:22:37.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Manila-CI/tree/gh-pages/59466675315a4fda95ca7aa38e01e5e2/","accounts_in_message":[],"_revision_number":2},{"id":"171bcefbdc9ea6793343ed51b0dae7266f84a55c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-19 07:32:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/59466675315a4fda95ca7aa38e01e5e2\n\n- pure-devstack-manila-tempest-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f243a40c55974195935f42c75d631069 : SUCCESS in 30m 15s\n- pure-devstack-manila-tempest-fa-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/955a1237ac694c3c8e0e30ebe7461cbd : FAILURE in 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d914ba4d40edd9010b58bd6377985d4bb75ee7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 07:40:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0be834de25142019a413c873a8cb359\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680f0a5e2ebb45c1b03bcb9ec1fcb683 : SUCCESS in 11m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/458a89e33c2449ba9a37e169c23a9f36 : SUCCESS in 33m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf8b31d9f04416e83a6244887298dae : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0676ba47b35d46d5b1aafb9605f186e9 : SUCCESS in 6m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dec8c1688492404d9f2dab0a88fdf4d4 : SUCCESS in 6m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16e08701b1444c8799b84d87be0a0f0f : SUCCESS in 13m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59375c5a9c824f4fb6a25b4d1e4f1ae0 : SUCCESS in 9m 14s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2fedb856f2af4b4d80296942308d13bf : SUCCESS in 4m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33b51538183242b1b0520397fe70d818 : SUCCESS in 5m 24s (non-voting)\n- manila-tox-py3-threading https://zuul.opendev.org/t/openstack/build/92f74b1b010d4cacad9e465678872e0e : FAILURE in 7m 56s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2e2ddadae2df4f35a111f84f726f890e : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c064643dba48b5b7bfd572b02e6c414ae33b499","author":{"_account_id":38657,"name":"ananta agarwalla","display_name":"HPE AlletraMP Manila CI","email":"ananta.agarwalla@hpe.com","username":"manillaci"},"date":"2026-07-19 08:23:23.000000000","message":"Patch Set 2:\n\nBuild Failed: Logs: https://github.com/hpe-storage/hpe_manila_logs/tree/main/new_logs/984239","accounts_in_message":[],"_revision_number":2},{"id":"23cce5e441c74fed053846d5e32c87d6139598a9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-19 08:49:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/984239/2/upstream-check-manila/manila-tempest-plugin-ontap-dhss/2ccf2f1/ : FAILURE in 1h 07m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"fe3a1a79cf8a2596c2ee8c3af0a02188e356b6cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-07-20 17:23:20.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/exascaler-manila-ci-gerrit/187/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"aef60dd278996e09b14c49860c0cebb455706661","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-24 05:27:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"51998c00ec00bbaeb02b5a9cff6fe527820abad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 05:28:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3f0be2ee8ea3055ca40f16a7a3e281f2e9b2cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 05:45:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84cc5d0cacb1405b8b0e084f99a915ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679881cdf6e74fe2b184487e560220ee : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8968a37f819c49b682fb56f8e1436269 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e71bd52001b948559dc49422b1a4a8c0 : SUCCESS in 8m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abdf68eb66884f6da7521ff36a7c725a : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28dc70862cab4e87bbc1937732ba2e62 : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":2},{"id":"aff9fc129ed973c9438dcf56dab2f67728e061b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 05:45:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a74e621415ff98f38a12642e7df814a51e1fac06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 05:46:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1c7c477b6534dc98158debcbe24271c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b56a22f58c37448488ae64d37fd881da : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbeef2162177bd03d88ddf4bc9601b0699c1fa0a","revisions":{"e2d00d96e751c2d56ce239079421380439a4386b":{"kind":"REWORK","_number":1,"created":"2026-04-12 14:33:20.000000000","uploader":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"ref":"refs/changes/39/984239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/984239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/984239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/984239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/984239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/984239/1"}}},"commit":{"parents":[{"commit":"201ac9b5bbdb9265dc0e442716709796b920fc6d","subject":"Merge \"Migrate grenade resources.sh from manila CLI to openstack CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/201ac9b5bbdb9265dc0e442716709796b920fc6d"}]}],"author":{"name":"Denver Baraka","email":"denverbaraka@gmail.com","date":"2026-04-12 14:27:13.000000000","tz":180},"committer":{"name":"Denver Baraka","email":"denverbaraka@gmail.com","date":"2026-04-12 14:27:13.000000000","tz":180},"subject":"Improve test coverage for share/rpcapi","message":"Improve test coverage for share/rpcapi\n\nSeveral methods had no unit tests in\nmanila/tests/share/test_rpcapi.py for the module\nmanila/share/rpcapi.py.\n\nThis patch adds unit tests for the following methods:\n- manage_share\n- unmanage_share\n- publish_service_capabilities\n- create_share_replica\n- create_backup\n- delete_backup\n- restore_backup\n\nCloses-Bug: #2148115\nChange-Id: I600be7c8f7cfe0060790e4e2d6f1386f20d9f5c5\nSigned-off-by: Denver Baraka \u003cdenverbaraka@gmail.com\u003e\nAssisted-by: ChatGPT\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e2d00d96e751c2d56ce239079421380439a4386b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e2d00d96e751c2d56ce239079421380439a4386b"}]},"branch":"refs/heads/master"},"cbeef2162177bd03d88ddf4bc9601b0699c1fa0a":{"kind":"REWORK","_number":2,"created":"2026-07-19 07:02:00.000000000","uploader":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"ref":"refs/changes/39/984239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/984239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/984239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/984239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/984239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/984239/2"}}},"commit":{"parents":[{"commit":"201ac9b5bbdb9265dc0e442716709796b920fc6d","subject":"Merge \"Migrate grenade resources.sh from manila CLI to openstack CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/201ac9b5bbdb9265dc0e442716709796b920fc6d"}]}],"author":{"name":"Denver Baraka","email":"denverbaraka@gmail.com","date":"2026-04-12 14:27:13.000000000","tz":180},"committer":{"name":"Denver Baraka","email":"denverbaraka@gmail.com","date":"2026-07-19 07:00:30.000000000","tz":180},"subject":"Improve test coverage for share/rpcapi","message":"Improve test coverage for share/rpcapi\n\nSeveral methods had no unit tests in\nmanila/tests/share/test_rpcapi.py for the module\nmanila/share/rpcapi.py.\n\nThis patch adds unit tests for the following methods:\n- manage_share\n- unmanage_share\n- publish_service_capabilities\n- create_share_replica\n- create_backup\n- delete_backup\n- restore_backup\n\nCloses-Bug: #2148115\nChange-Id: I600be7c8f7cfe0060790e4e2d6f1386f20d9f5c5\nSigned-off-by: Denver Baraka \u003cdenverbaraka@gmail.com\u003e\nAssisted-by: ChatGPT\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cbeef2162177bd03d88ddf4bc9601b0699c1fa0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cbeef2162177bd03d88ddf4bc9601b0699c1fa0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
